)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~834031","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I55a33f88356ff3d1ef6ed0ce83bb102172e46023","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"fip","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-03-22 10:00:50.000000000","reason":"Michał Dulko replied on the change"},"4727":{"account":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"last_update":"2022-03-23 12:21:01.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2022-03-23 10:43:41.000000000","reason":"Roman Dobosz replied on the change"}},"hashtags":[],"change_id":"I55a33f88356ff3d1ef6ed0ce83bb102172e46023","subject":"Use one pod to check connectivity in kuryr_restart","status":"MERGED","created":"2022-03-16 14:48:18.000000000","updated":"2022-03-23 12:22:49.000000000","submitted":"2022-03-23 12:21:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"834031-fip","meta_rev_id":"e5e05a506d965997ef0e4b1ef3f4e6277dd41fc7","_number":834031,"virtual_id_number":834031,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-23 12:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2022-03-22 10:00:50.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2022-03-23 10:45:48.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-23 10:43:41.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-23 10:43:41.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-16 15:02:23.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2022-03-16 17:15:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 07:42:11.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"CC"},{"updated":"2022-03-22 07:42:12.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"CC"},{"updated":"2022-03-23 10:43:41.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2022-03-23 10:45:48.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"}],"messages":[{"id":"181e5d1f72fcf363d6349d6b7614dec70770c264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-16 14:48:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3b97209461379322286e26cb8aa2cadb378f48c","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-16 15:02:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ec36ea135bb8f79fa07721dddb9e85cb95bd880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 17:15:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd438acfc1c0497b9d8af33645e7b139 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f023336c188b4cc7b39a73cd91e77c1a : SUCCESS in 5m 57s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/426970d7faa2495ebacc35d0305246ed : FAILURE in 1h 12m 52s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/4e7937f4b0a74b1298e336ddc13bb7fa : FAILURE in 1h 23m 51s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/5abb9d5e0cf14faf930222466c7a12a0 : SUCCESS in 1h 59m 19s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c41084d0eab84e8583ceef43055c264c : SUCCESS in 1h 11m 25s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/0e67c2fa0aa24be2a3a8dc13c989ae5a : FAILURE in 2h 02m 41s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/f127b1f0698b40079d07433e7fcee111 : SUCCESS in 1h 26m 19s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/dd0573f6577b4c58b18f040d21c02a9a : SUCCESS in 1h 28m 25s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/5c79eb6babd345398fdd7fd0d6b34d05 : FAILURE in 1h 43m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/30e08107bf4c412d9ac11b4843ccb77e : SUCCESS in 1h 34m 56s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/b0bae8c72a6b451abe3555983948451c : SUCCESS in 1h 32m 30s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/353b04361ed74528b90eb3b2a03c736f : SUCCESS in 1h 28m 28s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/cba70d823c0e49f4ad8d716d3116d0e4 : POST_FAILURE in 1h 41m 35s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/a28ebb977e2a45298fb8d24920637a8c : SUCCESS in 2h 20m 40s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/45e60a12156348c3b0c5711082332adc : SUCCESS in 1h 15m 58s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/fbf0feb08d344e48aeb91a07aa3a1285 : SUCCESS in 1h 13m 24s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/389e15558fa94f5cad269686882fdcbc : SUCCESS in 1h 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18c127929a2995f42e1c8d6f78385093f79985db","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-16 21:23:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe48ef09eb4546b534031e024b70d14b4d308e74","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-16 21:24:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbfda3daaa24a214759c26294cee294a6e895a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 00:19:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd626ef64ae4d64b41c4289cdc9c22b : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d0d11064c64cfb9c55002cd4e0dbd4 : SUCCESS in 5m 07s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/e556be338f2c465b805b26f9d1abeff6 : FAILURE in 51m 23s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/a0afd753d7c84db38564bc968376c5b5 : FAILURE in 1h 54m 31s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/0ec5938ae8a6453d9e11b2763423e8d7 : FAILURE in 1h 59m 58s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/631beac848124424834aa80b03918aeb : FAILURE in 1h 35m 58s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/34538c12ee0c4a60821925a6625d583a : FAILURE in 1h 38m 07s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/5fd6c535c1d84d728e7b0ab9da05d33f : SUCCESS in 1h 23m 20s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/d2b4167486664a3f98be0879a0896c9a : SUCCESS in 1h 21m 29s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/b1733399016f42499b191aa31ef8962a : SUCCESS in 1h 32m 57s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a2fcd6c05a4a4891b118202c73e36864 : SUCCESS in 1h 55m 10s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/251e4a5e4c7d4a6d95bfc70a4d87dafb : FAILURE in 1h 26m 54s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/aefba4c2de5543b1981fb75a0fb4a3f4 : SUCCESS in 1h 31m 03s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/165699f29469406f99b08a7beb5272c0 : POST_FAILURE in 2h 53m 06s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/6d8be7a496ed45639d816e9fd96043d6 : SUCCESS in 1h 38m 43s (non-voting)\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/7a96c4dda57b4c6d9a589c23afce101f : SUCCESS in 1h 30m 14s (non-voting)\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/470cda7c76294411905fc93c3370b6a7 : SUCCESS in 1h 10m 23s (non-voting)\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/c198b13a8c524f599c119e9cc9d58a3b : SUCCESS in 1h 07m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cbf30525f92a5114bec5325826ab3d6ced821a45","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-17 10:13:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"796be33063aedad4a983674c133c5b7607a92af5","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-17 10:46:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ec7c8c7ed7778df8cf312e1ab63c5aecb91a9ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-20 11:39:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7734cb5aedb5f57037fa6729135235730c44a3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-20 13:47:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50c7a5967cec3c96c5ac3c61b69d0ba9ad84835a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-20 16:19:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702e9ed14bd04d32ad5b7363ccc8218b : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffd9df0be89e42b99b53a640d016a7df : SUCCESS in 4m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/df89862b8e6149b7b2ae1cded94c53e2 : SUCCESS in 1h 47m 01s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/49f1fee2355d44818c5edb6a2603144c : SUCCESS in 1h 59m 40s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/550e68867f9d4038b202f3bccc0d014c : FAILURE in 1h 52m 18s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/9faff6d1616946a68228de3c9da0a7d4 : SUCCESS in 1h 33m 35s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/43b48117b7d04c1395e20e136dd6e009 : SUCCESS in 1h 17m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/e280be39e4394c3cbe6f6e04f36e8671 : SUCCESS in 1h 12m 34s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/eb8c46665fcf415893b4b6acfe5ba69d : SUCCESS in 1h 30m 21s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/31f4ca6d46204c62b3194b8d3826d31e : SUCCESS in 1h 23m 37s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/8ed6e088405042388e57f2ddf2f463c3 : SUCCESS in 1h 04m 01s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/f71672e5046041008016fceb29ba9d75 : FAILURE in 1h 40m 36s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/71b7747806f04edda4ecf05a6c271281 : SUCCESS in 1h 36m 48s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/d8655242f7a740d49addc125de8fca4e : POST_FAILURE in 2h 30m 13s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/968370088fe94485946a3a0869a0f5a4 : SUCCESS in 1h 04m 29s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/de859e38648c45a6be3399da9a45cae1 : SUCCESS in 1h 13m 58s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/8c8de93ae5d447b7b45776402eec0082 : SUCCESS in 1h 03m 47s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/373c558ffeea4ca4b4e38c18b4303a76 : SUCCESS in 50m 04s","accounts_in_message":[],"_revision_number":3},{"id":"a425d40d65320854fc4fed7839d676dea8ab4ba0","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-03-21 07:42:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e8891a2cd929d65d31bc5bef6c26cebb03d450ee","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-21 07:49:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"498d916640097d710d20c66d8676e7c8777e808f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 11:03:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1875c0a29ac427b996d767caff5cfca : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f6e0fabd8ac4d5c9d8eaedcfc73b0d5 : SUCCESS in 4m 42s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/8d2c8779e561424087f14521314cd806 : SUCCESS in 1h 35m 20s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/d5cd3f19bd9148ac8c27c46f0737aecb : SUCCESS in 1h 39m 18s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/e2f61adf806a40cabc0e8b1781f6892d : SUCCESS in 1h 21m 26s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/ca984beb48fe4403abecb644b6c90732 : SUCCESS in 1h 26m 13s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/97eab661c6f14ad49f4cc415a0f49fbc : SUCCESS in 1h 48m 18s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/797c017fb7b84c77b410c1744f35819d : SUCCESS in 1h 02m 39s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/85965f6db3e248dcbe1237a62a1a0529 : SUCCESS in 1h 00m 50s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/e97efb1ba6d346cd9ccc12b553bfb5b9 : SUCCESS in 1h 17m 28s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/67e37a536b2549c9975355bcd317b41d : SUCCESS in 1h 14m 32s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/eded3905c74c4f8b89e52d2b027151e3 : SUCCESS in 1h 42m 07s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/74dace4471744f2b917be434d57a4ca0 : SUCCESS in 1h 45m 55s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/f5a4b997ec304156a043797d69dee0d3 : POST_FAILURE in 3h 14m 55s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/01f25084c2d04439ae3cbabdf65cd8da : SUCCESS in 1h 59m 26s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/4516703ec2a14e2fbde2da28bbcf1ece : SUCCESS in 1h 09m 51s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/75aeadf82abc45008039f5a9870fda7e : SUCCESS in 1h 27m 24s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/eb94e3942daf487f8b623f0411fb997d : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":3},{"id":"661e6fc27daba28703c728701d3f2aa8f2839f58","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-21 11:16:02.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"64cb399b034bccee41a145db65d23de95a0099c7","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-21 13:28:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e24de65cc719eab397d1960db62aceb3f84d21c5","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-21 13:55:22.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f65c9e590d5f7bd1c7224f7e4d9c04e5810d5041","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-21 13:55:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1f7092e7706a1d7c6372b5a7615aa40d3d4633a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 16:45:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1dbaabaecae46279095a2e411d4e361 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a1cecb56ab84fa3a2cac5c0163f9190 : SUCCESS in 4m 56s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9eb5830ddee04ca8a4593373f1fac6b8 : SUCCESS in 1h 26m 45s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/60a026aebe7b4764ae40d1eea7efce12 : SUCCESS in 1h 35m 07s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/e7db3403c7c44a109978110a4f7d1036 : SUCCESS in 1h 18m 05s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/4aad6ae3a03847ebba03e57832a5ac8c : SUCCESS in 1h 16m 59s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/ba3e348e79114163aa7da902bcc009db : SUCCESS in 1h 30m 12s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/515fa89c3cc7473ebc5b01943a6d2ebc : SUCCESS in 1h 01m 05s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/b234531ed4874623ad6bd8059968e74f : SUCCESS in 1h 15m 54s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/1ad5017595e3437ba81190f433479f60 : SUCCESS in 1h 47m 23s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/0b870993473e414ca94af781573e61ce : SUCCESS in 1h 48m 34s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/f0a044e75188484eac62708ce0f05a6b : SUCCESS in 2h 32m 06s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/e7308cfae76f44ddb2335c2197144109 : SUCCESS in 2h 26m 06s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/91d43d7dae7f4582a7011b388fffce7f : POST_FAILURE in 2h 47m 14s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/1cacb0a48aa843ccbf34ba4039acfba7 : SUCCESS in 1h 42m 53s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/597ad12c82b44682a064e007fe089646 : SUCCESS in 1h 14m 36s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/76be192b63e64bfe8b31568335ab1c84 : SUCCESS in 57m 22s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/0ca4503990824d3a8ef91cfeefd0cc1b : SUCCESS in 1h 05m 26s","accounts_in_message":[],"_revision_number":4},{"id":"143f50b7d8cdfbd3ac62aaf22f497656bb33cbf3","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-22 10:00:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6dc58113aa3ea709abca199fd1301152d564d155","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-23 07:35:51.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"13250709b373d0057c03e828d585759dbb55e153","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-03-23 08:14:48.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"38b0a56b3f1f39ef0d1852d784005911ae47bd8a","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-03-23 08:30:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"18d2631e01fd41cec0fed2f49f9c2ed43e728644","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2022-03-23 10:29:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"10f537d0de6f421ff78369ccf452632e9656892f","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-03-23 10:43:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6db9fa9c18b3df01fe3217d9cbcfd74cba50bbdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 10:44:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e6673a1c930c9263d499750a1c09a125e46b7635","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-03-23 10:45:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c91bcf2b8028abb1632e4bab6017101bce236d44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 12:21:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270e7eadbd1e442c99ce8c67ee721576 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb9f99bc5d84ee488c6c1473affb430 : SUCCESS in 4m 48s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/56de340625534543b7bf2a2148139ecb : SUCCESS in 1h 35m 49s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/e0fb34841a1d494c803ae6168e3df849 : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":4},{"id":"23934d53a56a4ec24ab65801ed61f6053baae272","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 12:21:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e5e05a506d965997ef0e4b1ef3f4e6277dd41fc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 12:22:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66022347335141cf8c307c47b96ec29b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f6f77354b1cc424009b718517c338d5fc7669917","revisions":{"45fd82b9e3da4bccd4ac55f2b20e1e96603fa1ae":{"kind":"REWORK","_number":1,"created":"2022-03-16 14:48:18.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/31/834031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/31/834031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/1"}}},"commit":{"parents":[{"commit":"b667a5c6bc4845883cf10ea8817b97913c5e6b2c","subject":"Bump pod exec timeout in Service connectivity tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b667a5c6bc4845883cf10ea8817b97913c5e6b2c"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2022-03-16 14:47:04.000000000","tz":120},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2022-03-16 14:47:59.000000000","tz":120},"subject":"Delete the pod and fip test_kuryr_restart","message":"Delete the pod and fip test_kuryr_restart\n\nA pods is created and assigned a fip for each kuryr pod.\nDelete the pod and fip after they are created so as not to have a\nproblem in system with limited resources\n\nChange-Id: I55a33f88356ff3d1ef6ed0ce83bb102172e46023\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/45fd82b9e3da4bccd4ac55f2b20e1e96603fa1ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/45fd82b9e3da4bccd4ac55f2b20e1e96603fa1ae"}]},"branch":"refs/heads/master"},"814e6e68e1dea9df488717d1a22c1ffe0715f240":{"kind":"REWORK","_number":2,"created":"2022-03-20 11:39:29.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/31/834031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/31/834031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/2"}}},"commit":{"parents":[{"commit":"b667a5c6bc4845883cf10ea8817b97913c5e6b2c","subject":"Bump pod exec timeout in Service connectivity tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/b667a5c6bc4845883cf10ea8817b97913c5e6b2c"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2022-03-16 14:47:04.000000000","tz":120},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2022-03-20 11:36:19.000000000","tz":120},"subject":"Use one pod to check connectivity in kuryr_restart","message":"Use one pod to check connectivity in kuryr_restart\n\nIn test_kuryr_restart use one pod to check connectivity to the\ncreated pods from this pod instead of assigning a fip for each created\npod. This can solve problems when there is shortage of floating ips.\n\nChange-Id: I55a33f88356ff3d1ef6ed0ce83bb102172e46023\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/814e6e68e1dea9df488717d1a22c1ffe0715f240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/814e6e68e1dea9df488717d1a22c1ffe0715f240"}]},"branch":"refs/heads/master"},"bacbf215aad1047f240b6bc32a3ae2bac9453fec":{"kind":"REWORK","_number":3,"created":"2022-03-20 13:47:23.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/31/834031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/31/834031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/3"}}},"commit":{"parents":[{"commit":"0e68dcf892959326351c13ad103fb6a0a70e079a","subject":"Revert \"Temporarily disable stable jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0e68dcf892959326351c13ad103fb6a0a70e079a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2022-03-16 14:47:04.000000000","tz":120},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2022-03-20 13:46:41.000000000","tz":120},"subject":"Use one pod to check connectivity in kuryr_restart","message":"Use one pod to check connectivity in kuryr_restart\n\nIn test_kuryr_restart use one pod to check connectivity to the\ncreated pods from this pod instead of assigning a fip for each created\npod. This can solve problems when there is shortage of floating ips.\n\nAlso, Don\u0027t fail on deletion of a pod in cleanup if it\u0027s not found\n\nChange-Id: I55a33f88356ff3d1ef6ed0ce83bb102172e46023\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bacbf215aad1047f240b6bc32a3ae2bac9453fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/bacbf215aad1047f240b6bc32a3ae2bac9453fec"}]},"branch":"refs/heads/master"},"f6f77354b1cc424009b718517c338d5fc7669917":{"kind":"REWORK","_number":4,"created":"2022-03-21 13:55:46.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/31/834031/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/31/834031/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/834031/4"}}},"commit":{"parents":[{"commit":"0e68dcf892959326351c13ad103fb6a0a70e079a","subject":"Revert \"Temporarily disable stable jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0e68dcf892959326351c13ad103fb6a0a70e079a"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2022-03-16 14:47:04.000000000","tz":120},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2022-03-21 13:55:30.000000000","tz":120},"subject":"Use one pod to check connectivity in kuryr_restart","message":"Use one pod to check connectivity in kuryr_restart\n\nIn test_kuryr_restart use one pod to check connectivity to the\ncreated pods from this pod instead of assigning a fip for each created\npod. This can solve problems when there is shortage of floating ips.\n\nAlso, Don\u0027t fail on deletion of a pod in cleanup if it\u0027s not found\n\nChange-Id: I55a33f88356ff3d1ef6ed0ce83bb102172e46023\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f6f77354b1cc424009b718517c338d5fc7669917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/f6f77354b1cc424009b718517c338d5fc7669917"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
