)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~837812","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ia4591ef96ce642f53ead6147ff4b56960746e510","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"bp/specify-project-by-annotation","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-08-18 09:58:02.000000000","reason":"Change was submitted"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2022-07-11 08:48:45.000000000","reason":"\u003cGERRIT_ACCOUNT_27032\u003e replied on the change","reason_account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},"28329":{"account":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"last_update":"2022-08-18 09:58:02.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2022-07-21 12:04:26.000000000","reason":"\u003cGERRIT_ACCOUNT_13692\u003e replied on the change","reason_account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}},"hashtags":[],"change_id":"Ia4591ef96ce642f53ead6147ff4b56960746e510","subject":"Add annotation project driver scenario test cases","status":"MERGED","created":"2022-04-14 02:05:55.000000000","updated":"2022-08-18 09:59:20.000000000","submitted":"2022-08-18 09:58:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":51,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837812-bp/specify-project-by-annotation","meta_rev_id":"20a1a24e418e43fe6ad44b6dea3b26c56a36fa75","_number":837812,"virtual_id_number":837812,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 09:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2022-08-17 07:38:00.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2022-08-18 08:18:53.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":2,"date":"2022-08-18 07:50:28.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2022-08-18 08:18:53.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-14 02:19:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-25 11:37:54.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"CC"},{"updated":"2022-04-25 11:37:54.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2022-04-25 11:37:54.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2022-05-16 17:25:11.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"}],"messages":[{"id":"59838e7a2af67f3dba32c91f0f598b190d80c890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-04-14 02:05:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae89ba122e517c11c8ffb8e272e87b8f7b7f7adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 02:19:02.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"1bfcfd603918637eec8d9ab5d44bd5e265908988","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-04-14 02:42:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"51f8cc4b52394d0373bae729b03906cf09821d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 04:45:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb82f53cb944b7ea67ad8ce3a310dab : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e8e953293c4f66869b53275afd9d5d : SUCCESS in 3m 48s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/1421902e9b7142278664fb13b116d8f8 : SUCCESS in 1h 31m 37s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/9ccf142ddf974a1eb5eb48ddbcba5148 : SUCCESS in 1h 18m 10s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/48c78af874f44466996997bc346c3a73 : SUCCESS in 1h 16m 33s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/015cfa43f37849bfab21e1dd63377c8d : SUCCESS in 1h 26m 10s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/0e8c4bb0ba3347a7b59e34d85758a410 : SUCCESS in 1h 24m 51s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/67ca0ff359fb43d185cbb745db34c51f : SUCCESS in 1h 17m 53s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/7e9bd199c7a44b04ae7eae9a71fe158e : SUCCESS in 1h 13m 51s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/1dceb888969c4aeabbdd6062f949b388 : SUCCESS in 1h 32m 10s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/614892cb928b4ffc92de11eceea7ff5a : SUCCESS in 1h 43m 56s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/2604d6e161f64f8e84a75b2dd94eb832 : SUCCESS in 1h 56m 18s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/a8e81ae0061f483fb192fe0c3af4f34d : SUCCESS in 1h 42m 16s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/13fbe9b5173048ca8a6f687932454ca5 : POST_FAILURE in 1h 55m 31s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/34ebf9f52fc9425682c6135523c447e2 : SUCCESS in 1h 44m 59s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/d2e9dfc840fd404f8a28b18504433f54 : SUCCESS in 1h 12m 30s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/868d943f3776438bb48237e26b6afd73 : SUCCESS in 1h 18m 18s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/5ecb7906acbf42479bdb0aa438810f13 : FAILURE in 20m 16s","accounts_in_message":[],"_revision_number":2},{"id":"c002a1d2e89d03e62302d47aaa6b93c37224535c","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-04-15 06:28:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a64043ca014cfc3a31afacb9a72e77ad087875b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 09:09:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9beb9378104b97a181d14a5dc049e7 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf83028af7f49709e58b4538a999e4d : SUCCESS in 4m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/2ada1a85b3f44824a3375d8e6177e13b : SUCCESS in 1h 36m 28s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/6ffe31b7a09645f9a434b4613c8ca54e : SUCCESS in 1h 22m 35s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/a63f71f8ca234e54b580af6b980482c0 : SUCCESS in 1h 16m 15s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/cee915f30efe48e38df47eaa7d74341c : SUCCESS in 1h 10m 50s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/da149a8289c24f80afc7f3be4f436d2a : SUCCESS in 1h 26m 54s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/edfa18b617ff456793f14ad85358011a : SUCCESS in 58m 24s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/45a7a0a94a4e4463979ff4ac21297267 : SUCCESS in 54m 33s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/be14b959fe294b74b96457e1edf70138 : SUCCESS in 1h 05m 55s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/de30927cac6a4f55bd4257b31a9dd5b6 : SUCCESS in 1h 02m 38s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/6066235b0680412a8c6d60af7d420d3c : SUCCESS in 1h 40m 19s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/8bdde524684e45b48e2431ed4cf86dd7 : SUCCESS in 1h 50m 19s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/2217c8b70e044e82a9062f60cca5f436 : POST_FAILURE in 2h 37m 40s (non-voting)\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/da402eef4532416b80eac5735d60236e : SUCCESS in 1h 16m 15s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/d6796668986140eeaba1d47b1ae7fa68 : SUCCESS in 1h 01m 38s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/183bfabd206e457dbf302c7764b8f969 : SUCCESS in 59m 00s\n- kuryr-kubernetes-tempest-ussuri https://zuul.opendev.org/t/openstack/build/8ae8795094054bb9a8be9f46ad981553 : FAILURE in 11m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6e831420c0fd51b6ac84dc7bce50d750d22e5407","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-04-25 11:37:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97e63ff561dad60e9ec530d60a3a2e2339ee1664","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-04-29 06:07:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99689ab3a5fbc1ae3405f8204186e17a7e68f2a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-04-29 06:07:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"675c4de98120b3b564486f7cab6a8c14e95b179c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 07:23:20.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"4968591f7ee7a29a3f3a4a4bc4ae0380b0540203","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-04-29 09:33:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"014a021ec4d485d2095682f39dcb080e5f3504f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 11:19:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1110edd28f4c4e8f94de8c2941d9c3 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a17c5a0cf224881a136e5e8c29c856b : SUCCESS in 3m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7e7ffe4a34e443f1bc71b295e8d2b914 : SUCCESS in 1h 34m 49s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/e4987ed0d33044c382d725c468558175 : SUCCESS in 1h 39m 53s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/ed9358b32db549a8b2a2c98cc7446bcd : SUCCESS in 1h 09m 39s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/f91273c7d0a342c3afc6129512953562 : SUCCESS in 1h 22m 12s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/c3d1cbf4f8814a1aa36f16afca804db2 : SUCCESS in 1h 36m 24s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/a678d55174404765b15eb4c51df9ecee : SUCCESS in 58m 53s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/e8522f70a91b455584b2d742b90c1287 : SUCCESS in 53m 45s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/6626506474ba4b30b0627701f9a5a8ff : SUCCESS in 1h 03m 25s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/66cee6038eb0439baebc84261471551d : SUCCESS in 1h 08m 37s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/dafa986281e840808d52275455f4390d : SUCCESS in 1h 21m 54s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/644de1a5eb254c34b121be25b01ddef1 : SUCCESS in 1h 35m 26s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/bcf11f419c7f4154810ea2e66583fb99 : POST_FAILURE in 59m 11s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/1a80581e55a34a5996450ce79fa27d7b : SUCCESS in 1h 22m 26s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/396bf58af45a49919132b28a61588252 : SUCCESS in 1h 22m 07s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/4ed31c28be00449e852bd9151c5d77c2 : SUCCESS in 1h 31m 48s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/9e49dba6585043afb9ae0f1e12526e6b : SUCCESS in 1h 18m 11s","accounts_in_message":[],"_revision_number":3},{"id":"d9c6d278c2b37d034914f388884605f8d1b04e80","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-05-14 02:37:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a27ad8f1f454551668b298d073d9c3c3d683db89","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-05-16 17:25:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6427499879c515fa88b2b7133a048abc4bb7cbf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-05-21 08:46:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6bd1d788cc2c938ff7d9861ad66f8a6025d5c335","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-05-21 08:48:46.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e96bbdfebc57bdbc7a5569792089005b709c29df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-05-21 08:51:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"af9c46648b12c1a9e51288ccaee52060615c2dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 10:32:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d1ecc39f514400880ba528e53e87c5 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ffe722c720147f986e3f267b9842fd5 : SUCCESS in 3m 27s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d5e8ffa078874070885fa163c772e61e : FAILURE in 31m 00s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/5c92b5f4f3044091b6b193906c7460ed : FAILURE in 45m 12s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/bd99baf0fe084c8c9b7d4507ef77b698 : FAILURE in 18m 28s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/ef9d90ffe8254a689b8a65a4ce4daa06 : FAILURE in 30m 07s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/30288034d48742d6ba54497e2fae978d : FAILURE in 18m 02s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/1c9f7cd877b548b8bbf53663ff9054af : FAILURE in 18m 37s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/4387333a9faf435ebd2c4bc8c38c4368 : FAILURE in 15m 42s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/5ee2f233eb9c4891b1f13ad5acc9a783 : FAILURE in 14m 12s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/33a9cd187261402ba77d328ad8bb451b : FAILURE in 18m 12s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/8d9d78f779724e8cb8887f9247b11361 : FAILURE in 13m 35s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/2585bda01dd142a9a9b31866d5171242 : POST_FAILURE in 26m 11s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/5bd7db04525c4af18ffcfec28845acd2 : FAILURE in 27m 01s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/d3acd72af65149d098bcabafe6bce94d : SUCCESS in 1h 34m 46s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/6fbd83edf57d4afe8b613ccbbb806bf1 : SUCCESS in 1h 20m 08s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/d089fecbac354e8ab8e8d194fe66bd2b : SUCCESS in 1h 12m 56s","accounts_in_message":[],"_revision_number":5},{"id":"43fbb2d2c63eb1008a9bc5bffc244052bcd35efe","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-05-22 05:48:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"22a5fc7f580ade1baabcefd4e6795469fb10e923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 07:23:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cddd32421ef34c888dbad03c48ba106c : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf73073726f547a4b51f22a8759fec44 : SUCCESS in 3m 19s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d58accca951041c983c54ba4eab550f6 : FAILURE in 21m 10s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/1d90eb17c9494cb4b9136afd5197b4d8 : FAILURE in 28m 51s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/b4f91868e14a443ab19f87cd354c6a96 : FAILURE in 16m 32s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/3e5a7d8be92044dc84c1a17dd92718f5 : FAILURE in 19m 59s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/98f6903794b84ff98edace71ed11dc7a : FAILURE in 23m 25s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/ecca4e3af0be47da89f256e37a2f8e32 : FAILURE in 17m 38s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/8508a41e0184434bb424cf0ee6a7b689 : FAILURE in 18m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/7813131a0a794947b17d642e2a6684f8 : FAILURE in 16m 54s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/9c6746f40cc84a4d863f73612f63e5f5 : FAILURE in 18m 20s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/653192111f014431a49dcfca26fb3472 : FAILURE in 13m 37s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/183943eef34a4908908d11c2651ea14b : POST_FAILURE in 20m 37s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/c6514eb33f6940828f82ecbcb6dde797 : FAILURE in 26m 06s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/9692a620fc474951b94cf31921cb3bd5 : SUCCESS in 1h 33m 46s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/e204f7967c9740eca43d2b864f908d78 : SUCCESS in 1h 20m 29s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/99bc342ba70a40f8bb4a79e46de8fd67 : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":5},{"id":"ff37d2dd879070b84acf9852a4bdd3b53d472c37","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-05-23 06:17:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"45b7ec32a2e9512b8b531cd2c5c61e58cd32e395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 07:53:50.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4283adabf9984262bd0c2b3441fadfe8 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96d8f8554db549f0b05be1e454009017 : SUCCESS in 4m 28s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/87e2fa76cade431bb6111641586d5b64 : FAILURE in 25m 22s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/0be925857a8f4f9e84145ae5b9e216c9 : FAILURE in 36m 15s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/e0307a4415494f868f14826c1fc9fc10 : FAILURE in 18m 03s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/79b084ee55e24869af2df66ae3af2549 : FAILURE in 26m 23s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/daeb1b068eb64cb39b8884ceac1f4dac : FAILURE in 19m 27s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/dbd0b23af3f34efeb520a6b18cea7698 : FAILURE in 17m 55s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/2815f8de5d484ec79b5bb5bed384da94 : FAILURE in 18m 10s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/5c66743ce37b4d0dbd1613285a0c5b57 : FAILURE in 16m 27s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/c384d76c8dd242c5b00e1dbec5d62b35 : FAILURE in 18m 46s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/4d17d2dfa55045d68bf8ef0727f8663c : FAILURE in 17m 32s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/7c0897349cfb4a21ab072777840b84d6 : POST_FAILURE in 25m 21s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/ae6a944c3b0345ee89f9c19248964043 : FAILURE in 24m 42s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/3b9ee2a2920f43a6851a93af0bcbb29d : SUCCESS in 1h 31m 17s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/45c74479ee1f46cd970a0329f4b3362b : SUCCESS in 1h 00m 48s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/e339dc95dd69499ca6d9b33ac71bc30d : FAILURE in 1h 12m 03s","accounts_in_message":[],"_revision_number":5},{"id":"4397cb74cd04d6c8e22e14982821380e628f7b14","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-05-30 00:29:03.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9ea35b9a61acab88c44fd3fbb118caa2f82e66da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 02:15:50.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a36d31b576b43e6ba2e0613e837254d : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a781ba732e447fb4de3cb1aa522f83 : SUCCESS in 2m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/23c766e2331945448fdfb298704b5677 : FAILURE in 41m 39s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/bcab6ca0e5a94fe8a1db8355ed3362b7 : FAILURE in 41m 45s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/0c90f026a22e4002b25239e1cc4c022a : FAILURE in 32m 18s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/077416214c9c4b61a048aeadddd0f014 : SUCCESS in 1h 25m 51s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/0afbf02e4e434a29afc1a9f8723357e8 : FAILURE in 29m 31s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/02e03cbd48f947478032cde7e54431ce : FAILURE in 26m 32s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/5fd53947ec03444588c2e1eef25667bf : FAILURE in 26m 53s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/cc26a11dd7e34160b08c8142ae646ac2 : FAILURE in 25m 32s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/4e70cf1ff3bc44e0a194cf95693a238e : FAILURE in 27m 32s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/df473e42d2ce4d60a2e086b184e932fc : FAILURE in 26m 30s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/7cf7e1cf4cab461cb67794b1ec526fc5 : POST_FAILURE in 42m 20s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/173c8f2ee1a848089ad8a22053b8b30c : SUCCESS in 1h 00m 15s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/a22bf0f84bdd4205af831c1e4aeafc6b : SUCCESS in 1h 41m 36s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/06a12914d96e4a2a8082b6153dfd4e87 : SUCCESS in 1h 04m 01s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/bb50d15ce0a44680b5f5b4dbb44dec67 : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":5},{"id":"b90ecb1b7fc5c686e7d20435920130caf3ff2eba","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-06-03 09:59:55.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a225e823d90aa2ae91c019c1f91114fd0770e8e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-06-04 02:38:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a04bfe8020b8f5e22ad1b63c1213285e3212c770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 04:49:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5b37422873c4ae9aa846e5d79627c83 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68bd436e74ab4936b2b7edb4890a4dc1 : SUCCESS in 4m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/af54e6efc998477d9a299baeead374db : SUCCESS in 1h 21m 42s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/ab799860632a4670ba64548485a0b812 : SUCCESS in 1h 22m 43s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/b50cee71631e4a37b9c42c8e06951e3a : FAILURE in 24m 58s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c399fc36b41a4846999928233e8d7a36 : SUCCESS in 1h 09m 59s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/8eb0d8980d7347abaabe4bcc1e822663 : SUCCESS in 52m 54s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/a9a0825b6e664de0aeb8d566183208cd : SUCCESS in 51m 46s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/7244cedaa6bd451cac8102c4b3c323d4 : SUCCESS in 1h 13m 53s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a11f2320862e49d48b321212f0da0708 : SUCCESS in 1h 16m 40s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/4c27e958eb224d1b95e6d45bcb27e560 : SUCCESS in 2h 02m 16s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/8891c4892f164871bde1224453f97c9e : SUCCESS in 1h 59m 03s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/caa20974c36944f896ebbb6f9764bd70 : POST_FAILURE in 48m 49s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/b2e22106ff314c8b9d111f51357529ce : SUCCESS in 1h 23m 07s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/54bbf0c9ac4d44aba5e6af856a1e79ed : SUCCESS in 1h 22m 33s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/7dec2fa1795c4c268d03a76a634e8fac : SUCCESS in 1h 12m 43s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/654a6770c30a4f3aa9e5ef4b32b399ef : FAILURE in 1h 03m 59s","accounts_in_message":[],"_revision_number":6},{"id":"3f2cfb4e3822a5a101f4f968ca833a21775fbc8e","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-06-07 12:41:59.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0153d3c6885b11a731356f2b8831fe3ee8666ff6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-06-11 03:05:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d00255a4eb62a902065ab5a2221ca2cb54b0e220","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-06-11 03:09:27.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"2155df8be3a9d09a7f923d9a7114089b2941d892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-11 05:17:35.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71602d866e948db878563d5ff2e2b74 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ccf0ceb65e148c581e2afbc426cd0b8 : SUCCESS in 3m 18s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b16e6743849c4d03acd22d88f19b39d0 : SUCCESS in 1h 21m 05s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/7b92ba465cc6429b9e354890faf5204c : SUCCESS in 1h 20m 28s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/656d91ea8463403a884c48ebb16e5460 : FAILURE in 1h 29m 58s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/63ea7b3c95194ea1a5c102e784a8b084 : SUCCESS in 1h 13m 28s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/b18fe781540442b4ac2f2efe873e968e : SUCCESS in 1h 06m 57s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/cbada2b9c46c43cdaa29c92102883469 : SUCCESS in 1h 04m 21s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/fce98b1a09d444bab46766a6ac23ae36 : FAILURE in 1h 33m 50s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/6e0d0e15c6c640439a82ab830d344e84 : FAILURE in 1h 24m 52s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/24b3e866367f49fd9edfbbc54240eeb6 : SUCCESS in 2h 08m 49s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/f021e7f49ea941248d16c124696c32b6 : SUCCESS in 1h 47m 14s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/de9e42d5df7741eeb93cecc24e483f9e : POST_FAILURE in 46m 07s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/b170962334b444b4be233ab269ca10cb : FAILURE in 1h 04m 50s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/9be297e00db6428097a6bcbd7b1b9718 : SUCCESS in 1h 13m 01s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/1a45f6874e7746d287afc97c4f857e66 : SUCCESS in 1h 11m 36s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/18f2f8e97620445ab3875f55bdffc24e : FAILURE in 57m 22s","accounts_in_message":[],"_revision_number":7},{"id":"d36eb326b4cd57f56e1e511dc8e3ac81cf1ea0ed","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-06-14 02:12:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1705080d08522137ae19927395d208fe52d4151f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 04:15:02.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76a49285bd54b96912de9bc61a16da8 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43a283bfb6084480a9747485ae44f93f : SUCCESS in 3m 21s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/92ce02ab4ae144a08a0b9f9dc594709b : SUCCESS in 1h 19m 38s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/a7f2686510554f199b40f1f58252c5cf : SUCCESS in 1h 23m 35s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/980d9643a63a41049b1fc84fcd179e9a : SUCCESS in 1h 03m 44s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/7f80d9049343484780348fa04c1efd5a : SUCCESS in 1h 23m 26s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/8d3ff0a657a14acda84e24774101dab4 : SUCCESS in 52m 40s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/7702716a5c134b45aef9c6e50d3dae4e : SUCCESS in 50m 51s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/8c892945561d47a1863f031cb420900a : SUCCESS in 1h 07m 59s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/57f9ce1c3f0e4f4886ff598d38e564e2 : SUCCESS in 1h 04m 44s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/b1df6cd0d7654535b13eac2129fcc235 : FAILURE in 1h 38m 17s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/591d500df0c64005bdb1d2421a0f5bf6 : SUCCESS in 1h 57m 49s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/e1357e656657423d8cc591b02e17afb5 : POST_FAILURE in 1h 02m 04s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/0895aab9faf64e14ac85694fa2436a42 : FAILURE in 1h 10m 30s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/09dc46f95724444b9d3de1be186d0dac : SUCCESS in 1h 35m 01s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/ab94607674394b14b73b201dbc89474c : SUCCESS in 1h 11m 07s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/5fa865ed747d4e09bb6e7e1af6b41e5d : FAILURE in 1h 11m 04s","accounts_in_message":[],"_revision_number":7},{"id":"6c704e0c4b1a031cb21dbf0c3986a411ef6a0ca7","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-06-14 05:43:22.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"43513096839192e36ddf7c70588707b2c356d078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 08:25:00.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0247a9bf3c374c08b3f41565c1506f8a : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99396baca6744fd594e4aade5c20ac39 : SUCCESS in 3m 20s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/7b4d7cd8d4c649f69df17bc3b8a3c4e9 : SUCCESS in 1h 25m 42s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/e10475ae55f84a23973f27a59508bf63 : SUCCESS in 1h 20m 25s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/b96912562490475fbd155d7648bcf01f : SUCCESS in 1h 16m 03s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/2cd2f81d8d8d4f5cba781c4ddf455afc : SUCCESS in 1h 29m 40s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/735828023bf74332b89be4395da8b773 : SUCCESS in 53m 27s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/9f2e4b6fb4244985a1a9f0d45cdb71cf : SUCCESS in 1h 09m 35s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/2777bd4af21a476bac01a2edbc90cdfa : SUCCESS in 1h 05m 00s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a3084322ef264b12a314b9cb3e8c215e : SUCCESS in 1h 27m 09s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/365528a606e24662ac707e78935e6d15 : FAILURE in 2h 02m 33s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/46094788e772447dac119139d4375d0c : SUCCESS in 1h 14m 36s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/dfe3ad078bf94983afe0ca9766461df3 : POST_FAILURE in 54m 06s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/b2bc31d3d90744539a0ed287f3a87e75 : SUCCESS in 2h 36m 31s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/74576a2ea750444d8c74de21eb3f903c : SUCCESS in 1h 34m 10s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/c090eeb37d87437e9b16384e670a1892 : SUCCESS in 1h 35m 10s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/dd3c53cf891245cdbccd8c5bb35aa7fd : FAILURE in 1h 03m 17s","accounts_in_message":[],"_revision_number":7},{"id":"8e71c0d98edb0755c3ab13e8c1bcfa28e0596e17","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-06-20 08:47:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"90549d9dcac42ca267629d6358c77f5d3def2b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 10:14:54.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288989674c4e412681b99d428fdd4727 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b5a478aadae4f809e60a582aff406ea : SUCCESS in 3m 41s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/36700cc84a9a4affbab1fc3c650649d5 : SUCCESS in 1h 18m 43s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/dd30f102953441e78bbe91079062551d : SUCCESS in 1h 08m 54s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/5e5b432efa2342aab7a0192f3d99b698 : SUCCESS in 1h 05m 00s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/6833f3dd78f045dda9772172c1589293 : SUCCESS in 1h 11m 57s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/787561f6c0b5434a9b512301164d0d66 : SUCCESS in 1h 06m 33s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/cba0589a71814deba636f9bab225f786 : SUCCESS in 59m 18s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/9b9eecbd0aa04c93a8676010f3c6e3ca : SUCCESS in 54m 00s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/eb303e4ee7cb4349b4931bd99fcb1360 : SUCCESS in 44m 05s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/20c121bf94444331833f3d9411216702 : FAILURE in 56m 17s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/b67b7be3c0324c1b96466887e829c77a : SUCCESS in 1h 02m 59s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/8b4b5e13e13b4513bd3ede13177b0649 : POST_FAILURE in 45m 33s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/7abdb648d3ec45bc86f882dfcc53a610 : SUCCESS in 1h 24m 28s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/b8b29fe2e8e7442586b6004fd45babb7 : SUCCESS in 1h 06m 15s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/b3b9aec118eb45acaad74b6a55fdc0d5 : SUCCESS in 55m 43s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/591d5f6b2a4a4f70b6aa980fc3c655ff : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":7},{"id":"288e92e0bdf40d9b12ca0dabd2948b15ab090230","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-06-27 11:18:10.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6609224b2c74914d918a8b29c113e7687deeb0fd","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-07-07 06:23:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"90e1251510dd28d00877cc79472cfea0fd37acb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-07-07 06:23:24.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"6b40acfe17267b8e76d7c37d966ae637da01da9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 09:50:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73d2e1598ad342d194e9170ad4ae425d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824d57f5a4da49cd8c40ff4c1775b7f0 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee6ec9b765254d18946d2e5b4e84f374 : SUCCESS in 2m 35s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/db41aaf0cf3d49278fc5c0d1a78dabd1 : SUCCESS in 1h 10m 35s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/c66c7395242640ad93e15d212ac4d820 : SUCCESS in 1h 12m 40s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/d20b1cfed0834a6f8a4c025894223ce0 : SUCCESS in 55m 14s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/34764782e2464fe780e8e52314ac3267 : SUCCESS in 1h 10m 05s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/212d42aae72d4f33ad832bfb3e038928 : FAILURE in 21m 39s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/371614e0b69343a7b79e8980df488b6a : SUCCESS in 51m 53s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/42f1d93a449245efb39bc5cdb9cb3dcf : SUCCESS in 53m 39s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/236cb799dad44d1895da9882211df35f : SUCCESS in 55m 45s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/795a188934bc4cb79dfaf4c67efb275c : FAILURE in 17m 39s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/cebd9ffb6dae4d068eff0316229974ac : SUCCESS in 59m 46s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/497ae12262684c5ebc03e41b7ac28bbd : POST_FAILURE in 18m 58s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/4f82c8628ec643aabf52f80e3ee8c243 : TIMED_OUT in 3h 05m 02s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/a49b8d42b7e6451999ebc0d4a46ca9e5 : SUCCESS in 1h 11m 38s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/887ae1b9c48d463e998b6daf2cd60aeb : SUCCESS in 1h 02m 57s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/c537f450bf314b3c864a82e59cab54a8 : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":8},{"id":"afc724e81b63be3124fa266510b6320503f80cf5","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-07-07 10:21:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"57ab550692dfc8b5f9aa285d290ba72956e2fb3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 12:10:18.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/264197458da94fd080e6c3a1fd23f8fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/459739363388402baf2d3ea27e5ee8c0 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e83d3e7013284002962d47f2100ffd01 : SUCCESS in 2m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/286cdc324f4b40b59d206d84c68e5571 : SUCCESS in 1h 44m 25s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/1c8dd03f24e34299984c945fa87b4282 : SUCCESS in 1h 11m 31s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/5347b0b4fe654731bb17e81fbe1a0d93 : SUCCESS in 1h 02m 51s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/8be35eacbf2a44ef81318d0f1ccfc59a : SUCCESS in 1h 13m 00s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/fb320ee561224fac9c7e7bd506f86d90 : SUCCESS in 1h 12m 41s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/7cc5ebea3dbe447d93a2dfcf7b2e0b8e : SUCCESS in 55m 02s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/861dc6f920114534b4f87ef9b5ec28f6 : SUCCESS in 1h 01m 56s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a683f1f399824b29b448d3446988f244 : SUCCESS in 40m 25s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/5ee4f591974446daa322baf6d535e61e : SUCCESS in 1h 19m 25s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/af77e023aeda485fa05b6a014e060bf0 : SUCCESS in 1h 06m 26s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/4f1b26ffb8924952a5aa1ba93a2168d4 : POST_FAILURE in 54m 48s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/265690861112492686ba4d60bada786e : FAILURE in 1h 09m 18s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/d76d89bcef334a53ba3a032d0c131b6b : SUCCESS in 1h 20m 57s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/9dd43bf3dbbe40598e88ad7005823b46 : SUCCESS in 1h 03m 19s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/65f8643fa8ba45c7b1638bc5f5ed976d : SUCCESS in 1h 08m 44s","accounts_in_message":[],"_revision_number":8},{"id":"f8324ed195db01d709d7a727556ba0d766041c34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-07-08 01:43:54.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"50252fff132b4bf03975c96a134918ed0f4730d3","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-07-08 01:44:50.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bc2fd2b265fd9dc4c223fc4040f5825f4793baaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 04:54:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bcc9ddf565b429b9db4f7409feb6f92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a97044a4cbe54bfeae7c015c062df281 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80401f6a835e471f85920588cfaa8885 : SUCCESS in 2m 37s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/36873a12a56f4b6eab393b58a4368dd7 : SUCCESS in 1h 17m 47s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/1b84eb22d7244e57a8d1bbc5a9290ab1 : SUCCESS in 1h 16m 09s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/3880e126b9e74078b692e7672b8944e2 : SUCCESS in 59m 37s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/541b959b3b444693b60627d2e3e9db26 : SUCCESS in 1h 05m 43s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/f6dfa9a2581a4feeacd35bb4cec56a3d : SUCCESS in 54m 26s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/4f74ced4e08f4bc59b88241af996338c : FAILURE in 25m 34s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/161fd0d789324774839928295c70c256 : SUCCESS in 49m 54s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/1aed64b7c17046f884dbafb6436bd8e4 : SUCCESS in 41m 53s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/342957927f0d436db3ba760e6358783f : SUCCESS in 58m 33s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/972d141de8fd43db8b735e7a4d326426 : SUCCESS in 1h 03m 52s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/3669d04844c44f4d8cc0f26a96c62b60 : POST_FAILURE in 49m 10s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/23d4a2589815471c8d39ff5c7d528742 : TIMED_OUT in 3h 04m 47s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/944286226d0b42d8a70a75ab921e7099 : SUCCESS in 1h 20m 12s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/274e7b330ff14d7380a95426db9eca5f : SUCCESS in 1h 02m 06s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/9faa1eb2d5444683b452074893855ae2 : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":9},{"id":"1cb8985b17563e88718270b3e1dcc23826f002bb","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-07-08 05:40:07.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d8ee36f2ef622a4474acfb390a1bdfe8ac63839b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 08:51:13.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e82e500609142cd919a666e53b6513e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61423edb5dd3495a9c6a635f24867041 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/122f50bba5e24359843516b2a0ba4a1a : SUCCESS in 3m 37s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d1674001d116426bb1c16d2f4a2f0f30 : SUCCESS in 1h 41m 00s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/e261b132e9ef47d2ade9bbcc75102277 : SUCCESS in 1h 41m 17s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/09989bcb18ed4c39a65a964c7cce53fd : FAILURE in 59m 25s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/8890296086e746aea2338e247fe66a41 : SUCCESS in 1h 30m 39s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/23335cdeaa644bbe94128ea447401ad8 : SUCCESS in 1h 00m 56s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/412c8afb9586431289fd0b7ab5331315 : SUCCESS in 48m 15s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/922806abbd9b41df846e9cf9c7f5d05d : SUCCESS in 46m 38s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/48506cc10f33493cb8fe32c0e8218a97 : SUCCESS in 47m 09s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/27549185058d46c4b33b48e83498a7f2 : SUCCESS in 1h 11m 51s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/fb719cdaecd14103806a47527f14875b : SUCCESS in 1h 03m 21s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/20f831d7ed0d416aa96e5e5611609cdb : POST_FAILURE in 1h 01m 20s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/67d0afb07c02465aaf55209dae9a7b63 : TIMED_OUT in 3h 04m 10s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/e3161744f02545ce8554f5c40ae55113 : SUCCESS in 1h 29m 37s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/48a55b23e79f448b9f39b8ab5ed668fe : SUCCESS in 1h 11m 51s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/d49026b79daf4abd904be50d3cebcb84 : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":9},{"id":"bbe3019a4a26af88206e029f7a5516d752f8cbe7","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-07-11 08:48:45.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fee78ebed6f39ac9ab2a7cdf46c3a2661c3823c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 10:18:46.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09ba761d9df54942a7c30ac73e661598\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c674b88112494e01acb51ef6dad1b262 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ca5ec5c92294296982675e23bef9ea9 : SUCCESS in 3m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0a68b197ef324b5ab770a50b2d65228d : SUCCESS in 1h 29m 04s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/3e3af54a7ffb4fa38227bb57e13da6b3 : SUCCESS in 1h 14m 00s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/46b96d89b9804ff199787bf26abf82b5 : SUCCESS in 56m 14s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c9878de81f644b25a13e293c6e5e447d : SUCCESS in 1h 21m 03s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/be8f642b6d9b4234bcd7acdd91c10e6f : SUCCESS in 58m 05s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/8267bd6f9bf443f5a1585a48b6933d1b : SUCCESS in 49m 16s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/f1752d079e8f4c0ba40ae532fb77b70c : SUCCESS in 51m 46s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/54e9b3eb82734d0191ecb1b37e3cb64e : SUCCESS in 50m 16s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/cc7b663a4ad54480963c3bcab0535ad8 : SUCCESS in 1h 26m 04s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/10c9d8bfc2a64e68ad7c274811c39975 : SUCCESS in 1h 04m 56s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/d03bb097f05841508fa48be921ad412d : POST_FAILURE in 55m 32s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/fa1abe42e1454ccfa918e196eddb8cca : FAILURE in 1h 07m 23s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/bb69dab86e2b42c6afdc264aff2288f9 : SUCCESS in 1h 16m 52s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/6ab5bf7a00484585b789348f89aec966 : SUCCESS in 1h 03m 49s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/e7bdfb3ca3df452c8f8c7930497682a2 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":9},{"id":"d9b0b39ffa8642e4a2cb2440afc0d4b6db73b3e1","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-07-12 00:43:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"83276ba659dc6dab213d2a031aa134b6329cfeef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 03:52:17.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d73e6e7b81f64d6aa46e191a4c9b120c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e9215d4e0246c885b049210021116a : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5871e5d983fa42a4b171813442b85a8a : SUCCESS in 3m 17s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/459f7a42b0aa4f25abb7e0ad220be507 : SUCCESS in 1h 18m 52s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/bde4304432f446229aa4a201a7620e77 : SUCCESS in 1h 26m 13s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/e358d0b968064891baa79b446fb5deb9 : FAILURE in 18m 36s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/df5be4f3b97144a6a0536ecd7cab72f8 : FAILURE in 18m 34s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/05cd987a824446969496609c2e601ee8 : SUCCESS in 59m 51s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/9ff5907209644ea086f20354d88914c6 : SUCCESS in 55m 38s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/b5fdce4d24b14a5196ecd245acca9c16 : SUCCESS in 54m 53s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/79271216370442128f94a38833944219 : SUCCESS in 53m 39s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/5e80e4c149f64a13962694c71dd084ee : FAILURE in 1h 07m 40s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/5dcd37e5a95f4348b6ce11f61b26c04c : FAILURE in 1h 09m 08s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/dff0897b2ebd4782b71fe36b2ecb62ea : POST_FAILURE in 26m 18s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/cb8a2692178f4d96b46c01a464958640 : TIMED_OUT in 3h 07m 06s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/ba16fc30a1474622abf8591e005e291d : SUCCESS in 1h 48m 21s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/e27ffcda93f34cad802f6b72cbc1f3b5 : SUCCESS in 51m 58s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/e433c93f25ef4331be76d1848136958b : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":9},{"id":"52d2e071b92790b1000a47770356fb66e2aab029","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-07-21 12:04:26.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e984354714038e6fdd44e8699aad15559728b5c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 14:02:05.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24e72f20821f426b8db3a95a37681c3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c774b5c9ebd4282a5213f05331c9127 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30cfebf24a4a40b28234a733a7a97b06 : SUCCESS in 3m 32s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6e8b1667354f40119a5bceeb7073f5fe : SUCCESS in 1h 34m 39s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/216113ff759c41f5b3699b4341750c85 : SUCCESS in 1h 49m 54s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/721cbf9741564c1ebb8ba493fcbae85e : SUCCESS in 57m 17s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/65f02fdd49f64f31b55edc3f2fa25616 : SUCCESS in 1h 22m 43s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/10b85104e5874e82a140c4387752507e : SUCCESS in 1h 01m 42s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/fd225f7e0300418b9906486fd72cffc9 : SUCCESS in 58m 33s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a6dd50c5c83440739c6638a3b5908630 : SUCCESS in 56m 50s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/08a38f2442194b978b937dde00b56dc1 : SUCCESS in 55m 38s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/bdaaaf45bf3a451597c1531b29c8da36 : SUCCESS in 1h 07m 50s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/76f89c8619d345f4bbbb53a10fbb3d1c : SUCCESS in 1h 07m 18s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/8d55b8a81f994a8290e4dadd19d90643 : POST_FAILURE in 1h 08m 59s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/2ba7488c70ce4feb95ba054e6d4b654e : SUCCESS in 1h 36m 17s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/771082a62918499cbd7b8c49f8ad3cdb : FAILURE in 22m 36s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/7865aaae874c4fdcb489790575611d12 : FAILURE in 23m 28s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/dffd59a2397848989d6ab7abf9a807f8 : SUCCESS in 1h 10m 05s","accounts_in_message":[],"_revision_number":9},{"id":"d710a13f4b1b044d93926aeecca31e0085c11207","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-07-22 15:11:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"88d67b4c29d7259f4767ba898fe6839ede71676b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 17:26:05.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/625074f5e1be490f85d1e237dc8cd378\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b75dbaf83474c81a18b554478dcb92d : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b140bfa14b99487bac97e8c47e2723c5 : SUCCESS in 3m 24s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/4a281fc28989438e992203e9f86aac82 : SUCCESS in 1h 29m 35s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/bb968bd856244072812ee51dc6c966fd : SUCCESS in 1h 34m 27s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/b690cb3e77cf4aea850040aabbf00d97 : SUCCESS in 59m 29s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/33d72d5b67164b8a8382f7aef4e36e05 : SUCCESS in 1h 16m 00s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/3bd47536ae8749f7888caab48de31c1e : SUCCESS in 1h 10m 28s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/9ca01c58f3a7477ebd3bef6bf3bad427 : SUCCESS in 54m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/96c9878ac2e34a82899f929101482e8f : SUCCESS in 1h 16m 21s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/e4f44fe7d60e48e1a6f4d182cd032cdd : SUCCESS in 52m 20s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/1d81d626420f441281c87b029372be60 : SUCCESS in 1h 23m 55s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/79e1671d333848d8a1aa8ebc045791c4 : SUCCESS in 1h 16m 06s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/c543bf2be4324bc38fbdb60577fe3dd0 : POST_FAILURE in 57m 02s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/a8c79abbbe694af2b5140d3974bd573d : FAILURE in 2h 09m 09s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/2e33c87ca8ab49d094de36533987858c : SUCCESS in 1h 48m 00s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/5e01bde4fb94433aa3807d622972debc : SUCCESS in 1h 13m 58s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/2d1f3c549050453da751a8c95f7a13b8 : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":9},{"id":"ed27397f51a04030e38203d51faf263421ecb7bb","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-08-04 11:00:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"324335c995ba9be72d4c137b5193656cc9f0c69e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 12:44:08.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c90a2ce695341289ad6a26b9e7f189c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2700cb67c2614e329f39a01f2bedc85f : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f89cf4fc43fb4444ac40f406f0f2e694 : SUCCESS in 3m 44s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c3f828cdb4e8432987bf5bcd7c2ffa0a : FAILURE in 28m 13s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/4c4da96d9a764f639e71a30beaf4b46f : SUCCESS in 1h 24m 57s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/4a6b19383b37485cac78762d625a8e5a : FAILURE in 21m 47s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/9241f1fd6abd4dfb8bb213c6d25c1cad : FAILURE in 27m 12s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/cf8eb0889e7949dc9490f11143fa5e8e : SUCCESS in 1h 11m 12s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/1e2118d117864b47b19cd65942047fed : SUCCESS in 57m 25s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/6ae86fdbc9b4449b86b82ea9585c41fc : SUCCESS in 1h 00m 02s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/05d1a761b33440c1a60b5711ea20de52 : SUCCESS in 55m 34s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/9d1e7334d945463089f61521f3076ada : SUCCESS in 1h 01m 44s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/2d09154c2b2d421eab386410af2fad19 : FAILURE in 20m 18s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/b037754bc68547b5a9715462a6c8ca7d : POST_FAILURE in 25m 46s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/49bb13ec5035486382ceb1f7c241a814 : SUCCESS in 1h 33m 01s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/877d0775ce2b4e43a08008d032880b35 : SUCCESS in 1h 41m 21s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/5e8d207d7f334414a58b5907985a5dad : SUCCESS in 1h 03m 35s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/f5be2c89f1ce419a95ca5dc2395b1ff8 : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":9},{"id":"394aa4477c53b374c530a4dc1e1242befca134ea","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-08-06 00:54:45.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"394306529e44b414da8cff4412af7341af2411d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 03:52:00.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63bdfc1c052d4db6b749ec1580244a89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f22cc761767436f853cbbcd2a017e27 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ad6e88080f453894ac6ad672ca419d : SUCCESS in 6m 59s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/406854e330c84c83949baa5e13e5d2cf : SUCCESS in 1h 24m 08s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/efe11c2ceeed4b17b7229ac38fc3992a : SUCCESS in 1h 14m 40s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/12cc3445458842f4b7669268b8a1cb0a : FAILURE in 1h 06m 18s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/d02666814a2e4290aeaa7f4cbfa17267 : SUCCESS in 1h 02m 52s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/c65e82daab8c4892bb04c571874f1fcf : SUCCESS in 1h 05m 48s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/0c8fae4b1eb844f494df373c3c3aa4a1 : SUCCESS in 52m 39s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/1a17c61e92fc48e98b9a43e1e75da5c5 : SUCCESS in 52m 25s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/53f3a1e1c70047f9aec46879afc2c399 : SUCCESS in 51m 30s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/36e48ad0f1c142fd8e89cd5e435be63f : SUCCESS in 1h 08m 44s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/ad8992bf607f488da25b064015e4db0d : SUCCESS in 1h 04m 10s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/9e611d18042248a48b2fd2ae8b281bfc : POST_FAILURE in 38m 24s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/cadd4e3dea104f13adc237cf5c7248a2 : FAILURE in 2h 55m 47s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/013af1d5f14e4700ae2432ee5241c0c0 : SUCCESS in 1h 36m 10s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/7c8073dc556940d8a3ee9c8567e18aef : SUCCESS in 57m 28s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/98230d5e6b414067af01010ff9ee3d49 : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":9},{"id":"8d5d32d35a830494ae74ba0aa3ef2806ff2d5068","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-08-06 09:52:21.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"33548b92240dc3a6ef871154fef0d2dd9e81385a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 12:50:29.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/375b1d52243f4bc089276e3c4d271ed8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766070c71f0a423fa277852fb9ee73cd : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34012589e2f44b63aec806a57ea8606a : SUCCESS in 2m 30s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5c14f53a476a4cc3b483405709eef50c : SUCCESS in 1h 24m 48s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/42a486b79e7240599eee9560ec42ed3b : SUCCESS in 1h 36m 04s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/b54d82501dd84d20a2857efdfc56cb71 : SUCCESS in 1h 15m 44s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/44428042346b42e3abad149882eca2e3 : SUCCESS in 1h 31m 03s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/c6e505a380b94a7dbb1377af18993d25 : SUCCESS in 1h 27m 50s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/b8c5e428c5b64c20bca0a710eff3e9ab : SUCCESS in 1h 14m 06s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/a3c90ad64a544db4a6ca28ab5bdf8563 : SUCCESS in 1h 09m 47s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/6edd57e9ed4c499fa3ec31c7dc252bc9 : SUCCESS in 1h 05m 26s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/0e8eeea420634c579538522e862c5e9a : FAILURE in 1h 28m 00s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/807c3788c6334874be3a63cbf5569c3a : FAILURE in 1h 21m 22s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/bae488cdff3844f2b509107988f456d4 : POST_FAILURE in 22m 51s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/19df6f7c5f5c4178bab76d96cab31ac9 : FAILURE in 2h 55m 21s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/46aec8b3b65d4cd59f6a931b501314d9 : SUCCESS in 1h 34m 49s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/14f96d43a9844ff09f6409bf405fe9ba : SUCCESS in 1h 12m 51s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/0adf0ba26ea544eba483877e5a36f9df : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":9},{"id":"4119603c74608ef44c39424d5b3a31bba2133b26","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-08-11 14:56:25.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4abcb1ee491968baec0bab2200bc8bc1a2a54391","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-08-16 13:30:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"86364f55e8c5d6e2361dc634e1183c7f473d624c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-08-16 14:46:29.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"5ff45ef979b5b73ca48565c67d58d30f3092870a","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-08-16 14:47:34.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2fcdb9eb4aa342f18ea101b321013ad788147d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:36:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d73ef765a328406bac3a36f4064bfce6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3146425a30b043bd9e4c3c1e4db8d15c : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759d67efefa94c5badaff1d03002220c : SUCCESS in 6m 04s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9e8f922d103040deb5a1ae2f2b0dd939 : SUCCESS in 1h 38m 29s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/e755c2c795f34895962a5d4adbb80aa5 : SUCCESS in 1h 24m 12s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/77052014c9ee4878b9c74028fcfaffdd : SUCCESS in 1h 17m 30s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/cd94300a24b94e709a3ce71fa6bf8406 : SUCCESS in 1h 34m 51s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/94e01fd238674caebd1d849edfbd62b5 : SUCCESS in 1h 28m 45s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/e3a182f941c443058cb19568646229e9 : SUCCESS in 1h 15m 11s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/39729668c11f489e85b323a883a3c459 : SUCCESS in 1h 12m 38s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/04fb9f5eed714eafb2982f7aca713e8c : SUCCESS in 1h 12m 51s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/e013fed7b8ef41159bc301201454d30f : FAILURE in 1h 46m 49s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/554b2b31751443bcb829a8030960e8d0 : SUCCESS in 1h 31m 12s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/6e6ac99d18e643b784712bdad5a19155 : POST_FAILURE in 1h 00m 22s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/6a41c94589a94d9fb1b63a77502603af : FAILURE in 1h 19m 33s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/9611939762a947c5b6b4d37a9b950799 : SUCCESS in 1h 41m 40s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/cbf42b02192f4436886c5722a3c548fe : SUCCESS in 1h 20m 12s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/4c3f658eadf046a6a000cadf5200585b : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":10},{"id":"a265b79741e29adc43327af94dd3e83e549a980d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-08-16 19:14:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1a8378b2d45cdedd078edd379f42d2990bd6765a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 21:12:41.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b2cd5cdf58a4d1c818732cdb6364d45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9efbcb62abf49ba9bb8e0273e5c6478 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788e4d96601d4e738a40b95c6259213d : SUCCESS in 3m 32s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/b9db88c4204543f8aec9f66a38fcb888 : SUCCESS in 1h 34m 53s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/68900d9b895e41e28ef296b7b900bd3c : SUCCESS in 1h 30m 39s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/96e4b0ec032042269cd1a4066b2ce6fa : SUCCESS in 1h 00m 01s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/1d9d1b3cc647459fb60cce003c78d06d : SUCCESS in 1h 38m 56s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/2c4bd3b72e9d473d9e5f0e57d0794177 : SUCCESS in 1h 01m 40s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/65102ceabd344afe9cfa7960231441eb : SUCCESS in 55m 04s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/894ea876672d40eb9f1c685ef656cd36 : SUCCESS in 57m 27s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/d116439ed8db4081a39d8e22baa37be5 : SUCCESS in 47m 24s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/d3636baf44a84115b763b1a5143db9f9 : SUCCESS in 1h 10m 00s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/bac1c631e09141d09891cec9f0c5027a : SUCCESS in 1h 00m 35s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/74077954a14043f8911e5f8bdab5567f : POST_FAILURE in 57m 23s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/2a228d33348542f8b26e522f94813f66 : SUCCESS in 1h 33m 08s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/6e291580fe7c41e9a8ce19241369155a : SUCCESS in 1h 50m 45s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/ae492413c79843df945db5327c4a4ce8 : SUCCESS in 1h 22m 03s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/9208d67a6a3c40418f8ebaef042282a0 : SUCCESS in 1h 22m 07s","accounts_in_message":[],"_revision_number":10},{"id":"d883b18b8a8e7489f34abfc0f10ac1586cbe75d6","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-08-17 07:38:00.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e901d6854dc4ac3ca5094233e38ce133af858236","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2022-08-18 01:24:20.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"95f5fc2729baf2d2b34e5c94a8974d8d72d76b5e","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-08-18 07:50:28.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"198c2b53fd55a01ea8b1b117f345263fa3ae7834","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-08-18 08:18:53.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"dbfbacb5a9affac2a67da15b9244ad0c5f5d5432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 08:19:33.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"b6ed7f6ecbe005da980c2cd0560f486ae841126f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 09:58:02.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5977f6a214a4745bae4e0596e82391b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ff965a5459422bb5e05253b07f2c49 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bb1ee09e4a5404a84824926ff42276e : SUCCESS in 3m 49s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/bf071725d4124ccaa5e069e724bcafa8 : SUCCESS in 1h 34m 04s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/9c562dc07c864e17a3a006bb9eff09fc : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":10},{"id":"aa48b7eaf9393520ebcae29d4eaa60178d8e4c97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 09:58:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"20a1a24e418e43fe6ad44b6dea3b26c56a36fa75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 09:59:20.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9677eea7c8694d2885c10fee9165d52c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a26483121ae4644be5f39fd310ab2f9 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"4b0fceea709fe9fb23764174444b88087d262490","revisions":{"8eddd46d24b85b2be098ebef43fea19009476b2d":{"kind":"REWORK","_number":1,"created":"2022-04-14 02:05:55.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/12/837812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/1"}}},"commit":{"parents":[{"commit":"79917999e95b19b452d2d7839278fa24c62fe683","subject":"Merge \"Limit number of K8s API requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/79917999e95b19b452d2d7839278fa24c62fe683"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 02:05:48.000000000","tz":480},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8eddd46d24b85b2be098ebef43fea19009476b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8eddd46d24b85b2be098ebef43fea19009476b2d"}]},"branch":"refs/heads/master"},"44a7a18c64d33de6bd43b02837ff0741c71d7715":{"kind":"REWORK","_number":2,"created":"2022-04-14 02:42:25.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/12/837812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/2"}}},"commit":{"parents":[{"commit":"65f6cef16b2f3ca8eed2ba59800096dbad3d53e3","subject":"Merge \"Check that pod is deleted in delete_pod\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/65f6cef16b2f3ca8eed2ba59800096dbad3d53e3"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 02:42:14.000000000","tz":480},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/44a7a18c64d33de6bd43b02837ff0741c71d7715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/44a7a18c64d33de6bd43b02837ff0741c71d7715"}]},"branch":"refs/heads/master"},"24c7be13acbb7c5cdf241b1467b8b3e4ec82e075":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-29 06:07:38.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/12/837812/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/3"}}},"commit":{"parents":[{"commit":"962594cc7b3f27362d8fc7b4307ee8720e8e3381","subject":"Merge \"Adapt to new OS resource names.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/962594cc7b3f27362d8fc7b4307ee8720e8e3381"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-29 06:07:38.000000000","tz":0},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/24c7be13acbb7c5cdf241b1467b8b3e4ec82e075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/24c7be13acbb7c5cdf241b1467b8b3e4ec82e075"}]},"branch":"refs/heads/master","description":"Rebase"},"8461d6e1052c49e33e833d118ba3eb6a4d7b49b2":{"kind":"REWORK","_number":4,"created":"2022-05-21 08:46:42.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/12/837812/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/4"}}},"commit":{"parents":[{"commit":"61bfb58898815a37657a29d6a30809ba1411d09b","subject":"Fix the crd_lb_ip not being defined.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/61bfb58898815a37657a29d6a30809ba1411d09b"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-05-21 08:45:40.000000000","tz":480},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8461d6e1052c49e33e833d118ba3eb6a4d7b49b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/8461d6e1052c49e33e833d118ba3eb6a4d7b49b2"}]},"branch":"refs/heads/master"},"2ae4e0ee67c27903d04dee3be9310e698da553f2":{"kind":"REWORK","_number":5,"created":"2022-05-21 08:51:01.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/12/837812/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/5"}}},"commit":{"parents":[{"commit":"61bfb58898815a37657a29d6a30809ba1411d09b","subject":"Fix the crd_lb_ip not being defined.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/61bfb58898815a37657a29d6a30809ba1411d09b"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-05-21 08:50:28.000000000","tz":480},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ae4e0ee67c27903d04dee3be9310e698da553f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2ae4e0ee67c27903d04dee3be9310e698da553f2"}]},"branch":"refs/heads/master"},"d6ca30d2ab452ed65f940a9cdf2b503b1dbf707e":{"kind":"REWORK","_number":6,"created":"2022-06-04 02:38:12.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/12/837812/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/6"}}},"commit":{"parents":[{"commit":"61bfb58898815a37657a29d6a30809ba1411d09b","subject":"Fix the crd_lb_ip not being defined.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/61bfb58898815a37657a29d6a30809ba1411d09b"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-06-04 02:37:56.000000000","tz":480},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d6ca30d2ab452ed65f940a9cdf2b503b1dbf707e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d6ca30d2ab452ed65f940a9cdf2b503b1dbf707e"}]},"branch":"refs/heads/master"},"e69ba5501d8b2fbd89239f8b9a56073024b76e17":{"kind":"REWORK","_number":7,"created":"2022-06-11 03:05:48.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/12/837812/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/7"}}},"commit":{"parents":[{"commit":"61bfb58898815a37657a29d6a30809ba1411d09b","subject":"Fix the crd_lb_ip not being defined.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/61bfb58898815a37657a29d6a30809ba1411d09b"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-06-11 03:05:37.000000000","tz":480},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e69ba5501d8b2fbd89239f8b9a56073024b76e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/e69ba5501d8b2fbd89239f8b9a56073024b76e17"}]},"branch":"refs/heads/master"},"843062328821927d1b662bc1c82ba7146d4e6c85":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-07-07 06:23:24.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/12/837812/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/8"}}},"commit":{"parents":[{"commit":"d0840287fa0bd756bc65959015d97d36cb67d44c","subject":"Merge \"Change naming (again) for networks and subnets.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d0840287fa0bd756bc65959015d97d36cb67d44c"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-07-07 06:23:24.000000000","tz":0},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/843062328821927d1b662bc1c82ba7146d4e6c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/843062328821927d1b662bc1c82ba7146d4e6c85"}]},"branch":"refs/heads/master","description":"Rebase"},"a0f138e761fe78bf24ca76fdfbd84544b1309b42":{"kind":"REWORK","_number":9,"created":"2022-07-08 01:43:54.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/12/837812/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/9"}}},"commit":{"parents":[{"commit":"d0840287fa0bd756bc65959015d97d36cb67d44c","subject":"Merge \"Change naming (again) for networks and subnets.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d0840287fa0bd756bc65959015d97d36cb67d44c"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-07-08 01:43:37.000000000","tz":480},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a0f138e761fe78bf24ca76fdfbd84544b1309b42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/a0f138e761fe78bf24ca76fdfbd84544b1309b42"}]},"branch":"refs/heads/master"},"4b0fceea709fe9fb23764174444b88087d262490":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-08-16 14:46:29.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/12/837812/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/12/837812/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/12/837812/10"}}},"commit":{"parents":[{"commit":"d0840287fa0bd756bc65959015d97d36cb67d44c","subject":"Merge \"Change naming (again) for networks and subnets.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d0840287fa0bd756bc65959015d97d36cb67d44c"}]}],"author":{"name":"yangjianfeng","email":"yjf1970231893@gmail.com","date":"2022-04-14 01:59:26.000000000","tz":480},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-08-16 14:46:08.000000000","tz":120},"subject":"Add annotation project driver scenario test cases","message":"Add annotation project driver scenario test cases\n\nDepends-on: https://review.opendev.org/832768\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/853261\nImplements: blueprint specify-project-by-annotation\nChange-Id: Ia4591ef96ce642f53ead6147ff4b56960746e510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4b0fceea709fe9fb23764174444b88087d262490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/4b0fceea709fe9fb23764174444b88087d262490"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}}]}],"submit_requirements":[]}
