)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~838093","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Id7434270af5fc654fc9bff9e0105e2408db392f9","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"ussuri-em","attention_set":{},"removed_from_attention_set":{"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2022-04-18 13:58:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7434270af5fc654fc9bff9e0105e2408db392f9","subject":"Update \u0027Maintained\u0027 stable branch jobs on master gate","status":"MERGED","created":"2022-04-15 07:18:23.000000000","updated":"2022-04-18 14:00:43.000000000","submitted":"2022-04-18 13:58:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838093-ussuri-em","meta_rev_id":"94c09e1349b7dc266ea7e6ed174f94656fb033a0","_number":838093,"virtual_id_number":838093,"owner":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-18 13:58:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":2,"date":"2022-04-18 12:26:10.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":1,"date":"2022-04-18 12:26:10.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-15 10:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-18 12:26:10.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"}],"messages":[{"id":"b190ddc0c22be48f9e0a324ef26a1c9d8d560fb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2022-04-15 07:18:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"694467eb034eb1bda273889bc01705998404f27d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 10:14:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e467c7443e8349bb9c6741da1d00d6e9 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bafd7bd871344e9bad805f138813d116 : SUCCESS in 3m 21s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/05abfd280a704e37a951ecb07d80aa32 : SUCCESS in 1h 44m 39s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/3bcd9384eef841698ea2da9dd4bb9d52 : SUCCESS in 1h 40m 47s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/3187e8186d134dc0adc76fbe14449d24 : SUCCESS in 1h 16m 33s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/e87a53ea0505456fb8c8be646a1fce91 : SUCCESS in 1h 15m 10s\n- kuryr-kubernetes-tempest-lower-constraints https://zuul.opendev.org/t/openstack/build/ecf3977e1c5b43978d7e0783f613d77d : SUCCESS in 1h 35m 30s (non-voting)\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/72008c8db456448dbcfe37e824babb8e : SUCCESS in 58m 51s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/be53acd1c883454196f0e300e5352eec : SUCCESS in 54m 25s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/fd6c51d8a7ca49eeba9dd1de61c9d633 : SUCCESS in 1h 02m 45s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/ca3ef046e9fd42628944c717bcdc1175 : SUCCESS in 1h 01m 11s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/ae336fb365564c3ba963e7b85b0ac77f : SUCCESS in 1h 29m 28s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/a254c38782024cd68badaa33192d9416 : SUCCESS in 1h 38m 24s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/b69170189e4a451c929ad002a2072ed3 : SUCCESS in 2h 50m 47s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/29ef17a992874989bf10220d76b998fb : SUCCESS in 1h 34m 58s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/27e81e4f66ba422eb4a10afe91782e6e : SUCCESS in 1h 33m 23s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/01428b703a164385bec5a48d2dc7bcff : SUCCESS in 1h 10m 24s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/4e855d3768954b5ab8e920d72a703d21 : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f2e2d6058cd1b8a89ac84581f57dc11bd3e15692","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-04-18 12:26:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d9c222681321648f8f08e9576754d4d8baa6bb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 12:27:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"811282f1cc145c6d3f80a3d94fa7c5942196ab4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 13:58:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb918d2669b14833929d93431bf0f41a : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98ce83d95d7340eebee48277b882cc50 : SUCCESS in 3m 08s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/0b791532bec64007ac2056bc47878f37 : SUCCESS in 1h 30m 25s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/cf9cf759ad1f4588a4d37755dfe918e1 : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"af6fb8d8a387aac33ebe29cb8b7fe75f1caec512","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 13:58:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"94c09e1349b7dc266ea7e6ed174f94656fb033a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 14:00:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8ad7248e8042aa990b8092ad156473 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4debdd1711ac84aaea15ee3d77aea006389315d","revisions":{"d4debdd1711ac84aaea15ee3d77aea006389315d":{"kind":"REWORK","_number":1,"created":"2022-04-15 07:18:23.000000000","uploader":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"ref":"refs/changes/93/838093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/93/838093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/838093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/838093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/838093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/93/838093/1"}}},"commit":{"parents":[{"commit":"65f6cef16b2f3ca8eed2ba59800096dbad3d53e3","subject":"Merge \"Check that pod is deleted in delete_pod\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/65f6cef16b2f3ca8eed2ba59800096dbad3d53e3"}]}],"author":{"name":"Roman Dobosz","email":"gryf73@gmail.com","date":"2022-04-15 07:17:04.000000000","tz":120},"committer":{"name":"Roman Dobosz","email":"gryf73@gmail.com","date":"2022-04-15 07:17:04.000000000","tz":120},"subject":"Update \u0027Maintained\u0027 stable branch jobs on master gate","message":"Update \u0027Maintained\u0027 stable branch jobs on master gate\n\nWe have stable/yoga released and stable/ussuri is in\n\u0027Extended Maintenance\u0027 state. Current \u0027Maintained\u0027\nstable branches are yoga, xena, wallaby, and victoria.\n\nAs per tempest stable branch testing policy[1], adding the\nnewly released stable/yoga job and remove the EM stable/ussuri\njob.\n\n[1] https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: Id7434270af5fc654fc9bff9e0105e2408db392f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d4debdd1711ac84aaea15ee3d77aea006389315d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/d4debdd1711ac84aaea15ee3d77aea006389315d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
