)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~852327","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I6feaac2d786a46593377331afe3bfbe8eceaf0ef","project":"openstack/kuryr-tempest-plugin","branch":"master","attention_set":{"35067":{"account":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"last_update":"2023-10-11 12:34:41.000000000","reason":"\u003cGERRIT_ACCOUNT_13692\u003e replied on the change","reason_account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6feaac2d786a46593377331afe3bfbe8eceaf0ef","subject":"Use py3 as the default runtime for tox","status":"NEW","created":"2022-08-06 01:39:41.000000000","updated":"2024-01-10 09:09:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c682a4a75292cedf91005e893c8b7ed1e6ad9601","_number":852327,"virtual_id_number":852327,"owner":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-09 07:31:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":2,"date":"2024-01-03 20:30:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":2,"date":"2023-10-11 12:34:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":1,"date":"2024-01-03 20:30:02.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}],"REVIEWER":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-06 04:46:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 12:34:41.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2024-01-03 20:30:02.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2024-01-10 06:04:54.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"CC"}],"messages":[{"id":"ca26fcbe9ee9aa4961a8580872c261089ea5a200","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2022-08-06 01:39:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"850c098ee95ec18b2e9048541b1b7639b3e3be92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 04:46:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee6c0deac4594cc1ad9c8c51c3399625\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ecc845ab7a40b9a19cc475c98ba36b : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a99f04d3fd4f55906689eaf35852f8 : SUCCESS in 3m 15s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/6f810d447b7744e3ab40deda8a10ddd4 : SUCCESS in 1h 25m 33s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/efde893d56c5456cbe4d402daaca055d : SUCCESS in 1h 20m 15s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/149410aa71b34ab7b7b76cf4e27c2e71 : SUCCESS in 1h 02m 11s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/71221534f93d465f851163a1d84019ff : SUCCESS in 1h 16m 42s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/2b60ed13a8f94749970242745ac4e5f6 : SUCCESS in 1h 06m 27s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/717c2ac434f5426b801191fcd8270e68 : SUCCESS in 55m 40s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/5d930cca373f41ce83f393831804e06b : SUCCESS in 57m 10s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/46f2d0d0904343ae9ded56013d46900f : SUCCESS in 1h 11m 31s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/59d10e811ef044469de8090ded3a5367 : SUCCESS in 1h 30m 24s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/84e327000ad84d7ba7e02427b1e54772 : SUCCESS in 1h 39m 53s (non-voting)\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/b46cf5e453934e4d940f26046de3f994 : POST_FAILURE in 45m 53s (non-voting)\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/ed698b4b003c4e7588333d53ad8b5329 : FAILURE in 3h 04m 59s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/c951c28aa4b942dca219f2607fceee2f : SUCCESS in 1h 28m 35s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/da7fe3ab66344caeb172e3674c0ea145 : SUCCESS in 1h 12m 59s\n- kuryr-kubernetes-tempest-victoria https://zuul.opendev.org/t/openstack/build/308e9fea22de40d4ad9f3070a636598f : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f73519ce221233c2aae1502ec40850c0a92e9652","author":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"date":"2023-08-09 05:42:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"636ae235a286499d866c930d90cee399578585d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:31:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbb13179985b47a29add630e1d3e5389\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c832d6b50fb14643b7648fda05e0a604 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7597ae5177e489eb2eff52e09326c49 : SUCCESS in 2m 31s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/328005bf3e664f2cb9eff036c0723547 : SUCCESS in 1h 20m 31s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/22441ed1be0c438eacc66f2179e0f871 : SUCCESS in 1h 35m 30s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/7e7a732ecea44e7eba2cd5e315a7bc71 : SUCCESS in 1h 04m 35s (non-voting)\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/c87b3951bb634edf9b6649a9bb9bb12f : SUCCESS in 1h 34m 24s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/9a440b5a235643e59944db4b56935d8e : SUCCESS in 1h 13m 33s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/3371d88b9c134240ba16223a3355d1ea : SUCCESS in 59m 46s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/b542beac8411496b80337576b0414fa3 : SUCCESS in 46m 16s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/60da69f80bec4f02b717172127c5f667 : SUCCESS in 54m 44s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/f25f66c6ec184f6283ac0514ef5addc0 : FAILURE in 1h 07m 44s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/3c76de8397a34c7a935d3a94ce602113 : SUCCESS in 54m 07s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/4f1b7df1f77d430496332370824964d9 : SUCCESS in 1h 42m 06s\n- kuryr-kubernetes-tempest-2023-1 https://zuul.opendev.org/t/openstack/build/b1c9ea4ec77549a4950286cb6f85ca4b : SUCCESS in 1h 16m 57s\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/7bd10f0e2c314097a41db908dc267c07 : SUCCESS in 1h 30m 09s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/d4aff73699ae4b3aa5e7851c410c4c83 : SUCCESS in 1h 42m 05s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/504a53607b45462183f1e7294e5803cc : SUCCESS in 1h 22m 55s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/8bc94df67aab43a1b36ae5aa20b35dfe : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7a4ee4e173c6aee02a243ddbae2754b401829b34","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2023-10-11 12:34:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2307ae5c744ecd99081f517ec15f5cd5f8806680","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2024-01-03 20:30:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"328c50688e2d9f50db0b7987e88981b3577075b2","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2024-01-10 06:04:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c682a4a75292cedf91005e893c8b7ed1e6ad9601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 09:09:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65d344958eb8405a9123040a6b745648\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a270ffa64a4d4ccc9face27b57a8b404 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1768bef7e45d482bb118cbd6959d9214 : SUCCESS in 4m 18s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/93501e871e504f81b37aea05a4059844 : SUCCESS in 1h 01m 29s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/7c4dfc31d16b487399abd0e649a40ad5 : SUCCESS in 1h 18m 36s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/9eb94bb53440431d915c2000b2fed360 : SUCCESS in 51m 35s (non-voting)\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/60071bd587b44b46ad1e4a3f69281ebc : SUCCESS in 1h 32m 27s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/3088bea0308642f39a8a6203ffcd2697 : SUCCESS in 56m 48s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/e4ee2c96c60849ae955c5babe9e858d5 : SUCCESS in 48m 36s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/aeca8d0093a548ba9193192c5974b839 : TIMED_OUT in 3h 02m 53s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/ae1b58abd43742778db58196dbfed275 : SUCCESS in 47m 41s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/0af819e4f4e54bfba150e43746ee77c3 : SUCCESS in 53m 43s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/b616c3d615254a76963887c011fdc066 : SUCCESS in 54m 28s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/26720953fd3f4311a5f24e1822822c20 : SUCCESS in 1h 30m 07s\n- kuryr-kubernetes-tempest-2023-1 https://zuul.opendev.org/t/openstack/build/fa939450a9f24be994c0ea76f74f149a : SUCCESS in 1h 28m 59s\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/a02f67015e884d72abeb72ca5a6e9473 : SUCCESS in 1h 19m 11s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/07328e78056a406aaea25f356bcd0dce : SUCCESS in 1h 43m 08s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/b5a605e0a8234a6a8e156195cab24685 : SUCCESS in 1h 16m 19s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/397dde5aaecb46b9b3e6d64ffa7465b2 : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee68cd22875fda36e49316ada29a282966fd243a","revisions":{"ee68cd22875fda36e49316ada29a282966fd243a":{"kind":"REWORK","_number":1,"created":"2022-08-06 01:39:41.000000000","uploader":{"_account_id":35067,"name":"wangjiaqi","email":"wangjiaqi07@inspur.com","username":"jiaqi07"},"ref":"refs/changes/27/852327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/27/852327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/27/852327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/27/852327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/27/852327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/27/852327/1"}}},"commit":{"parents":[{"commit":"c1b5419ca90761f4484e2c6d33e81bb7138a84f9","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/c1b5419ca90761f4484e2c6d33e81bb7138a84f9"}]}],"author":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-06 01:38:49.000000000","tz":480},"committer":{"name":"jiaqi07","email":"wangjiaqi07@inspur.com","date":"2022-08-06 01:38:56.000000000","tz":480},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update\nthis at each new cycle.\n\nChange-Id: I6feaac2d786a46593377331afe3bfbe8eceaf0ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ee68cd22875fda36e49316ada29a282966fd243a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ee68cd22875fda36e49316ada29a282966fd243a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
