)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~861531","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I04b4acadd2864ab3160a97888dbb912c18951907","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"zed-stable-job","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2022-10-21 13:28:26.000000000","reason":"\u003cGERRIT_ACCOUNT_11600\u003e replied on the change","reason_account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-10-21 16:45:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I04b4acadd2864ab3160a97888dbb912c18951907","subject":"Add stable/zed jobs on master gate","status":"MERGED","created":"2022-10-16 04:04:11.000000000","updated":"2022-10-21 16:47:36.000000000","submitted":"2022-10-21 16:45:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861531-zed-stable-job","meta_rev_id":"92449fd4812ed897b0454159cb5da1f299eb366e","_number":861531,"virtual_id_number":861531,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-21 16:45:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2022-10-20 10:52:00.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2022-10-19 15:48:12.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":1,"date":"2022-10-20 05:47:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2022-10-19 15:48:12.000000000","_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-16 05:42:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-17 09:46:33.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2022-10-19 14:05:50.000000000","updated_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2022-10-20 05:47:06.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"8850d3b004c207ead2c669413bbf1867bd6f006d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-16 04:04:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9557f0a5808f494422f389abc246ada00b1de89c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-16 05:42:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2600dc23ddde4b77bea0471ff7776852\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f027785d6eb848358dd288b0caa0b4de : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd932aba68e4d20b552fa3f13a8c2ec : SUCCESS in 3m 01s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/a63c4d4b9adb42c683799e0b40bf244f : SUCCESS in 1h 14m 23s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/a135c0c0538144b0a9cb49a8d8e8cc92 : SUCCESS in 1h 23m 35s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/1f02d280e57947248e24683bd754c1f1 : SUCCESS in 55m 14s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/9b5a01dc68934abdbd9a360c8408f589 : SUCCESS in 1h 16m 43s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/88f06c4b3d4c4175b922220cbd328ced : SUCCESS in 56m 14s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/975a17622c0a47a191f58bd3834fb12a : SUCCESS in 46m 41s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/91f8ce0a20e4441e8299971703a48e9a : SUCCESS in 47m 16s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/d892d128cc4546c6b202db2804ee8e21 : SUCCESS in 45m 20s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/151896fbccf648178cc5251ef33c2b84 : SUCCESS in 1h 07m 02s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/d5a5c293bdc54b72acd7d13e7c84e489 : SUCCESS in 1h 04m 22s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/baf495e1337548d690b766358ee1bc5d : SUCCESS in 1h 33m 50s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/7c1c768955b9495f8c5b3b3f603391f9 : POST_FAILURE in 55m 14s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/18d5c37145f04308bcf059872ea9e32a : SUCCESS in 1h 22m 53s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/908b82fb87b4464b94380600afc4a739 : SUCCESS in 1h 21m 28s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/69b04367c818457aa6bde10f8c18d45b : FAILURE in 31m 06s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/ba3ee230721847e3bd5118e341a13ead : SUCCESS in 1h 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"aefa4d0025a9e88ca8da433bf08cd4302702882d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-10-17 09:46:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c511a8da8d7674d36da0b5640f62fd74472de4e","author":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"date":"2022-10-19 14:05:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea46c53b0c08ed91e691dd3558ef7c00a95218fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 15:29:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2c7eb5b3764438a8306982debfe42f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ed587bae0f40cbb4f246b05aab7c54 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92f178e42a4e4e1cbeb8f40c388716ed : SUCCESS in 2m 35s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/543958b7f4a948fe90320d70e4ea9fe8 : SUCCESS in 1h 14m 08s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/63d4163dad384c059ef7803f605692d6 : SUCCESS in 1h 01m 51s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/8ab7cbe21f7f4db9862f367fe1087431 : SUCCESS in 58m 49s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/eb8f5f675f27432984f9ab211f7b5633 : SUCCESS in 1h 06m 34s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/c2ae14b3e1cc4ad8a1bbc01867d4bcaf : SUCCESS in 58m 11s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/0cba7e8e8568492ab0110c2b255921d1 : SUCCESS in 43m 36s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/21f4f832c06a4a48ab43b70dbd880d90 : SUCCESS in 45m 33s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/fcd108b5cbb64957a1c249e08ee18463 : SUCCESS in 47m 03s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/9859be290dc944f88919c52ae4023471 : SUCCESS in 1h 01m 45s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/5d870ca284074373be08468d308f3460 : FAILURE in 1h 03m 29s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/88688fb7f321454ab37ed404c1e14284 : SUCCESS in 1h 14m 57s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/678c120bf9f9401193e09dca7bcf827f : POST_FAILURE in 41m 08s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/4e8529fc6899489c8713421f251ac90f : SUCCESS in 1h 09m 00s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/76a1dcb5c8ee490db7ab29f079651062 : SUCCESS in 1h 16m 11s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/99cf8c486abf42eaabe48fc4531d78ab : FAILURE in 23m 12s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/bcfe2a2d224b4b84beaec6dbe99a3902 : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8f2fc55d33e3b358b7cc6524d6080b3491aefeb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-10-19 15:48:12.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"86f61b7a7db85ef86c3331ce226ecd83641fa7d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:39:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/926a88f274604fa584814d0330f32562\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84d511eb3824a0cabb5737f8542b89a : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9e5132d51549a1a1b495fe297b3bde : SUCCESS in 2m 35s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/f262836471e148b0a73d4c37c0543919 : SUCCESS in 1h 26m 50s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/8e6325ea69fa480e83e8837afa6391a3 : FAILURE in 1h 46m 06s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/907ac8212be34058a1e38c9f91f644e9 : SUCCESS in 57m 05s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/f4964d99f5c846a2a1cdc55989effa12 : SUCCESS in 1h 37m 49s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/931c4d5507494ce5b41806e04e2217fb : SUCCESS in 58m 05s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/e056ea60b83b47578d66938726938618 : SUCCESS in 45m 59s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/6f79274e401741bb95f6a0825a48bc8d : SUCCESS in 50m 20s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/ada70eed9a9f4f34a84a2fd80f94337a : SUCCESS in 48m 24s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/0439d6e3e93a4d5b8af2fe1f2c0ee450 : FAILURE in 1h 07m 19s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/df86e280d9d74770a8b0072a4861538e : SUCCESS in 1h 01m 05s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/5b0b35ed12df4c53add744e008c335da : FAILURE in 9m 37s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/cc1d417fd9c6478f8ee4252309ef55ae : POST_FAILURE in 44m 42s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/ca363bedb5864ee6a44ee7b5fc0dbee5 : FAILURE in 1h 23m 35s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/6bb5f6ed493645f9a1c2511e20387e04 : SUCCESS in 1h 15m 51s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/be027809c79c41c0be8e8cd1b3c6fea4 : FAILURE in 1h 07m 49s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/49fe431c3d354dae941dc4e09f0447f7 : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f59045e4ac59f6a86d50bea6ef10e8a284765510","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-20 05:47:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3cc212546712db4310c360c36cc4b2efc7ffad97","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-10-20 10:52:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bd6de892a4b2fefd93d139a35c33e753dcb6015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:38:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35749835fc0f4adbb29deb28b6c58682\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8c10a8e9214d438e28ba5fc252bdc1 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f3912290944e2f9eacb22f2494a6c3 : SUCCESS in 2m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/403ea001e1e948468570cab12caeab79 : SUCCESS in 1h 30m 22s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/ea40c5b584494afc85a8237a031068a9 : SUCCESS in 1h 28m 15s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/17f8a2663cb1464aaf5e800b14d2dec5 : SUCCESS in 1h 02m 37s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/2a887b645df243339a8a9d641a8abc01 : SUCCESS in 1h 13m 41s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/b7d74f210f7b449ea1a3f1715807f91c : SUCCESS in 1h 11m 27s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/ca88a5c98bb0458ca86db1c9099794c1 : FAILURE in 20m 38s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/308d9d5e35aa4ad5b09112c4b7e2c238 : FAILURE in 1h 04m 16s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/fc4fc2a9b6d6476cb113bf077826f3e2 : SUCCESS in 1h 03m 11s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/c9b89202519e4b648ed2825a180cedb9 : SUCCESS in 1h 10m 22s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/a45c1c09c56c408fa93eb8126ef02bad : SUCCESS in 1h 14m 21s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/40bf8622358043c4a688172e77533e11 : SUCCESS in 1h 34m 48s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/c51a7c8a58284eb9b0c2a1b5097388f5 : POST_FAILURE in 45m 07s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/524e5279eeb348e6ad83d94c544e6e8f : SUCCESS in 1h 38m 55s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/76d11917284946e3a89f26eb7ec8bb6d : SUCCESS in 1h 40m 43s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/e612b74423b349c38d3caa2b5673aa8e : SUCCESS in 1h 06m 23s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/c4a5b49e94dd478cb7fe9b5dce314b9e : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":2},{"id":"61d823a3520eb2ca87e62e06572fbd4838db3495","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:39:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e6af54b368895e36cd4c7b51e4f08100159ed103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 14:25:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff5097a9a0a94e7f9fa65fcf21a11747\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf8ea3cf8584423a3732fd80143b194 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce1937491484e9688811fd9d5e7f449 : SUCCESS in 2m 43s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/355db178db734ea7aec62dbb451597fa : FAILURE in 1h 30m 48s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/aeb86a26ae91490399af980efeae90bd : SUCCESS in 1h 39m 11s","accounts_in_message":[],"_revision_number":2},{"id":"b4ae427b9a4adaaf6744401e6b91b9d53ad1a235","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-10-20 14:47:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65a4635577131a5b925148562e9c37a06cfca007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 16:12:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5a4b9ca0a3040da93a1381b7d362c77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41965a532f54412795e4fe5e1751d59d : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783dea316be84b1e8b95c4a600b3aef9 : SUCCESS in 2m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/c3d606ca58724e009007b045231192e8 : SUCCESS in 1h 16m 38s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/b46ef1d3fa9f4e25b7fe1af66d92347f : SUCCESS in 1h 20m 54s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/286ce71e85574d2d96e864148be93631 : SUCCESS in 1h 02m 43s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/aab6847603994fbdbacbbbe24506e8c7 : SUCCESS in 1h 19m 32s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/68ef38e45e0f4045a57dedaf873d1406 : SUCCESS in 1h 13m 30s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/e3173acbe5c14fc5bd8a6fef38d4216e : SUCCESS in 48m 05s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/036209a216c0477e9039202670dfbddc : SUCCESS in 56m 09s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/1d085ead838d44afac21872d0ac193bc : SUCCESS in 46m 51s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/121a02b0cd7d46d5b612ea86a92d0692 : FAILURE in 1h 16m 07s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/6fa3faefb4da41f283b327cb1afd7b37 : SUCCESS in 1h 01m 20s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/936ced4974214e45b99caa9b552367df : SUCCESS in 1h 16m 52s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/94aff89f9ed44dc9bfde946ea9a497fc : POST_FAILURE in 44m 16s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/795e4e431bbf417aaa0f52ed1680978b : SUCCESS in 1h 13m 37s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/56bdda15c6da4a68bac18308f361fc35 : SUCCESS in 1h 20m 21s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/1423a074b5eb4ce8b0b79b43537a12ce : SUCCESS in 1h 17m 13s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/44c6b71f158540d2af3b0e6741cd9a7f : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3b4ebc54573524038d81cd5d41c04a52ab763f30","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-10-20 16:21:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb1284585750ec6f0a44e8e7400cd045b490e22d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 18:20:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1be137ff1bcf4487af5ba739dafa0a3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94ee9d63bef45d986e3e4ff28a0a112 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89fdfb6821c744ecaeb324398b2ab298 : SUCCESS in 2m 36s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/fe53eef3a723497aa1e82d7dee05b505 : SUCCESS in 1h 35m 50s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/2daa16438c554b72a8351969a6772a4e : SUCCESS in 1h 26m 10s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/8ce323a517b04936a54287d06738e610 : FAILURE in 55m 11s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/0476d8d6fe6a4ab18635ed96223eed32 : SUCCESS in 1h 26m 55s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/d21911aab6f84d74941e884bcf287ae1 : SUCCESS in 59m 38s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/d3db2fc924654859b65e3be7e4ec8785 : SUCCESS in 51m 40s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/f05def8e39714a2eaf948b108444419b : SUCCESS in 44m 40s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/b5523ea4495f42aead9f56df00281528 : SUCCESS in 48m 59s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/eaa8d99e4d294e96940b2cfa1829c8fc : FAILURE in 1h 16m 08s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/b281f44eb28a4cbbb7bb06be5150243a : SUCCESS in 57m 18s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/fb05db1e723843bca0da889505c0eb01 : SUCCESS in 1h 52m 25s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/47aa88ed82e24c159195fcfad4bc3e4f : POST_FAILURE in 56m 40s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/ed4143b8d9004601b1ef6725b3da09c5 : SUCCESS in 1h 41m 32s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/6677ce906e414a9fbc1fc3db92d40268 : SUCCESS in 1h 31m 33s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/a8d97b5487b74c6da65c27c8884efb26 : SUCCESS in 1h 34m 28s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/4902aa3a85f54e53903ac3d30455ecac : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b97d8c4096e525950c7f0b7b516746afd04309f5","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2022-10-21 13:28:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"498e0dd21956c90e5ff38f784211d2fa386214d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 15:00:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5457dab527a244cea486dd5c7fca3b6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4751efbb84e40d286c38ff9c03f2315 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6344c8ac281d4bb1adff2c7d816ddc28 : SUCCESS in 2m 52s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/d535707a2acc49ae93f57f2929f3649b : SUCCESS in 1h 15m 24s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/0371ea7780eb46099d83a00d1a2d7d2b : SUCCESS in 1h 23m 13s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/e479004df4e04283919383d8d346124c : SUCCESS in 54m 07s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/11bc98bc9d9c42ba8d1dd7392e685794 : SUCCESS in 1h 06m 21s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/3a81e5c645c843dcb8b00c0896aaf403 : SUCCESS in 56m 45s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/ed8c556aa6b248eaa412baafcd845436 : SUCCESS in 56m 54s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/23e0b965c26544a0a18c8a4edb34ce97 : SUCCESS in 51m 15s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/a2ac2cf514eb4e84a5a3e7fb03580c91 : SUCCESS in 1h 11m 24s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/6572ff5e4fc44723b1cc89b76a7fac4c : SUCCESS in 55m 29s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/8dd7569817e64398ba92000daf8407a6 : SUCCESS in 1h 28m 59s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/64c884f8065644acb4530254e709bebe : SUCCESS in 1h 30m 30s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/3553159962bb466a8cce4f216577181f : POST_FAILURE in 39m 13s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/f646be4bd3934b8e97264e31914f0b4b : SUCCESS in 1h 24m 14s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/6788fd742fbc424eb65c6c3a5269dbf1 : SUCCESS in 1h 26m 34s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/48eab5bb3b7647efa06a463ddea3183d : SUCCESS in 1h 12m 07s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/69a4bbd2daa44fafa98cd38e73798ac8 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":2},{"id":"1e7a490eeb3dfbb886dba9a2b5be0983360760cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 15:01:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4d192dc8a9b8bc4eb90c3ee15a46350cdcb73222","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:45:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd46ef20e2ce4bc5ba4d012b1e80afa2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c587b141634de8956e1da9a419fca3 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23114c5710bf40ccae30c3a1139fe86b : SUCCESS in 3m 18s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/3c953759eaac499cb0ab218205237d22 : SUCCESS in 1h 40m 10s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/85f310c4dc924c2daf6fd6f00f565c57 : SUCCESS in 1h 28m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8e884195be4db27c1c3f225e84b5ea62b0ff7834","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:45:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"92449fd4812ed897b0454159cb5da1f299eb366e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 16:47:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fe238664f844f18bf55a2c27dc451c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d8a97dc064f427d865e4a60f55e8430 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"34255b30a21971bb4b99a6f39973d0bc24d7e113","revisions":{"5893aa751348b4e4da8c27a6ba556ca85c998a20":{"kind":"REWORK","_number":1,"created":"2022-10-16 04:04:11.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/31/861531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/31/861531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/861531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/861531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/861531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/861531/1"}}},"commit":{"parents":[{"commit":"050861ca88330eb3b7347e76d1205c67837f2dad","subject":"Pin stable branch jobs nodeset to Ubuntu Focal (20.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/050861ca88330eb3b7347e76d1205c67837f2dad"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-16 03:50:27.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-16 03:50:27.000000000","tz":-300},"subject":"Add stable/zed jobs on master gate","message":"Add stable/zed jobs on master gate\n\nAs zed is released, we should add its job on master\ngate to keep branchless tempest plugins compatible\nto stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I04b4acadd2864ab3160a97888dbb912c18951907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5893aa751348b4e4da8c27a6ba556ca85c998a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/5893aa751348b4e4da8c27a6ba556ca85c998a20"}]},"branch":"refs/heads/master"},"34255b30a21971bb4b99a6f39973d0bc24d7e113":{"kind":"NO_CHANGE","_number":2,"created":"2022-10-19 15:48:12.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/31/861531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/31/861531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/861531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/861531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/861531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/31/861531/2"}}},"commit":{"parents":[{"commit":"0863d4ad32d1ffb83d27096abeb2a3c54487681e","subject":"Pin stable branch jobs nodeset to Ubuntu Focal (20.04)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/0863d4ad32d1ffb83d27096abeb2a3c54487681e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-16 03:50:27.000000000","tz":-300},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2022-10-19 15:48:01.000000000","tz":120},"subject":"Add stable/zed jobs on master gate","message":"Add stable/zed jobs on master gate\n\nAs zed is released, we should add its job on master\ngate to keep branchless tempest plugins compatible\nto stable branch.\n\nRef: Tempest plugins guide for stable branch testing:\n- https://docs.openstack.org/tempest/latest/stable_branch_testing_policy.html\n\nChange-Id: I04b4acadd2864ab3160a97888dbb912c18951907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/34255b30a21971bb4b99a6f39973d0bc24d7e113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/34255b30a21971bb4b99a6f39973d0bc24d7e113"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}}]}],"submit_requirements":[]}
