)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~868947","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~Ibbca8bf7ddc1550eb6e0c3f0fcebbc31c9b77e4c","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"subnet-name","attention_set":{},"removed_from_attention_set":{"11600":{"account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"last_update":"2023-01-02 16:16:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11600\u003e replied on the change","reason_account":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},"9284":{"account":{"_account_id":9284,"name":"Michal Dubiel","email":"md@semihalf.com","username":"md"},"last_update":"2023-01-03 14:41:13.000000000","reason":"Change was submitted"},"4727":{"account":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"last_update":"2023-01-03 14:41:13.000000000","reason":"Change was submitted"},"27032":{"account":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"last_update":"2023-01-03 14:41:13.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2023-01-02 11:33:02.000000000","reason":"\u003cGERRIT_ACCOUNT_13692\u003e replied on the change","reason_account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}},"hashtags":[],"change_id":"Ibbca8bf7ddc1550eb6e0c3f0fcebbc31c9b77e4c","subject":"Fix get subnet name in get_port_pool","status":"MERGED","created":"2023-01-01 10:36:24.000000000","updated":"2023-01-03 14:42:11.000000000","submitted":"2023-01-03 14:41:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868947-subnet-name","meta_rev_id":"72cc3e06a0b9715cc592c26b19882777949b656f","_number":868947,"virtual_id_number":868947,"owner":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-03 14:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2023-01-02 12:49:16.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2023-01-02 11:33:02.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":1,"date":"2023-01-02 12:49:16.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9284,"name":"Michal Dubiel","email":"md@semihalf.com","username":"md"},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}],"reviewers":{"REVIEWER":[{"_account_id":9284,"name":"Michal Dubiel","email":"md@semihalf.com","username":"md"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-01 10:39:27.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":27032,"name":"Maysa de Macedo Souza","email":"maysa.macedo95@gmail.com","username":"maysa"},"state":"REVIEWER"},{"updated":"2023-01-01 10:39:27.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":9284,"name":"Michal Dubiel","email":"md@semihalf.com","username":"md"},"state":"REVIEWER"},{"updated":"2023-01-01 10:39:27.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2023-01-01 12:07:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-02 10:50:12.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"}],"messages":[{"id":"3d4cbcc91faa872d47a58802fa17d129b45fb9fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2023-01-01 10:36:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21c06fd0634676e35746a4600e0d0e4b5b42d14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 12:07:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c33683d0238e4890adb6da16241eb3b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40b526af1fe4467aa812b6930c2936c : FAILURE in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e131d890a0064d819213d01b28f7b814 : FAILURE in 1m 50s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/ea070533a74e47b5b50edf3b97bfcb72 : SUCCESS in 1h 08m 11s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/8832e3cb97654994a75bfc039ae12c9c : SUCCESS in 1h 06m 09s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/824cb3896e0e4826b1612140fe620203 : SUCCESS in 1h 01m 05s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/004157bbe28f470a905aef21769c0784 : SUCCESS in 57m 12s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/bba5cb10f50e4356af94f0ceec3fdd74 : SUCCESS in 1h 01m 32s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/4a81446951414853b51c7c2dec26da9e : SUCCESS in 52m 00s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/465d880861b74693b0ba441ca55d7b5b : SUCCESS in 54m 12s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/53ae7d241dc941ffa25745dad1e71be2 : SUCCESS in 53m 14s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/31d07adaba93454d9ddae82ef05aeae8 : SUCCESS in 1h 15m 18s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/9a354415c7554ef9bb381899d9b7f950 : SUCCESS in 1h 01m 39s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/aa3d9d277a754ff79ed509e584a3d934 : SUCCESS in 1h 01m 40s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/64da2723f6674fcfa7f11dbee77aefc9 : POST_FAILURE in 34m 04s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/2fd7b3f1e2db480c92a3e619c4d6bdcd : SUCCESS in 1h 06m 41s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/012368aab1534183aa1613742ad52a33 : SUCCESS in 1h 16m 16s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/9ccc24b23f59478e965694cf81443b7e : SUCCESS in 1h 29m 00s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/ac6b01ec31254e9b82a527e2d79ffee9 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8a7c889d6f17850780497d85f8d2754e2a9f2b5d","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2023-01-02 10:50:12.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8df7c39148ddfb78dff8046a431160f975c771a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2023-01-02 10:55:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8da1627bbf69fc02478e66c002bc42f56f9c4d9d","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2023-01-02 11:33:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b6e6df7e98677baf0f250a111468f6481500c20f","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2023-01-02 12:49:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7018acb56edd284938200f31ee7bddf2ad8ccd45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 12:51:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e9a3385ed1d4718a9333d29cc2d1b30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49315bca7014d379be3a9428b25347a : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e091eb2660364ba2bb3496590d0ca71e : SUCCESS in 3m 00s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/5b7d29eecf3746849265ff7b5e32ead5 : SUCCESS in 1h 25m 15s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/f9c33d55623d43b1a12b271f1440d761 : SUCCESS in 1h 19m 55s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/53065597a0cc4f1ebf22b8b549e1249f : SUCCESS in 47m 06s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/0b69bad2114044f19d9408acdf679c92 : SUCCESS in 1h 08m 30s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/aea0183ee2024fad9caab5b5b9a738f9 : SUCCESS in 49m 15s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/2b2a4b79583844e29188a75ae0e4ef3b : SUCCESS in 43m 30s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/6884ef9cb1654c99990b7a0ad1535f27 : SUCCESS in 41m 07s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/795516cc1af846c3b3ead130d171b55b : SUCCESS in 40m 10s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/fd6ddf3ad0004eeb8ba097aa81e67927 : FAILURE in 55m 28s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/00a1c2b3219640efbaa79f2151a2e6c9 : SUCCESS in 1h 01m 28s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/f4d6685bf84b4cd092afc78e741266a9 : SUCCESS in 1h 39m 07s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/70a9b90d1243467fa872efa1b8484cbd : POST_FAILURE in 48m 46s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/c99d0330f81740afb69e771ed9fc6832 : SUCCESS in 1h 30m 56s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/c4f48594195440899da84b198aabcf55 : SUCCESS in 1h 47m 15s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/d61cff1bd43f4410bf0c486a81d431cb : SUCCESS in 1h 17m 54s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/6220b24485db4b828b165a16da34cadc : SUCCESS in 1h 19m 36s","accounts_in_message":[],"_revision_number":2},{"id":"0676b1e466f4528ef3a22dfc2cf793bb00e7a648","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2023-01-02 13:27:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d8e74976d90ece25912f681fe24fc96202220453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 15:26:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/012cb544949c4ecfaa83235df0587328\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee24a95815248139e5f8ca79372705d : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6071f8fa93394f29a1c5941cc5e86815 : SUCCESS in 2m 22s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/db6ac1efe8ec4e379d32c9bd8d6731aa : SUCCESS in 1h 41m 46s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/af93766775404a07bafe03d8a8ddfd66 : SUCCESS in 1h 40m 29s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/7ebcf9c393354cd5a01707c0bc7f678d : FAILURE in 1h 20m 26s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/236f8ca1665c4380841c0dba58a99563 : SUCCESS in 1h 37m 09s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/41a37d69a2d143a2950110b7be28fa30 : FAILURE in 1h 27m 56s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/cb9c3dc2a357462b926ab441943eca8b : FAILURE in 23m 08s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/4f21f9f1db4640958d23ec65ef9b556a : SUCCESS in 1h 08m 20s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/c8c95d8913d544c4aeae620e61ab6b22 : SUCCESS in 1h 00m 37s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/29af23ee78a14309a0e5e61aba1961a5 : FAILURE in 1h 23m 01s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/24bf06f64dca40a9b224eb07fce684bb : SUCCESS in 1h 15m 55s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/fd5b137507ec4584aae550750c8a1e7a : SUCCESS in 1h 51m 43s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/75742e00c26248ee82b5991fb960734b : POST_FAILURE in 1h 12m 06s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/e7ce7fe769f0467d8e26539b11f8401f : SUCCESS in 1h 30m 36s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/5c84536a1aa84a769c6798adf838bf87 : SUCCESS in 1h 51m 25s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/4313e024304341f5b772ba8ac35843e2 : SUCCESS in 1h 38m 35s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/d3df3a88148641d3b6e23c1cbb0cd775 : SUCCESS in 1h 30m 05s","accounts_in_message":[],"_revision_number":2},{"id":"20f883cf18d1e2036684f1fe5efb695fd27b6690","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2023-01-02 16:16:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ce828b6fb4f0b363a2a852a19b0d80db1541e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 18:10:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e5f90531f1a47b3b17f83603cd44bb9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44cd0ea8da694ed0b2a5c9097c4dccf0 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c6366ad676450b8c52c372db0f11b6 : SUCCESS in 2m 22s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/dabf3aa2748e47dfa4399cd09eaa58fc : SUCCESS in 1h 22m 37s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/d0ce215960414e11a713e759d3050ed6 : SUCCESS in 1h 37m 31s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/f868cc3037ec45f2a3de95b4a30eb128 : SUCCESS in 57m 12s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/79faf95f883346038bc7ea3a437474c2 : SUCCESS in 1h 48m 13s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/269e553ff657400d8520ab609615034c : SUCCESS in 1h 13m 17s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/c8668ca5e20d4e45aa244f05d8169529 : SUCCESS in 51m 51s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/5b127aaede3d407a9eced241a40090fa : FAILURE in 17m 39s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/4b1aed2805b04affa4eab495a48c5962 : SUCCESS in 48m 28s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/27d37839593b45ca889bec51f3649bd6 : SUCCESS in 1h 15m 25s\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/08a43abd3a074435b26ff005ed16b869 : SUCCESS in 1h 06m 13s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/9304cdf141f64314884e86b06a2f31e3 : SUCCESS in 1h 25m 21s\n- kuryr-kubernetes-e2e-np https://zuul.opendev.org/t/openstack/build/16c4d2d8773b4df5b8e924d6b51795a6 : POST_FAILURE in 57m 23s (non-voting)\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/5b7f872a3a354b3fa810231891d57a73 : SUCCESS in 1h 28m 40s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/2b1062a4a9c2483aaecfa3335558942d : SUCCESS in 1h 39m 41s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/4102e63c5f4d4d6087f6cd6314ade84d : SUCCESS in 1h 27m 12s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/487ec5373a5b4d82808fa0c5be26cdc7 : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0ab46fe1af0528d95e2e58dcc6e4620461657302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 12:57:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a4f1a5fb439863154fbac3427cabff1d7cfb6c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 14:41:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9232ed8aade4e24b16dfda96ddc3f56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb662fe0ca74439908e74f354ee2bea : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a6704250c04bd8ac6a8c0fd4b3c58f : SUCCESS in 2m 41s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/9892d5dcbb4e4398b7de78b136f59f2a : SUCCESS in 1h 38m 58s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/8577d67d64ba4471a9f589a6ad117c52 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":2},{"id":"c69ee38cda3e16f06154c19b75edafad58c71607","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 14:41:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"72cc3e06a0b9715cc592c26b19882777949b656f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 14:42:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56b8d00c3caa47b28ed058863c206720\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5c4f1736b094be8bffaec38c299a6ff : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"76f0c5ac74e2a52c4fb9301e6fc7ca3a92b8b345","revisions":{"2b540bb4cd372111c56854a641835c688687ac43":{"kind":"REWORK","_number":1,"created":"2023-01-01 10:36:24.000000000","uploader":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"ref":"refs/changes/47/868947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/47/868947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/47/868947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/47/868947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/47/868947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/47/868947/1"}}},"commit":{"parents":[{"commit":"34255b30a21971bb4b99a6f39973d0bc24d7e113","subject":"Add stable/zed jobs on master gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/34255b30a21971bb4b99a6f39973d0bc24d7e113"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2023-01-01 10:34:32.000000000","tz":120},"committer":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2023-01-01 10:34:32.000000000","tz":120},"subject":"Fix get subnet name in get_port_pool","message":"Fix get subnet name in get_port_pool\n\nSubnet name in the form of ns_uid/ns_name was added\n\nChange-Id: Ibbca8bf7ddc1550eb6e0c3f0fcebbc31c9b77e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b540bb4cd372111c56854a641835c688687ac43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/2b540bb4cd372111c56854a641835c688687ac43"}]},"branch":"refs/heads/master"},"76f0c5ac74e2a52c4fb9301e6fc7ca3a92b8b345":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-01-02 10:55:18.000000000","uploader":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"ref":"refs/changes/47/868947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/47/868947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/47/868947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/47/868947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/47/868947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/47/868947/2"}}},"commit":{"parents":[{"commit":"82934553ab8afa5109ea971b5798c407318584f8","subject":"Update tox.ini for new tox version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/82934553ab8afa5109ea971b5798c407318584f8"}]}],"author":{"name":"Itzik Brown","email":"itzikb@redhat.com","date":"2023-01-01 10:34:32.000000000","tz":120},"committer":{"name":"Michał Dulko","email":"mdulko@redhat.com","date":"2023-01-02 10:55:18.000000000","tz":0},"subject":"Fix get subnet name in get_port_pool","message":"Fix get subnet name in get_port_pool\n\nSubnet name in the form of ns_uid/ns_name was added\n\nChange-Id: Ibbca8bf7ddc1550eb6e0c3f0fcebbc31c9b77e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/76f0c5ac74e2a52c4fb9301e6fc7ca3a92b8b345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/76f0c5ac74e2a52c4fb9301e6fc7ca3a92b8b345"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"}}]}],"submit_requirements":[]}
