)]}'
{"id":"openstack%2Fkuryr-tempest-plugin~904667","triplet_id":"openstack%2Fkuryr-tempest-plugin~master~I2687b8b310bc50efb077b638ba4842419789fa84","project":"openstack/kuryr-tempest-plugin","branch":"master","topic":"py3.11","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-01-05 09:16:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2687b8b310bc50efb077b638ba4842419789fa84","subject":"Update python classifier in setup.cfg","status":"MERGED","created":"2024-01-04 07:57:17.000000000","updated":"2024-01-05 09:17:12.000000000","submitted":"2024-01-05 09:16:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904667-py3.11","meta_rev_id":"0f1c397c84d735d2c58ca701029135d0769571df","_number":904667,"virtual_id_number":904667,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-05 09:16:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":2,"date":"2024-01-04 10:00:50.000000000","_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":2,"date":"2024-01-05 07:45:56.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-05 02:59:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"all":[{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":1,"date":"2024-01-05 07:45:56.000000000","_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-04 09:48:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-04 10:00:50.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2024-01-05 02:59:43.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2024-01-05 07:45:56.000000000","updated_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"}],"messages":[{"id":"8ad6bcf10b5b4156b399556d6f6b9be15c885d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-04 07:57:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80922fd0847693cde6d220535e33ed21659ebb30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 09:48:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bdeef9ded00471a9eddb7967d64f570\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f56ab46882d644ccadec29383f49a112 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6dc0d184e44e4aa8846a921788999a : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcaed6bfff18420cad4b53aa1e8b15d4 : SUCCESS in 3m 40s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/decdae65aae04dfaa6386ffbc4cfef44 : SUCCESS in 1h 13m 00s\n- kuryr-kubernetes-tempest-docker https://zuul.opendev.org/t/openstack/build/df8e281b2bba4e21bffa107648a44587 : SUCCESS in 1h 43m 02s\n- kuryr-kubernetes-tempest-defaults https://zuul.opendev.org/t/openstack/build/d4d6b9d52767496db64231b4dc3de6e3 : FAILURE in 1h 07m 44s (non-voting)\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/bd57262684d548c0945387d4c1ff094b : SUCCESS in 1h 14m 29s\n- kuryr-kubernetes-tempest-multinode https://zuul.opendev.org/t/openstack/build/99beddf86ba04b2e9d0f337580900556 : SUCCESS in 53m 44s (non-voting)\n- kuryr-kubernetes-tempest-multinode-ovs https://zuul.opendev.org/t/openstack/build/f5e34a79e15c4cbe996148c371fb7dd3 : SUCCESS in 56m 07s (non-voting)\n- kuryr-kubernetes-tempest-ipv6 https://zuul.opendev.org/t/openstack/build/c0786468c08946c896a97be1b5cc6bed : SUCCESS in 55m 26s (non-voting)\n- kuryr-kubernetes-tempest-ipv6-ovs https://zuul.opendev.org/t/openstack/build/6216867915ff4a9987b5aaead184af06 : SUCCESS in 48m 17s (non-voting)\n- kuryr-kubernetes-tempest-amphora https://zuul.opendev.org/t/openstack/build/4d05d2cb258d47248fd089e2470a1ecd : FAILURE in 1h 09m 44s (non-voting)\n- kuryr-kubernetes-tempest-amphora-ovs https://zuul.opendev.org/t/openstack/build/26c8908502ba489ea0e287ce054d920b : SUCCESS in 1h 01m 55s (non-voting)\n- kuryr-kubernetes-tempest-annotation-project-driver https://zuul.opendev.org/t/openstack/build/8ea86aade0ae436eb15924b7ebb1d880 : SUCCESS in 1h 04m 55s\n- kuryr-kubernetes-tempest-2023-1 https://zuul.opendev.org/t/openstack/build/0f09258c27b246c2b91150c2bd310ab7 : SUCCESS in 1h 31m 12s\n- kuryr-kubernetes-tempest-zed https://zuul.opendev.org/t/openstack/build/f3387c39cf594e36ace06b506d87831b : SUCCESS in 1h 22m 28s\n- kuryr-kubernetes-tempest-yoga https://zuul.opendev.org/t/openstack/build/a80c7bc31c504c91ac2cb746d83a4db3 : SUCCESS in 1h 26m 08s\n- kuryr-kubernetes-tempest-xena https://zuul.opendev.org/t/openstack/build/5dd486a9ba37427f923f8c892d1a4559 : SUCCESS in 1h 30m 44s\n- kuryr-kubernetes-tempest-wallaby https://zuul.opendev.org/t/openstack/build/5330accd046b4a788ccd1e1a4301e302 : SUCCESS in 1h 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"77699064924684f6401568135a1c9bea38a1675f","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2024-01-04 10:00:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b8ae8dea1958e32c9a1a537bef5dda2687b9808","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-01-05 02:59:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"98b07c56c2c302f773b2cc935c31ea8820937c0f","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2024-01-05 07:45:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3835bea3471305c050f9dbd3f3a88fed371c8c9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 07:46:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb24cdf09f8eafa83f1f788610179fdb74434d53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 09:16:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a797c22ef3c74211b326e3a0af3af6d9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/646b566eaf2d4d5cad93ee9dd3684296 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124ec1d365034d31af1b3b010c55a21a : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ce10008c684e86b9444544a1156cb8 : SUCCESS in 3m 45s\n- kuryr-kubernetes-tempest https://zuul.opendev.org/t/openstack/build/922332654b104f10b6bfec1948f91b78 : SUCCESS in 1h 24m 14s\n- kuryr-kubernetes-tempest-systemd https://zuul.opendev.org/t/openstack/build/1760ba872e344fe38282885a669f6e05 : SUCCESS in 1h 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"adbedd6a4d9d5db7120413faa61e74cdbedc1a10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 09:16:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0f1c397c84d735d2c58ca701029135d0769571df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 09:17:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/405ea85a7de84e6e8ca736dbfcff7b58\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99fba5f6dc464612ac81ebfc6165102c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"115d6bd490e5afff2a40df40121e8c04894d9551","revisions":{"115d6bd490e5afff2a40df40121e8c04894d9551":{"kind":"REWORK","_number":1,"created":"2024-01-04 07:57:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/67/904667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr-tempest-plugin","ref":"refs/changes/67/904667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/67/904667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/67/904667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/67/904667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr-tempest-plugin refs/changes/67/904667/1"}}},"commit":{"parents":[{"commit":"ab45b2fadbbda3d1d97856121673e874b41de209","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/ab45b2fadbbda3d1d97856121673e874b41de209"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 07:25:02.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-01-04 07:57:16.000000000","tz":-480},"subject":"Update python classifier in setup.cfg","message":"Update python classifier in setup.cfg\n\nAs per the current release tested runtime, we test\npython version from 3.8 to 3.11 so updating the\nsame in python classifier in setup.cfg\n\nChange-Id: I2687b8b310bc50efb077b638ba4842419789fa84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/115d6bd490e5afff2a40df40121e8c04894d9551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr-tempest-plugin/commit/115d6bd490e5afff2a40df40121e8c04894d9551"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
