)]}'
{"id":"openstack%2Fkuryr~755049","triplet_id":"openstack%2Fkuryr~master~Icfb3d042ccb29b1879d77671d937c8df56f017e9","project":"openstack/kuryr","branch":"master","hashtags":[],"change_id":"Icfb3d042ccb29b1879d77671d937c8df56f017e9","subject":"bump py37 to py38 in tox.ini","status":"MERGED","created":"2020-09-29 14:35:44.000000000","updated":"2020-09-30 08:21:41.000000000","submitted":"2020-09-30 08:19:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755049-1601453995178-9817b790","meta_rev_id":"fea0981fcf0ad345c86c900ea4bd092a4fa54975","_number":755049,"virtual_id_number":755049,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-30 08:19:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"all":[{"value":2,"date":"2020-09-30 05:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":2,"date":"2020-09-30 08:09:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-30 07:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"all":[{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":1,"date":"2020-09-30 08:09:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 05:01:03.000000000","updated_by":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2020-09-30 07:07:42.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2020-09-30 08:09:56.000000000","updated_by":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"reviewer":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"state":"REVIEWER"},{"updated":"2020-09-30 08:19:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"238ce3cf6854196410b8ba34a8dd8d5b6b2e513e","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-29 14:35:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4367f0cfbf94f481bad20c37142243a53cee8f76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 14:53:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1a5850e3e2bf4632bb10c9d6fb5e2190 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/909287db1bd5476da4a0f91be9dcfe79 : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b15f08280e74074a3f474cdc4afadf8 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad4b2e7d7cd4dd6914294a2dd6c158f : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c107a6cb5fa044a9a07be389795c885b : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c63d38d711f94ce29b6b4fe310cce711 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2372c556c3d9421289c75b6863e0c0fc : SUCCESS in 4m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ce94fdb0b44451d934e270cdc08eced : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10b977064b7434eb8b9696eb1762f38 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"428ea10950b1ccc2f5208ecd86d2b281b7026b53","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-09-30 05:01:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8b8dcc695f814ab5696adcd4a42e2a240f39367","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2020-09-30 07:07:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1c3adde3ffaf907ca82796294b39a2718b846fd5","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-09-30 08:09:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3db85570d4ad39b0fad68e5976535df82fdf84d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 08:10:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8a7a217cfdff2e39c34037a09614895ecfff6d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 08:19:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/730139645bcc4e8c9e020b5d8a778a30 : SUCCESS in 2m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/232b8819bad3469ea0c9d252b09c0e36 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e2144ed89941538bb9d1cf25c00527 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a00b90eace6d4f48bf53e8a16162a7c7 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7ed4baf4d8a4318bfdc158f3d58930b : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f346a6ad0b744e9b69d082dd9cb00ea : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9c004bfad848a1803d16579fabf36b : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7b74d30ee517a8f17140978c15fe95df04a3f0eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 08:19:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fea0981fcf0ad345c86c900ea4bd092a4fa54975","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 08:21:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73cb3d0a49d14240ac89eeda703f49aa : SUCCESS in 1m 14s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17c4951441a46e0b3704d7af955e366 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46b1e6edbcf2d7ef92db694ba052e84e63e1361b","revisions":{"46b1e6edbcf2d7ef92db694ba052e84e63e1361b":{"kind":"REWORK","_number":1,"created":"2020-09-29 14:35:44.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/49/755049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kuryr","ref":"refs/changes/49/755049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kuryr refs/changes/49/755049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kuryr refs/changes/49/755049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kuryr refs/changes/49/755049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kuryr refs/changes/49/755049/1"}}},"commit":{"parents":[{"commit":"d9f4b79160c6b184a982fc1d5f50d313889f7ad5","subject":"Merge \"Update master for stable/victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr/commit/d9f4b79160c6b184a982fc1d5f50d313889f7ad5"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-29 14:35:06.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-29 14:35:06.000000000","tz":480},"subject":"bump py37 to py38 in tox.ini","message":"bump py37 to py38 in tox.ini\n\nin W cycle, we should test py38 by default.\n\nChange-Id: Icfb3d042ccb29b1879d77671d937c8df56f017e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr/commit/46b1e6edbcf2d7ef92db694ba052e84e63e1361b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kuryr/commit/46b1e6edbcf2d7ef92db694ba052e84e63e1361b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
