)]}'
{"id":"openstack%2Flarge-scale~847951","triplet_id":"openstack%2Flarge-scale~master~I5001fba0e9d6e2f26d99127abe9f97b79f46c233","project":"openstack/large-scale","branch":"master","topic":"add/largescale_scaleup_documentation","attention_set":{},"removed_from_attention_set":{"34533":{"account":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"last_update":"2022-06-28 13:59:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5001fba0e9d6e2f26d99127abe9f97b79f46c233","subject":"Transfer scale up page to rst format","status":"MERGED","created":"2022-06-28 11:03:15.000000000","updated":"2022-06-28 14:00:41.000000000","submitted":"2022-06-28 13:59:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847951-add/largescale_scaleup_documentation","meta_rev_id":"1253fb87227d7ba8e852293258fbe33ec2cb69f0","_number":847951,"virtual_id_number":847951,"owner":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-28 13:58:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-28 13:54:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-28 13:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 11:08:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-28 13:54:17.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"c6dde283c720b4d5d3b79413be84883b4e583a63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"date":"2022-06-28 11:03:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8cdf6ab5a48031c36fc6c13e6c3d9f37fdc9938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 11:08:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f3aed9dfa34fd2b3fbb6e5c45efdcf : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ec5a10ed45405a2ecbdc51e7dee66ac91eef6fb4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2022-06-28 13:54:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b95c8a853b86b10e17aa07af74644fd783e2917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:54:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e0132afae059917017412952000533ef864ff2a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:58:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d5a36f9a234e2fb192ef4322193a70 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b0f451bf0e7e4c40fbdfd733e6c3350b2d7d16c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:59:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1253fb87227d7ba8e852293258fbe33ec2cb69f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 14:00:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/668508a998dd4e4fa51f0f38f7b87f08 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8fb990ed2d58cfd0116686025436ceb22e3c0b86","revisions":{"8fb990ed2d58cfd0116686025436ceb22e3c0b86":{"kind":"REWORK","_number":1,"created":"2022-06-28 11:03:15.000000000","uploader":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"ref":"refs/changes/51/847951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/51/847951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/51/847951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/51/847951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/51/847951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/51/847951/1"}}},"commit":{"parents":[{"commit":"561faae3b3972284537c68db379d780baa982d72","subject":"Adding empty files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/561faae3b3972284537c68db379d780baa982d72"}]}],"author":{"name":"Ramona Rautenberg","email":"rautenberg@osism.tech","date":"2022-06-28 11:02:07.000000000","tz":120},"committer":{"name":"Ramona Rautenberg","email":"rautenberg@osism.tech","date":"2022-06-28 11:02:07.000000000","tz":120},"subject":"Transfer scale up page to rst format","message":"Transfer scale up page to rst format\n\nSigned-off-by: Ramona Rautenberg \u003crautenberg@osism.tech\u003e\nChange-Id: I5001fba0e9d6e2f26d99127abe9f97b79f46c233\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/8fb990ed2d58cfd0116686025436ceb22e3c0b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/8fb990ed2d58cfd0116686025436ceb22e3c0b86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[]}
