)]}'
{"id":"openstack%2Flarge-scale~848082","triplet_id":"openstack%2Flarge-scale~master~Icf2d38621f283d02f2d71a17ca2b668db94efd92","project":"openstack/large-scale","branch":"master","topic":"add/largescale_scaleout_documentation","attention_set":{},"removed_from_attention_set":{"34533":{"account":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"last_update":"2022-06-30 08:37:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf2d38621f283d02f2d71a17ca2b668db94efd92","subject":"Transfer scale out page to rst format","status":"MERGED","created":"2022-06-29 07:57:25.000000000","updated":"2022-06-30 08:38:20.000000000","submitted":"2022-06-30 08:37:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848082-add/largescale_scaleout_documentation","meta_rev_id":"7d4ec074488ee37bb8248109ce33c5d00ac58d9d","_number":848082,"virtual_id_number":848082,"owner":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-30 08:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-06-30 06:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-06-30 08:27:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-06-30 08:27:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 08:02:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 06:27:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-06-30 08:27:38.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"c6139f367e3620d7b24264fa3cff571eaf4a30e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"date":"2022-06-29 07:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cf1be5f4e1e36ce06e6069e710d158e484d71d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 08:02:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f05ced1a5234075bb41a69843a4f8e0 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"bee8a12b8a86f5f901d991ddf52172d56adf66f7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-30 06:27:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bfb45c3631df1b1a553e78849f4a7ba1f6c9a97e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-06-30 08:27:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1376d9eadd14171a6f12fb97d7abfa76cf1698b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:28:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b473e2237aa822841343af19dde2bff6e3a9bf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:37:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3a5d48d1e0b4e16af9fa0876ce07d5f : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"62547ae242d22181bdbe65d9d044304bd7f53de3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:37:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d4ec074488ee37bb8248109ce33c5d00ac58d9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:38:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/1b163666cc734d25b2480c78a813f1cd : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c0c072b8df3ca8f68de2426391b5cc3920eed5f","revisions":{"4c0c072b8df3ca8f68de2426391b5cc3920eed5f":{"kind":"REWORK","_number":1,"created":"2022-06-29 07:57:25.000000000","uploader":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"ref":"refs/changes/82/848082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/82/848082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/82/848082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/82/848082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/82/848082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/82/848082/1"}}},"commit":{"parents":[{"commit":"8fb990ed2d58cfd0116686025436ceb22e3c0b86","subject":"Transfer scale up page to rst format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/8fb990ed2d58cfd0116686025436ceb22e3c0b86"}]}],"author":{"name":"Ramona Rautenberg","email":"rautenberg@osism.tech","date":"2022-06-29 07:56:32.000000000","tz":120},"committer":{"name":"Ramona Rautenberg","email":"rautenberg@osism.tech","date":"2022-06-29 07:56:32.000000000","tz":120},"subject":"Transfer scale out page to rst format","message":"Transfer scale out page to rst format\n\nSigned-off-by: Ramona Rautenberg \u003crautenberg@osism.tech\u003e\nChange-Id: Icf2d38621f283d02f2d71a17ca2b668db94efd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/4c0c072b8df3ca8f68de2426391b5cc3920eed5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/4c0c072b8df3ca8f68de2426391b5cc3920eed5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[]}
