)]}'
{"id":"openstack%2Flarge-scale~848098","triplet_id":"openstack%2Flarge-scale~master~I6b616407e0bf7fa303a52908f8257651d7f59457","project":"openstack/large-scale","branch":"master","topic":"add/largescale_upgradeandmaintain_documentation","attention_set":{},"removed_from_attention_set":{"34533":{"account":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"last_update":"2022-06-30 08:37:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b616407e0bf7fa303a52908f8257651d7f59457","subject":"Transfer upgrade and maintain page to rst format","status":"MERGED","created":"2022-06-29 11:13:46.000000000","updated":"2022-06-30 08:39:34.000000000","submitted":"2022-06-30 08:37:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848098-add/largescale_upgradeandmaintain_documentation","meta_rev_id":"27af3947edd32c59dced2b3dd1ed564be486f214","_number":848098,"virtual_id_number":848098,"owner":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-30 08:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-06-30 06:28:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-06-30 08:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-06-30 08:27:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 11:17:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 06:28:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-06-30 08:27:56.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"b8c194922fef001e670f3919386f079fad36a8b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"date":"2022-06-29 11:13:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d01b6aefb4cd46f1fd2631ff9427b1841728b075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 11:17:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3392a0e296949afaf70a83e90b0b099 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"795fcfa18c044657aa3ea112419af4b960c24b9a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-30 06:28:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"62111486f186db25590a19365a56b7ff065eac42","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-06-30 08:27:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a81be9c50303bf15fa62e4cee72f3f098ef97747","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:28:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b385ab537bfc56cf32e5385d5872e07549dfc88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:37:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d85edbb84f5476c8e819ed3e03728dd : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e54b8dd2c8c6e08901c5fca1dc754d083febb0c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:37:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"27af3947edd32c59dced2b3dd1ed564be486f214","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:39:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/69afc6391e144a29b41b5ddee0d1b07d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b80378e50dc7c7e6eca0a5812ebe0edf2fe6eac","revisions":{"5b80378e50dc7c7e6eca0a5812ebe0edf2fe6eac":{"kind":"REWORK","_number":1,"created":"2022-06-29 11:13:46.000000000","uploader":{"_account_id":34533,"name":"Ramona Beermann","display_name":"ra-beer","email":"ramona.beermann@osism.tech","username":"ramonarautenberg"},"ref":"refs/changes/98/848098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/98/848098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/98/848098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/98/848098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/98/848098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/98/848098/1"}}},"commit":{"parents":[{"commit":"8fb990ed2d58cfd0116686025436ceb22e3c0b86","subject":"Transfer scale up page to rst format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/8fb990ed2d58cfd0116686025436ceb22e3c0b86"}]}],"author":{"name":"Ramona Rautenberg","email":"rautenberg@osism.tech","date":"2022-06-29 11:12:52.000000000","tz":120},"committer":{"name":"Ramona Rautenberg","email":"rautenberg@osism.tech","date":"2022-06-29 11:12:52.000000000","tz":120},"subject":"Transfer upgrade and maintain page to rst format","message":"Transfer upgrade and maintain page to rst format\n\nSigned-off-by: Ramona Rautenberg \u003crautenberg@osism.tech\u003e\nChange-Id: I6b616407e0bf7fa303a52908f8257651d7f59457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/5b80378e50dc7c7e6eca0a5812ebe0edf2fe6eac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/5b80378e50dc7c7e6eca0a5812ebe0edf2fe6eac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[]}
