)]}'
{"id":"openstack%2Flarge-scale~897546","triplet_id":"openstack%2Flarge-scale~master~I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c","project":"openstack/large-scale","branch":"master","attention_set":{},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-11-11 16:58:40.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-11-04 04:37:25.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-11-11 16:54:41.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c","subject":"stories: discuss various latency issues encountered","status":"MERGED","created":"2023-10-06 13:11:46.000000000","updated":"2023-11-14 16:33:51.000000000","submitted":"2023-11-11 16:58:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"897546","meta_rev_id":"285b3d6b7e4f01302b4f800d0bad9a62f311bcc6","_number":897546,"virtual_id_number":897546,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-11 16:58:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-04 04:37:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-11 16:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-11 16:54:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 13:22:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-06 13:48:19.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2023-10-09 08:30:11.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2023-11-01 07:43:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-04 04:37:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-11-11 16:54:41.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"}],"messages":[{"id":"e2354880b4771e07946858f69cf241fc75ea95c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-10-06 13:11:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b81b22e4d3b126e8f71edae3bef0f0d6d2d6b7b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 13:22:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2aa7ea10f364a2cb39c35eb748927de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13481d4e1f7a488fad8e7811df712aef : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7fe4924e7f223283cf37b1f7da44abfbe648ffe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-10-06 13:34:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6ceb81ed10a9743a457de30cfb9097a979eb9144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-10-06 13:37:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e718a764511a4863d1267a1712ead40d202a2b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 13:38:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64c96be27bda497eb67d6d8d16403d54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddcf8a16a9154d0db63ce7131f65ca84 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"96a9376bd18d11b1a9656148f3f77f2fc3e4ba3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 13:45:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5c47f49b8ae435a8058dd92249e49cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f20ca7014a40fa8ac4d8a184f6363d : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":3},{"id":"9725348bc73efb306ba0baccefe3186d48a3a94f","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-10-06 13:48:19.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fb5e54ca3db1f12eec076950c5c5c9299e3415c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-10-08 17:25:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2a59843711ce87459acbbd0f692f1645bfb0a1a3","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-10-08 17:25:44.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f961120f64ede4b939ccd4f3e277819fdfd6eb84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 17:36:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11483c2f968748c08fd32765f9dc9d5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c977ed1dbe3f4ff3bb8d471f3449f191 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":4},{"id":"5c23041cb18776cbaf3a698645eb09b66f9f15d6","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-10-09 12:24:15.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"1d21f2f05a5a66d8ef771fdf4097bdcfa3988fa2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-10-31 08:57:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8104c2481bda7d1caaffdcb7d54b734325b44008","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-10-31 08:58:26.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"e47114f49747e716b8b5a5ec25e281c4943e9975","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-10-31 08:59:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5b09d752583d5bb481c1ecf10965b9a572771474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:01:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f3ec40e032d497683f64d8f127b2039\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c23494c089465b9e11ad5f5c32dee3 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":5},{"id":"b876b00bd9c65928c8fd8811a98bc290ca2a16a5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-01 07:43:20.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"dec143bebca382c7f3317984bbcc2b9b278a5d82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-11-02 09:13:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"306ecc48edaf5593890c9542dae11211b51eea71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-11-02 09:16:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"206cfcf0bb236f33bf3c0fe42eec992f9b8e4fb6","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-11-02 09:19:00.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ef591daba64f7fc0ea26dab5499e8c6f5d3f032f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 09:27:58.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e31ec105ed30448394b0c672f6532988\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/862de73f47044b558bc2274a16c22404 : FAILURE in 2m 47s","accounts_in_message":[],"_revision_number":7},{"id":"e2b07bd309612322511ce828d70a14855d7e883b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-11-02 09:59:30.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"bc8b09df2d0d5f94b9e7ff6ef52cb1cb92a2cfb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 10:06:40.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be783ce3489b4e59b6f3bc386a231938\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9a010394334e108eacbd264f633284 : FAILURE in 2m 30s","accounts_in_message":[],"_revision_number":8},{"id":"f2acb6c4cb19438da8ad37605f3c519deecbc44c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-11-02 14:09:02.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"3533ec66afd8825b9b439acecdea3dcbe583dd66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 14:13:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8879993a050b470caee24f0033dc83b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78a08c423abd4c79a9954f3a94af9af3 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":9},{"id":"641bab43841fdf3879f64f5c628713ba1af382ad","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-04 04:37:25.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c29c7062b5eb812c509be59de441c40d3479253c","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-11 16:54:41.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cc7c91cb3dfd273b3e43f39c2ae0cd913c8fb308","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-11-11 16:54:46.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"26e3f28a5c899810a3e7ecab0eee5a02899addce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 16:55:04.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"32ed7dff88c6e7c24f3413c90727860e1e94a102","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 16:58:40.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/960ae8554e5e4960bf4fc99461d6cc70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a5899528aba4c36910ab75c61b2094f : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":9},{"id":"b566cbf94f71606043f78e7d664c823394251ca1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 16:58:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"ed8d31a6aa2e3d0c5da5fa777edba7ac90e720bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 16:59:38.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9dca0082aa647de805a9cc60963d8e5\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/c33a7823972141f5bcfd8d826e603e2f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":9},{"id":"285b3d6b7e4f01302b4f800d0bad9a62f311bcc6","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-11-14 16:33:51.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"af1088c46839c26fca92f771fcabbbee35aa0342","revisions":{"baf9b011db79ed9947a1a6b830101adc7bc3bfdd":{"kind":"REWORK","_number":1,"created":"2023-10-06 13:11:46.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/46/897546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/46/897546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/46/897546/1"}}},"commit":{"parents":[{"commit":"9e54292102f90ec05111e4cce1184722af5231fc","subject":"Add information about nova setting update_resources_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9e54292102f90ec05111e4cce1184722af5231fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 09:41:03.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 13:11:45.000000000","tz":120},"subject":"stories: discuss various latency issues encountered","message":"stories: discuss various latency issues encountered\n\nChange-Id: I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/baf9b011db79ed9947a1a6b830101adc7bc3bfdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/baf9b011db79ed9947a1a6b830101adc7bc3bfdd"}]},"branch":"refs/heads/master"},"b589f7fa2ae49af07ef572df842c77367485577e":{"kind":"REWORK","_number":2,"created":"2023-10-06 13:34:27.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/46/897546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/46/897546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/46/897546/2"}}},"commit":{"parents":[{"commit":"9e54292102f90ec05111e4cce1184722af5231fc","subject":"Add information about nova setting update_resources_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9e54292102f90ec05111e4cce1184722af5231fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 09:41:03.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 13:34:20.000000000","tz":120},"subject":"stories: discuss various latency issues encountered","message":"stories: discuss various latency issues encountered\n\nChange-Id: I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/b589f7fa2ae49af07ef572df842c77367485577e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/b589f7fa2ae49af07ef572df842c77367485577e"}]},"branch":"refs/heads/master"},"1e499fc753c19420a47a24f0dd56d352bfe5b381":{"kind":"REWORK","_number":3,"created":"2023-10-06 13:37:59.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/46/897546/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/46/897546/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/46/897546/3"}}},"commit":{"parents":[{"commit":"9e54292102f90ec05111e4cce1184722af5231fc","subject":"Add information about nova setting update_resources_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9e54292102f90ec05111e4cce1184722af5231fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 09:41:03.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 13:37:52.000000000","tz":120},"subject":"stories: discuss various latency issues encountered","message":"stories: discuss various latency issues encountered\n\nChange-Id: I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/1e499fc753c19420a47a24f0dd56d352bfe5b381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/1e499fc753c19420a47a24f0dd56d352bfe5b381"}]},"branch":"refs/heads/master"},"0e13d9e375742eebb5a1ada33cf4b82bbef1f329":{"kind":"REWORK","_number":4,"created":"2023-10-08 17:25:27.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/46/897546/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/46/897546/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/46/897546/4"}}},"commit":{"parents":[{"commit":"9e54292102f90ec05111e4cce1184722af5231fc","subject":"Add information about nova setting update_resources_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9e54292102f90ec05111e4cce1184722af5231fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 09:41:03.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-08 17:25:22.000000000","tz":120},"subject":"stories: discuss various latency issues encountered","message":"stories: discuss various latency issues encountered\n\nChange-Id: I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/0e13d9e375742eebb5a1ada33cf4b82bbef1f329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/0e13d9e375742eebb5a1ada33cf4b82bbef1f329"}]},"branch":"refs/heads/master"},"d17c2806b72ff10a08c8f6e667d11b278e3a7a6a":{"kind":"REWORK","_number":5,"created":"2023-10-31 08:57:04.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/46/897546/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/46/897546/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/46/897546/5"}}},"commit":{"parents":[{"commit":"9e54292102f90ec05111e4cce1184722af5231fc","subject":"Add information about nova setting update_resources_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9e54292102f90ec05111e4cce1184722af5231fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 09:41:03.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-31 08:56:58.000000000","tz":60},"subject":"stories: discuss various latency issues encountered","message":"stories: discuss various latency issues encountered\n\nChange-Id: I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/d17c2806b72ff10a08c8f6e667d11b278e3a7a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/d17c2806b72ff10a08c8f6e667d11b278e3a7a6a"}]},"branch":"refs/heads/master"},"5dcebd91bee6ba81529e573abf936db8a63e126e":{"kind":"REWORK","_number":6,"created":"2023-11-02 09:13:47.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/46/897546/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/46/897546/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/46/897546/6"}}},"commit":{"parents":[{"commit":"9e54292102f90ec05111e4cce1184722af5231fc","subject":"Add information about nova setting update_resources_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9e54292102f90ec05111e4cce1184722af5231fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 09:41:03.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-11-02 09:13:33.000000000","tz":60},"subject":"stories: discuss various latency issues encountered","message":"stories: discuss various latency issues encountered\n\nChange-Id: I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/5dcebd91bee6ba81529e573abf936db8a63e126e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/5dcebd91bee6ba81529e573abf936db8a63e126e"}]},"branch":"refs/heads/master"},"8986640bb751bbab4f3fdf18e1e8dbefad3917db":{"kind":"REWORK","_number":7,"created":"2023-11-02 09:16:03.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/46/897546/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/46/897546/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/46/897546/7"}}},"commit":{"parents":[{"commit":"9e54292102f90ec05111e4cce1184722af5231fc","subject":"Add information about nova setting update_resources_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9e54292102f90ec05111e4cce1184722af5231fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 09:41:03.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-11-02 09:15:52.000000000","tz":60},"subject":"stories: discuss various latency issues encountered","message":"stories: discuss various latency issues encountered\n\nChange-Id: I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/8986640bb751bbab4f3fdf18e1e8dbefad3917db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/8986640bb751bbab4f3fdf18e1e8dbefad3917db"}]},"branch":"refs/heads/master"},"c3da7e6586f272aae63bdcc8691a78cd0cbb4b6f":{"kind":"REWORK","_number":8,"created":"2023-11-02 09:59:30.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/46/897546/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/46/897546/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/46/897546/8"}}},"commit":{"parents":[{"commit":"9e54292102f90ec05111e4cce1184722af5231fc","subject":"Add information about nova setting update_resources_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9e54292102f90ec05111e4cce1184722af5231fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 09:41:03.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-11-02 09:59:22.000000000","tz":60},"subject":"stories: discuss various latency issues encountered","message":"stories: discuss various latency issues encountered\n\nChange-Id: I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/c3da7e6586f272aae63bdcc8691a78cd0cbb4b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/c3da7e6586f272aae63bdcc8691a78cd0cbb4b6f"}]},"branch":"refs/heads/master"},"af1088c46839c26fca92f771fcabbbee35aa0342":{"kind":"REWORK","_number":9,"created":"2023-11-02 14:09:02.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/46/897546/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/46/897546/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/46/897546/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/46/897546/9"}}},"commit":{"parents":[{"commit":"9e54292102f90ec05111e4cce1184722af5231fc","subject":"Add information about nova setting update_resources_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9e54292102f90ec05111e4cce1184722af5231fc"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-10-06 09:41:03.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-11-02 14:08:55.000000000","tz":60},"subject":"stories: discuss various latency issues encountered","message":"stories: discuss various latency issues encountered\n\nChange-Id: I12c73133c67c2ebb64217d6deeb51fd57ffc5d2c\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
