)]}'
{"id":"openstack%2Flarge-scale~907409","triplet_id":"openstack%2Flarge-scale~master~Ife47a45740c35e431a3d72709f6a91fce4c9dab7","project":"openstack/large-scale","branch":"master","topic":"dhcp","attention_set":{},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2024-02-29 13:06:05.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2024-02-28 16:20:39.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Ife47a45740c35e431a3d72709f6a91fce4c9dab7","subject":"neutron: adds elements regardings neutron dhcp agent az aware","status":"MERGED","created":"2024-02-01 11:20:40.000000000","updated":"2024-02-29 13:07:03.000000000","submitted":"2024-02-29 13:06:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907409-dhcp","meta_rev_id":"b209c486626d2b94085a8d9ed29b034fc2ecaf0e","_number":907409,"virtual_id_number":907409,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-29 13:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2024-02-29 12:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-28 16:20:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2024-02-29 12:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 11:24:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-02 10:15:20.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2024-02-28 16:20:39.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2024-02-29 12:58:21.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"51adf9aea6c32293c4eed34fc23a07cabed9f891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-01 11:20:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f35faa518592f52d0c497a0fad37d073b78ead90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 11:24:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0aa0b154c364a2ab919e1d7dbe04447\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8d3c28119e4e3eb01dfc99dbe56277 : FAILURE in 2m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5f14c8aefd1aad6272d8e7297a6223a8f61fcd8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-01 12:53:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d6f3d0387eb2421eda7a2cff2fcab8b14b530e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 13:01:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b7476ff26bd46fdad08bcdb1f962937\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ce51d40378472d91db5869480b6c78 : FAILURE in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"4c416396d9ad8d6543bf7eb9b178b230b837c2f4","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-01 14:07:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7f2a7aac035c3a708a646eb8d1fcb64706e8b78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-01 16:00:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0c09f4539c6d2bedb8755bb7de1cc81fa38eb961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 16:04:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba2803b15c6648278d3081f04fc3d8f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380f436d989d4cbdaac8264d46f59788 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3c5c4bd7379d1e7de456ca728eecff25add06239","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-01 19:40:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97223b8c0918b57e27ec764c31deac50012f49e9","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-02 10:15:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b318c1a98849fd04c29e79f98a0405769a4df162","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-02 10:17:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2aaf99ddcdc859e6db4db8324da0ff7048312967","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-05 08:50:14.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c434e5ffab367ba8c37b76606b2431df5403ef76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-05 13:41:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7186bcee284f8802434c483e741e3aab4b0fc333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 13:44:56.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/069b5646ca104d5d8bb093cf13551fa6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5ee68bb555416f82cd1ca70b275b5d : FAILURE in 2m 40s","accounts_in_message":[],"_revision_number":4},{"id":"7e351f3002e663bb4d5417cf491ca774f1a1682e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-05 14:47:29.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"46a961e6a22ae9fe2c0bfc5e1614f8656ddbe049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 14:51:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b503408fcb88412290a4c2d763e3fb25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02313e43663491bbf8e706d2c633158 : FAILURE in 2m 36s","accounts_in_message":[],"_revision_number":5},{"id":"4671c7bd057d3b32d3f4e7f48278666ae4774b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-06 07:52:31.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bb297d512d60ac169d6c5e74d1652903944ed852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 07:57:42.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdf44f2ab4a74837bddc0cfd4b7d5fdf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2535439b4649420c8b816f44eb0bd651 : FAILURE in 3m 32s","accounts_in_message":[],"_revision_number":6},{"id":"56c0bf8badb528c40b7fab4ad1a59ee7cb192ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-06 09:54:34.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"49f359d6ecaa1dc94ebf8d29b650b484798219f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 10:00:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3617f4264f864c18a26c62e8fa731078\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f16e9192f7ba4b99b8e7916505ca64e0 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":7},{"id":"59eacdd5f8c4f62dcca2c244243a0885f918e5cd","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-09 08:54:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"083fb7fb5c2e334359df536346553680d888173d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-27 10:42:01.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c66f2fd1ac56bb54e091efe339dc464a4c087fee","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2024-02-27 10:42:15.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"80a03112ecb4988388f6314f631a19788d7e30c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 10:50:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/284cf449adbb4ab1af45f3e62eee9d4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9270eef6284940a42e3df527a022cc : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":8},{"id":"741cc657caabbc3974c8b70e296a868ac6c21c6f","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-28 16:20:39.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"de9f04939f3377c2f200f46f4ac277c78228e2bf","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-02-29 12:58:21.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7704993c5d4de29d4044a7aa6f6046fbfc41982e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 12:58:42.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f5b6e9cbce6f2f0f7ff1f02270a6d9375a58f3b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:06:05.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2516496efd34245b0d0a94ced3a4a24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7acada9f08304fe8975ae8e0a884c6a4 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":8},{"id":"00611a43056aa37ff91a98ed25d9792be2053c3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:06:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"b209c486626d2b94085a8d9ed29b034fc2ecaf0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:07:03.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad6924cf8cc14aa0a4579ff09e1da5b0\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/83d2eb03965749a1a72765e0d680b510 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"884a0b774917c5486a3b3832450a8be3558e8b7f","revisions":{"623761acfa2e25d157aa1a2634bbf54ae8fe696d":{"kind":"REWORK","_number":1,"created":"2024-02-01 11:20:40.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/09/907409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/09/907409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/09/907409/1"}}},"commit":{"parents":[{"commit":"af1088c46839c26fca92f771fcabbbee35aa0342","subject":"stories: discuss various latency issues encountered","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 11:18:00.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 11:20:14.000000000","tz":60},"subject":"neutron: adds elements regardings neutron dhcp agent az aware","message":"neutron: adds elements regardings neutron dhcp agent az aware\n\nChange-Id: Ife47a45740c35e431a3d72709f6a91fce4c9dab7\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/623761acfa2e25d157aa1a2634bbf54ae8fe696d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/623761acfa2e25d157aa1a2634bbf54ae8fe696d"}]},"branch":"refs/heads/master"},"24885822e1adc030e24a463920ec0cc0aa90638b":{"kind":"REWORK","_number":2,"created":"2024-02-01 12:53:20.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/09/907409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/09/907409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/09/907409/2"}}},"commit":{"parents":[{"commit":"af1088c46839c26fca92f771fcabbbee35aa0342","subject":"stories: discuss various latency issues encountered","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 11:18:00.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 12:53:13.000000000","tz":60},"subject":"neutron: adds elements regardings neutron dhcp agent az aware","message":"neutron: adds elements regardings neutron dhcp agent az aware\n\nChange-Id: Ife47a45740c35e431a3d72709f6a91fce4c9dab7\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/24885822e1adc030e24a463920ec0cc0aa90638b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/24885822e1adc030e24a463920ec0cc0aa90638b"}]},"branch":"refs/heads/master"},"bc878b988719f7c57f2512a88321e9673581cd44":{"kind":"REWORK","_number":3,"created":"2024-02-01 16:00:02.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/09/907409/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/09/907409/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/09/907409/3"}}},"commit":{"parents":[{"commit":"af1088c46839c26fca92f771fcabbbee35aa0342","subject":"stories: discuss various latency issues encountered","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 11:18:00.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 15:59:55.000000000","tz":60},"subject":"neutron: adds elements regardings neutron dhcp agent az aware","message":"neutron: adds elements regardings neutron dhcp agent az aware\n\nChange-Id: Ife47a45740c35e431a3d72709f6a91fce4c9dab7\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/bc878b988719f7c57f2512a88321e9673581cd44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/bc878b988719f7c57f2512a88321e9673581cd44"}]},"branch":"refs/heads/master"},"9b4e2f94b4efb2407f7c9e8d7702e14d680c2b34":{"kind":"REWORK","_number":4,"created":"2024-02-05 13:41:13.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/09/907409/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/09/907409/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/09/907409/4"}}},"commit":{"parents":[{"commit":"af1088c46839c26fca92f771fcabbbee35aa0342","subject":"stories: discuss various latency issues encountered","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 11:18:00.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-05 13:04:36.000000000","tz":60},"subject":"neutron: adds elements regardings neutron dhcp agent az aware","message":"neutron: adds elements regardings neutron dhcp agent az aware\n\nChange-Id: Ife47a45740c35e431a3d72709f6a91fce4c9dab7\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9b4e2f94b4efb2407f7c9e8d7702e14d680c2b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/9b4e2f94b4efb2407f7c9e8d7702e14d680c2b34"}]},"branch":"refs/heads/master"},"a9308c2fccf769db1c8dce7ddc11e07575f4ee22":{"kind":"REWORK","_number":5,"created":"2024-02-05 14:47:29.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/09/907409/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/09/907409/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/09/907409/5"}}},"commit":{"parents":[{"commit":"af1088c46839c26fca92f771fcabbbee35aa0342","subject":"stories: discuss various latency issues encountered","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 11:18:00.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-05 14:47:22.000000000","tz":60},"subject":"neutron: adds elements regardings neutron dhcp agent az aware","message":"neutron: adds elements regardings neutron dhcp agent az aware\n\nChange-Id: Ife47a45740c35e431a3d72709f6a91fce4c9dab7\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/a9308c2fccf769db1c8dce7ddc11e07575f4ee22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/a9308c2fccf769db1c8dce7ddc11e07575f4ee22"}]},"branch":"refs/heads/master"},"d47a18d806bb3294fe3da744afc1caa2a656b4ba":{"kind":"REWORK","_number":6,"created":"2024-02-06 07:52:31.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/09/907409/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/09/907409/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/09/907409/6"}}},"commit":{"parents":[{"commit":"af1088c46839c26fca92f771fcabbbee35aa0342","subject":"stories: discuss various latency issues encountered","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 11:18:00.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-06 07:52:25.000000000","tz":60},"subject":"neutron: adds elements regardings neutron dhcp agent az aware","message":"neutron: adds elements regardings neutron dhcp agent az aware\n\nChange-Id: Ife47a45740c35e431a3d72709f6a91fce4c9dab7\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/d47a18d806bb3294fe3da744afc1caa2a656b4ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/d47a18d806bb3294fe3da744afc1caa2a656b4ba"}]},"branch":"refs/heads/master"},"1bc68dfa838c2b37b92382f290d62b0e95f641d1":{"kind":"REWORK","_number":7,"created":"2024-02-06 09:54:34.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/09/907409/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/09/907409/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/09/907409/7"}}},"commit":{"parents":[{"commit":"af1088c46839c26fca92f771fcabbbee35aa0342","subject":"stories: discuss various latency issues encountered","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 11:18:00.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-06 09:54:28.000000000","tz":60},"subject":"neutron: adds elements regardings neutron dhcp agent az aware","message":"neutron: adds elements regardings neutron dhcp agent az aware\n\nChange-Id: Ife47a45740c35e431a3d72709f6a91fce4c9dab7\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/1bc68dfa838c2b37b92382f290d62b0e95f641d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/1bc68dfa838c2b37b92382f290d62b0e95f641d1"}]},"branch":"refs/heads/master"},"884a0b774917c5486a3b3832450a8be3558e8b7f":{"kind":"REWORK","_number":8,"created":"2024-02-27 10:42:01.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/09/907409/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/large-scale","ref":"refs/changes/09/907409/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/large-scale refs/changes/09/907409/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/large-scale refs/changes/09/907409/8"}}},"commit":{"parents":[{"commit":"af1088c46839c26fca92f771fcabbbee35aa0342","subject":"stories: discuss various latency issues encountered","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/af1088c46839c26fca92f771fcabbbee35aa0342"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-01 11:18:00.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2024-02-27 10:41:54.000000000","tz":60},"subject":"neutron: adds elements regardings neutron dhcp agent az aware","message":"neutron: adds elements regardings neutron dhcp agent az aware\n\nChange-Id: Ife47a45740c35e431a3d72709f6a91fce4c9dab7\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/884a0b774917c5486a3b3832450a8be3558e8b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/large-scale/commit/884a0b774917c5486a3b3832450a8be3558e8b7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
