)]}'
{"id":"openstack%2Fldappool~996942","triplet_id":"openstack%2Fldappool~master~I4fe1307bd5f2e43c343ec6fda202d967f78615aa","project":"openstack/ldappool","branch":"master","hashtags":[],"change_id":"I4fe1307bd5f2e43c343ec6fda202d967f78615aa","subject":"Remove Python 3.10 support","status":"NEW","created":"2026-07-12 13:53:48.000000000","updated":"2026-07-12 14:03:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c291322257e85651827e2806ea53d7681d93733","_number":996942,"virtual_id_number":996942,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-12 14:03:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-12 14:03:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34928b12ac3d24aa8ec9e8cdcccc76f133579849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 13:53:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbf10164e684536270ab584b1fade931f46736b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 13:55:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c291322257e85651827e2806ea53d7681d93733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 14:03:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d88defb0598047199eea0857e0302001\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/27a641635dd54c7cb2cc5ffb60c628fe : SUCCESS in 2m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d582766e57740b583b79ea318420434 : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4b42970f9994b05996252c14588a369 : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52839e4689c54665883358fcce5b5893 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e84858ab9964cd8ba7c4e1e9d75857f : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a75734897db43118fa8cfd56a95190c : SUCCESS in 4m 59s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c807aecdf0ba4beabd6b4257f0c620f6 : SUCCESS in 4m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/046fb7e0490148ceb2a8ebea042a4a89 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e3602d62c73f4ec0da9b2894a985148a31966904","revisions":{"373c69a04afac8ecb5409d789880121fc0694368":{"kind":"REWORK","_number":1,"created":"2026-07-12 13:53:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/996942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ldappool","ref":"refs/changes/42/996942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ldappool refs/changes/42/996942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ldappool refs/changes/42/996942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ldappool refs/changes/42/996942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ldappool refs/changes/42/996942/1"}}},"commit":{"parents":[{"commit":"ba36471a8d11c5811ee53b1eb9966a6782007dec","subject":"Merge \"Drop flake8-docstrings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ldappool/commit/ba36471a8d11c5811ee53b1eb9966a6782007dec"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 13:52:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 13:53:40.000000000","tz":540},"subject":"Update supported python versions","message":"Update supported python versions\n\n... according to the current tested runtimes.\n\nAlso add the job template to test more recent python versions.\n\nChange-Id: I4fe1307bd5f2e43c343ec6fda202d967f78615aa\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ldappool/commit/373c69a04afac8ecb5409d789880121fc0694368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ldappool/commit/373c69a04afac8ecb5409d789880121fc0694368"}]},"branch":"refs/heads/master"},"e3602d62c73f4ec0da9b2894a985148a31966904":{"kind":"REWORK","_number":2,"created":"2026-07-12 13:55:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/996942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ldappool","ref":"refs/changes/42/996942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ldappool refs/changes/42/996942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ldappool refs/changes/42/996942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ldappool refs/changes/42/996942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ldappool refs/changes/42/996942/2"}}},"commit":{"parents":[{"commit":"04f23ae51dc04ab0133231ec2aa6a2486a104e76","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ldappool/commit/04f23ae51dc04ab0133231ec2aa6a2486a104e76"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 13:54:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 13:55:06.000000000","tz":540},"subject":"Remove Python 3.10 support","message":"Remove Python 3.10 support\n\n... according to the current tested runtimes.\n\nAlso add the job template to test more recent python versions.\n\nChange-Id: I4fe1307bd5f2e43c343ec6fda202d967f78615aa\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ldappool/commit/e3602d62c73f4ec0da9b2894a985148a31966904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ldappool/commit/e3602d62c73f4ec0da9b2894a985148a31966904"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
