)]}'
{"id":"openstack%2Floci~1005783","triplet_id":"openstack%2Floci~master~Ifc1d38aeb0ea14051744ce7ac09c827f224dc85d","project":"openstack/loci","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2026-09-15 22:06:19.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"Ifc1d38aeb0ea14051744ce7ac09c827f224dc85d","subject":"Use only compute-kit jobs for testing images","status":"MERGED","created":"2026-09-15 18:43:47.000000000","updated":"2026-09-15 23:33:19.000000000","submitted":"2026-09-15 23:17:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005783","meta_rev_id":"66246d810e9cf9772240bae92c7ae77faf7be2c7","_number":1005783,"virtual_id_number":1005783,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 23:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"all":[{"value":0,"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":2,"date":"2026-09-15 21:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"value":1,"date":"2026-09-15 22:06:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 20:48:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-15 21:18:47.000000000","updated_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"reviewer":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"state":"REVIEWER"}],"messages":[{"id":"2ac2637d2ea31a59405c53dca6b5867cf70c2bfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-15 18:43:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d573bfc93dc207d7ab316daadb6b7ae89e5dd9a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 20:48:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ed5ff2d542d4c2ba70dfb3e07e79b90\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/97a5441cb5ed4218bf02202ae5b5e678 : SUCCESS in 1h 58m 26s (non-voting)\n- loci-build-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/8baf51009e404896a774d4d4dfabd95e : SUCCESS in 30m 45s\n- loci-compute-kit-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/d0d54c3d6a5d48409f272cab726420fd : SUCCESS in 1h 17m 37s\n- loci-build-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/eb1c15293aca45b09386d01387d603f6 : SUCCESS in 30m 14s\n- loci-compute-kit-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/3fabf4fe769d4d44b5115cd4faaf006e : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4ee8ba53ce5106f8e101d0aaaaaa4c8178c4ab3c","author":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"},"date":"2026-09-15 21:18:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"80e1edae7f42f0ec90b6e71827a37d5baa51c239","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2026-09-15 22:06:19.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac46ac8ee123e56464023df11adc5772359f59d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:07:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8566a644fa44015ddfad904dee95053906324c8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 23:17:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7248488931714a5eacc7665b7293584c\n\n- loci-buildset-registry https://zuul.opendev.org/t/openstack/build/bdced30f14b44ccf85cfe2a90656f6e9 : SUCCESS in 1h 09m 37s (non-voting)\n- loci-upload-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/57a37d5e594142dda37045739db4f1da : SUCCESS in 33m 40s\n- loci-upload-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/1747df3989ce4bb886de040fae14bbde : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":1},{"id":"efd5f392454c484efc98df0ddc88a36d14e609bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 23:17:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66246d810e9cf9772240bae92c7ae77faf7be2c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 23:33:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/399614d4832645e48034b6442ebdb58f\n\n- loci-promote-2025-1-ubuntu_noble https://zuul.opendev.org/t/openstack/build/ad6b175400de468dae0b4c5468d0e38b : SUCCESS in 10m 09s\n- loci-promote-2025-2-ubuntu_noble https://zuul.opendev.org/t/openstack/build/9e0777fec1604df29d7a6563b60afeb3 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e6210c14d4f7a0329708aeaaf7a7a19bc146b1a","revisions":{"3e6210c14d4f7a0329708aeaaf7a7a19bc146b1a":{"kind":"REWORK","_number":1,"created":"2026-09-15 18:43:47.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/83/1005783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/83/1005783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/83/1005783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/83/1005783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/83/1005783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/83/1005783/1"}}},"commit":{"parents":[{"commit":"55ada58cc625755c54d0f956f4246d6cb8bf5aaa","subject":"Remove python-vitrageclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/55ada58cc625755c54d0f956f4246d6cb8bf5aaa"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-09-15 18:43:40.000000000","tz":-300},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2026-09-15 18:43:42.000000000","tz":-300},"subject":"Use only compute-kit jobs for testing images","message":"Use only compute-kit jobs for testing images\n\nThe openstack-helm repo is being updated so there are\nfewer jobs with larger test cases. The compute-kit job\nwill deploy rook and cinder in addition to what it used\nto deploy: nova, neutron, placement, glance, heat.\n\nThis PS drops the loci-cinder-* jobs from the check and\nperiodic pipelines and removes their job definitions so\nnothing refers to the openstack-helm-cinder-*_loci\nparent jobs any more. The loci-promote-* jobs now depend\non the compute-kit jobs only.\n\nThe cinder image itself is still built, uploaded and\npromoted.\n\nSigned-off-by: Vladimir Kozhukalov \u003ckozhukalov@gmail.com\u003e\nChange-Id: Ifc1d38aeb0ea14051744ce7ac09c827f224dc85d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3e6210c14d4f7a0329708aeaaf7a7a19bc146b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3e6210c14d4f7a0329708aeaaf7a7a19bc146b1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34520,"name":"Sergiy Markin","email":"smarkin@mirantis.com","username":"sm515x"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
