)]}'
{"id":"openstack%2Floci~632129","triplet_id":"openstack%2Floci~master~I8f42c6f87263398afb097eb76290dc9a0edea0b8","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I8f42c6f87263398afb097eb76290dc9a0edea0b8","subject":"requirements: do not remove anything from the image","status":"ABANDONED","created":"2019-01-21 16:36:01.000000000","updated":"2019-11-29 11:18:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"605d650a2c8f675ced0f9c7bac1569e83a15b334","_number":632129,"virtual_id_number":632129,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2019-01-24 09:42:52.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":-1,"date":"2019-01-21 17:21:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2019-01-23 21:10:24.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-21 17:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-01-24 09:42:52.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"}],"messages":[{"id":"e6bdb1dec0831aa344e440196c3ae6c9e0695740","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-01-21 16:36:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0668f1d80fde396fa529010717bc75a108894553","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-21 17:21:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/29/632129/1/check/loci-cinder/8e3f024/ : POST_FAILURE in 28m 36s\n- loci-glance http://logs.openstack.org/29/632129/1/check/loci-glance/7bf4f33/ : POST_FAILURE in 35m 37s\n- loci-heat http://logs.openstack.org/29/632129/1/check/loci-heat/630280d/ : POST_FAILURE in 37m 16s\n- loci-horizon http://logs.openstack.org/29/632129/1/check/loci-horizon/757bdf4/ : POST_FAILURE in 32m 52s\n- loci-ironic http://logs.openstack.org/29/632129/1/check/loci-ironic/37fce02/ : POST_FAILURE in 32m 20s\n- loci-keystone http://logs.openstack.org/29/632129/1/check/loci-keystone/f15cb0c/ : POST_FAILURE in 29m 31s\n- loci-neutron http://logs.openstack.org/29/632129/1/check/loci-neutron/802f073/ : POST_FAILURE in 39m 38s\n- loci-nova http://logs.openstack.org/29/632129/1/check/loci-nova/3eafbfa/ : POST_FAILURE in 38m 50s\n- loci-octavia http://logs.openstack.org/29/632129/1/check/loci-octavia/ca7cfd4/ : POST_FAILURE in 31m 46s\n- loci-placement http://logs.openstack.org/29/632129/1/check/loci-placement/0f4797d/ : POST_FAILURE in 26m 18s (non-voting)\n- loci-requirements http://logs.openstack.org/29/632129/1/check/loci-requirements/e63de63/ : POST_FAILURE in 24m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bf252c723cec9b3c88099deb97a2398ec63efdf2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-01-22 16:34:03.000000000","message":"Patch Set 1:\n\nGood catch on the initial issue. I\u0027d rephrase the commit message, as it was a little hard to parse for me :p\n\nWe need to make sure this passes tests too.","accounts_in_message":[],"_revision_number":1},{"id":"23d791a068ef4324ea653fa56db28d1b742515d7","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-01-23 19:34:35.000000000","message":"Patch Set 1: Code-Review-1\n\nDefinitely a problem of removing too much, but I feel this solution goes to the opposite extreme and removes too much.\n\nKeep in mind that every Loci project build re-downloads requirements, which can be considered a bandwidth leak (it helped cause me to overrun ISP quota, so an issue I\u0027m sensitive to).\n\nLet\u0027s look at the criteria we can use to prune the wheels to keep the image minimal but not break architectural compatibility.","accounts_in_message":[],"_revision_number":1},{"id":"04fdb5b28d5231327413062deb2d046162628310","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-01-23 21:10:24.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"abd3bd1f68a5c279042bba044adf5c609b8829a1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-01-23 21:10:47.000000000","message":"Patch Set 1:\n\nwill look into going other way","accounts_in_message":[],"_revision_number":1},{"id":"65d3faddc70d497c60e851ff50b7425f65845cff","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-01-24 09:42:52.000000000","message":"Patch Set 1:\n\n\u003e Definitely a problem of removing too much, but I feel this solution\n \u003e goes to the opposite extreme and removes too much.\n \u003e \n \u003e Keep in mind that every Loci project build re-downloads\n \u003e requirements, which can be considered a bandwidth leak (it helped\n \u003e cause me to overrun ISP quota, so an issue I\u0027m sensitive to).\n \u003e \n \u003e Let\u0027s look at the criteria we can use to prune the wheels to keep\n \u003e the image minimal but not break architectural compatibility.\n\nWe indeed need to be more surgical.","accounts_in_message":[],"_revision_number":1},{"id":"d18b366533ec2656f68cc6775ff1a73412c8613c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-02-11 17:45:02.000000000","message":"Patch Set 1:\n\nOne of solutions would be drop of -P20 and do all in one thread to make sure that we do not have parallel building wheel of the same package.\n\nNow \u0027cffi\u0027 is built twice.","accounts_in_message":[],"_revision_number":1},{"id":"b9b8495c0912e506b1adc4c4dfff64ed97a434ec","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-11-07 21:08:01.000000000","message":"Removed reviewer Chris Hoge with the following votes:\n\n* Code-Review-1 by Chris Hoge \u003cchris@hogepodge.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"605d650a2c8f675ced0f9c7bac1569e83a15b334","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-29 11:18:15.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/688052 does it better","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80c0f056ebd28f1153c829ea5bf40abf15330ef6","revisions":{"80c0f056ebd28f1153c829ea5bf40abf15330ef6":{"kind":"REWORK","_number":1,"created":"2019-01-21 16:36:01.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/632129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/29/632129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/632129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/632129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/29/632129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/29/632129/1"}}},"commit":{"parents":[{"commit":"6c7bddb8b2f3912fcf1ebaf8b79162b46b56f746","subject":"Merge \"Make fetch_wheels.py python3 compatible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6c7bddb8b2f3912fcf1ebaf8b79162b46b56f746"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-01-21 16:35:38.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-01-21 16:35:38.000000000","tz":60},"subject":"requirements: do not remove anything from the image","message":"requirements: do not remove anything from the image\n\nRequirements image is meant to build wheels for all next images. By default\nsome of files are removed. They are supposed to be only those we did not\nbuild...\n\nBut also those which are attempted to be built several times are dropped.\n\nOn AArch64 it means \u0027cffi\u0027 so each image build fails:\n\nCollecting cffi\u003d\u003d\u003d1.11.5 (from -c /tmp/wheels/upper-constraints.txt (line 517))\nCould not find a version that satisfies the requirement cffi\u003d\u003d\u003d1.11.5 (from -c /tmp/wheels/upper-constraints.txt (line 517)) (from versions: )\nNo matching distribution found for cffi\u003d\u003d\u003d1.11.5 (from -c /tmp/wheels/upper-constraints.txt (line 517))\nThe command \u0027/bin/sh -c /opt/loci/scripts/install.sh\u0027 returned a non-zero code: 1\n\nSide effects?\n\n1. requirements image grows from 225 to 316 megabytes\n2. other images nearly do not fetch from pypi\n\nChange-Id: I8f42c6f87263398afb097eb76290dc9a0edea0b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/80c0f056ebd28f1153c829ea5bf40abf15330ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/80c0f056ebd28f1153c829ea5bf40abf15330ef6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
