)]}'
{"id":"openstack%2Floci~655456","triplet_id":"openstack%2Floci~master~I33064d2e8b8a4b460afc89d54e8a2622fd5c7fc2","project":"openstack/loci","branch":"master","topic":"dnm","hashtags":[],"change_id":"I33064d2e8b8a4b460afc89d54e8a2622fd5c7fc2","subject":"DNM -- Try to build with LXD.","status":"ABANDONED","created":"2019-04-24 14:24:30.000000000","updated":"2019-04-25 09:10:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f5947f11c23880117104556c896297e090c46e06","_number":655456,"virtual_id_number":655456,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-04-24 15:59:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 15:59:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee7c0dcc66a6f25b0cd588576e8454e4ffd64a63","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-24 14:24:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4e099edd03b1cb15f9eb306c1fcefc9ca69563d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:36:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/56/655456/1/check/loci-cinder/5746139/ : FAILURE in 2m 17s\n- loci-glance http://logs.openstack.org/56/655456/1/check/loci-glance/fd475b7/ : FAILURE in 2m 36s\n- loci-heat http://logs.openstack.org/56/655456/1/check/loci-heat/bc01f75/ : FAILURE in 2m 59s\n- loci-horizon http://logs.openstack.org/56/655456/1/check/loci-horizon/380c872/ : FAILURE in 3m 05s\n- loci-ironic http://logs.openstack.org/56/655456/1/check/loci-ironic/365ff05/ : FAILURE in 2m 57s\n- loci-keystone http://logs.openstack.org/56/655456/1/check/loci-keystone/1930987/ : FAILURE in 4m 38s\n- loci-neutron http://logs.openstack.org/56/655456/1/check/loci-neutron/564f935/ : FAILURE in 2m 39s\n- loci-nova http://logs.openstack.org/56/655456/1/check/loci-nova/eda0bc6/ : FAILURE in 2m 54s\n- loci-octavia http://logs.openstack.org/56/655456/1/check/loci-octavia/e7458d2/ : FAILURE in 2m 46s\n- loci-placement http://logs.openstack.org/56/655456/1/check/loci-placement/e18d5de/ : FAILURE in 2m 52s (non-voting)\n- loci-requirements http://logs.openstack.org/56/655456/1/check/loci-requirements/a3f75f5/ : FAILURE in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c62a437abde4841cfc446a36c1724b3bdc713410","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-24 14:46:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d1baf8a8344e64bbea2550fde0efd78600beffb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 15:59:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder http://logs.openstack.org/56/655456/2/check/loci-cinder/b7815e6/ : SUCCESS in 58m 54s\n- loci-glance http://logs.openstack.org/56/655456/2/check/loci-glance/d702eb4/ : SUCCESS in 30m 43s\n- loci-heat http://logs.openstack.org/56/655456/2/check/loci-heat/8c42280/ : SUCCESS in 34m 40s\n- loci-horizon http://logs.openstack.org/56/655456/2/check/loci-horizon/faf2480/ : SUCCESS in 42m 05s\n- loci-ironic http://logs.openstack.org/56/655456/2/check/loci-ironic/e1fe980/ : SUCCESS in 38m 21s\n- loci-keystone http://logs.openstack.org/56/655456/2/check/loci-keystone/7038732/ : SUCCESS in 35m 38s\n- loci-neutron http://logs.openstack.org/56/655456/2/check/loci-neutron/dc02b05/ : SUCCESS in 35m 48s\n- loci-nova http://logs.openstack.org/56/655456/2/check/loci-nova/efbfca3/ : SUCCESS in 41m 29s\n- loci-octavia http://logs.openstack.org/56/655456/2/check/loci-octavia/4fa9339/ : SUCCESS in 40m 36s\n- loci-placement http://logs.openstack.org/56/655456/2/check/loci-placement/83a87d4/ : SUCCESS in 44m 32s (non-voting)\n- loci-requirements http://logs.openstack.org/56/655456/2/check/loci-requirements/3cc76be/ : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":2},{"id":"31578bbb8f663662bb16247d3a4c786d4f366fee","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-25 09:10:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9fec13ff39299129d1b7a4ab78d615e6351b8852","revisions":{"1b04685f38d37143f0f008994482a9a8e80f6cc7":{"kind":"REWORK","_number":1,"created":"2019-04-24 14:24:30.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/56/655456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/56/655456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/56/655456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/56/655456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/56/655456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/56/655456/1"}}},"commit":{"parents":[{"commit":"59e3b985526136389a4edabc8f7b3835f5dd6cd7","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/59e3b985526136389a4edabc8f7b3835f5dd6cd7"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-04-24 14:24:05.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-04-24 14:24:05.000000000","tz":120},"subject":"DNM -- Try to build with LXD.","message":"DNM -- Try to build with LXD.\n\nChange-Id: I33064d2e8b8a4b460afc89d54e8a2622fd5c7fc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1b04685f38d37143f0f008994482a9a8e80f6cc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1b04685f38d37143f0f008994482a9a8e80f6cc7"}]},"branch":"refs/heads/master"},"9fec13ff39299129d1b7a4ab78d615e6351b8852":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-24 14:46:23.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/56/655456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/56/655456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/56/655456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/56/655456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/56/655456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/56/655456/2"}}},"commit":{"parents":[{"commit":"086ace64e90e234c915aa19a63a1dc365cc2a590","subject":"Move to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/086ace64e90e234c915aa19a63a1dc365cc2a590"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-04-24 14:24:05.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-04-24 14:46:23.000000000","tz":0},"subject":"DNM -- Try to build with LXD.","message":"DNM -- Try to build with LXD.\n\nChange-Id: I33064d2e8b8a4b460afc89d54e8a2622fd5c7fc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9fec13ff39299129d1b7a4ab78d615e6351b8852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9fec13ff39299129d1b7a4ab78d615e6351b8852"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
