)]}'
{"id":"openstack%2Floci~660493","triplet_id":"openstack%2Floci~master~I16fbc8ec5fd2e246e15f0b391a2ce6f5ced3ff31","project":"openstack/loci","branch":"master","topic":"test","hashtags":[],"change_id":"I16fbc8ec5fd2e246e15f0b391a2ce6f5ced3ff31","subject":"do not merge","status":"ABANDONED","created":"2019-05-21 18:52:03.000000000","updated":"2019-06-07 15:08:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70e5520e4bc849a9af8c64cd42aede4c530e36de","_number":660493,"virtual_id_number":660493,"owner":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-05-21 19:37:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-05-21 18:52:19.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 19:37:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9ac73ed82204c8c0c10441bf27750905d8bb9b2","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-21 18:52:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8937edcaf4eba083601b567fe482d3c525b14f0","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-21 18:52:19.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b39aeae45ed4e4566d59a4f7daa62019178c0bce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 19:37:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/93/660493/1/check/loci-cinder/2e7b20a/ : FAILURE in 27m 05s\n- loci-glance http://logs.openstack.org/93/660493/1/check/loci-glance/fe93c75/ : FAILURE in 31m 28s\n- loci-heat http://logs.openstack.org/93/660493/1/check/loci-heat/102e2b9/ : FAILURE in 27m 14s\n- loci-horizon http://logs.openstack.org/93/660493/1/check/loci-horizon/64993a7/ : FAILURE in 26m 27s\n- loci-ironic http://logs.openstack.org/93/660493/1/check/loci-ironic/38837a3/ : FAILURE in 27m 02s\n- loci-keystone http://logs.openstack.org/93/660493/1/check/loci-keystone/880db38/ : FAILURE in 27m 06s\n- loci-neutron http://logs.openstack.org/93/660493/1/check/loci-neutron/91c87ed/ : FAILURE in 28m 34s\n- loci-nova http://logs.openstack.org/93/660493/1/check/loci-nova/4b364a4/ : FAILURE in 26m 40s\n- loci-octavia http://logs.openstack.org/93/660493/1/check/loci-octavia/bdee5e1/ : FAILURE in 25m 44s\n- loci-placement http://logs.openstack.org/93/660493/1/check/loci-placement/f8e89a9/ : FAILURE in 26m 46s (non-voting)\n- loci-requirements http://logs.openstack.org/93/660493/1/check/loci-requirements/e217427/ : FAILURE in 28m 28s","accounts_in_message":[],"_revision_number":1},{"id":"780e601bcfb3d69ab1061994dc69344d4dedf6b9","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-24 11:56:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73b9cdfdc837a4682e8c0cbedb166748e284bebf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 12:02:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/93/660493/1/check/loci-cinder/780e8be/ : FAILURE in 3m 56s\n- loci-glance http://logs.openstack.org/93/660493/1/check/loci-glance/6a7e340/ : FAILURE in 3m 45s\n- loci-heat http://logs.openstack.org/93/660493/1/check/loci-heat/54e31e3/ : FAILURE in 3m 34s\n- loci-horizon http://logs.openstack.org/93/660493/1/check/loci-horizon/d8e48d2/ : FAILURE in 3m 57s\n- loci-ironic http://logs.openstack.org/93/660493/1/check/loci-ironic/c728df5/ : FAILURE in 3m 48s\n- loci-keystone http://logs.openstack.org/93/660493/1/check/loci-keystone/133d6cc/ : FAILURE in 3m 39s\n- loci-neutron http://logs.openstack.org/93/660493/1/check/loci-neutron/bf9fe31/ : FAILURE in 3m 48s\n- loci-nova http://logs.openstack.org/93/660493/1/check/loci-nova/f791c5c/ : FAILURE in 3m 55s\n- loci-octavia http://logs.openstack.org/93/660493/1/check/loci-octavia/e4865e4/ : FAILURE in 3m 51s\n- loci-placement http://logs.openstack.org/93/660493/1/check/loci-placement/d0db4d2/ : FAILURE in 3m 41s (non-voting)\n- loci-requirements http://logs.openstack.org/93/660493/1/check/loci-requirements/7780be3/ : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"acb8caf8d1a4c67818890ca9a4726837dbdf4689","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-24 17:59:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0e4cbddf66e6d422b7a77e123da6e22754943275","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 18:22:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/93/660493/1/check/loci-cinder/ec9dffa/ : FAILURE in 4m 36s\n- loci-glance http://logs.openstack.org/93/660493/1/check/loci-glance/d256131/ : FAILURE in 3m 51s\n- loci-heat http://logs.openstack.org/93/660493/1/check/loci-heat/a53552f/ : FAILURE in 4m 02s\n- loci-horizon http://logs.openstack.org/93/660493/1/check/loci-horizon/a622c8e/ : FAILURE in 4m 08s\n- loci-ironic http://logs.openstack.org/93/660493/1/check/loci-ironic/e5aead3/ : FAILURE in 4m 00s\n- loci-keystone http://logs.openstack.org/93/660493/1/check/loci-keystone/72bd901/ : FAILURE in 3m 58s\n- loci-neutron http://logs.openstack.org/93/660493/1/check/loci-neutron/f4fa9f4/ : FAILURE in 3m 49s\n- loci-nova http://logs.openstack.org/93/660493/1/check/loci-nova/d3d20ba/ : FAILURE in 4m 41s\n- loci-octavia http://logs.openstack.org/93/660493/1/check/loci-octavia/2fe728c/ : FAILURE in 4m 01s\n- loci-placement http://logs.openstack.org/93/660493/1/check/loci-placement/dc07251/ : FAILURE in 5m 13s (non-voting)\n- loci-requirements http://logs.openstack.org/93/660493/1/check/loci-requirements/8c20a7a/ : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"39903eb65272e936452b3069157e986eaa4b5a86","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-28 15:09:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"58a3ccce8a85fc5fe722d17d691439c6c1a615a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 16:01:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/93/660493/1/check/loci-cinder/3c46c89/ : FAILURE in 27m 29s\n- loci-glance http://logs.openstack.org/93/660493/1/check/loci-glance/f48dac6/ : FAILURE in 32m 27s\n- loci-heat http://logs.openstack.org/93/660493/1/check/loci-heat/cfdcab5/ : FAILURE in 26m 43s\n- loci-horizon http://logs.openstack.org/93/660493/1/check/loci-horizon/91de423/ : FAILURE in 26m 37s\n- loci-ironic http://logs.openstack.org/93/660493/1/check/loci-ironic/494614c/ : FAILURE in 29m 08s\n- loci-keystone http://logs.openstack.org/93/660493/1/check/loci-keystone/09a8880/ : FAILURE in 38m 43s\n- loci-neutron http://logs.openstack.org/93/660493/1/check/loci-neutron/2e04387/ : FAILURE in 29m 55s\n- loci-nova http://logs.openstack.org/93/660493/1/check/loci-nova/42e7048/ : FAILURE in 40m 55s\n- loci-octavia http://logs.openstack.org/93/660493/1/check/loci-octavia/aaafd91/ : FAILURE in 38m 47s\n- loci-placement http://logs.openstack.org/93/660493/1/check/loci-placement/ae4d5b5/ : FAILURE in 29m 59s (non-voting)\n- loci-requirements http://logs.openstack.org/93/660493/1/check/loci-requirements/d9bd136/ : FAILURE in 38m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6327bddece7b672c867ad65e689f09fdd149aa33","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-29 15:15:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"70799cd0208ad561905a12ebf40d22da5f479f9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 16:04:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/93/660493/1/check/loci-cinder/a1297f9/ : FAILURE in 28m 05s\n- loci-glance http://logs.openstack.org/93/660493/1/check/loci-glance/2b4b294/ : FAILURE in 26m 34s\n- loci-heat http://logs.openstack.org/93/660493/1/check/loci-heat/440c812/ : FAILURE in 20m 44s\n- loci-horizon http://logs.openstack.org/93/660493/1/check/loci-horizon/7709b4a/ : FAILURE in 25m 23s\n- loci-ironic http://logs.openstack.org/93/660493/1/check/loci-ironic/32fc7db/ : FAILURE in 25m 19s\n- loci-keystone http://logs.openstack.org/93/660493/1/check/loci-keystone/6dadb44/ : FAILURE in 26m 03s\n- loci-neutron http://logs.openstack.org/93/660493/1/check/loci-neutron/bf9f6e2/ : FAILURE in 28m 52s\n- loci-nova http://logs.openstack.org/93/660493/1/check/loci-nova/a272c24/ : FAILURE in 23m 17s\n- loci-octavia http://logs.openstack.org/93/660493/1/check/loci-octavia/9cb0d18/ : FAILURE in 23m 09s\n- loci-placement http://logs.openstack.org/93/660493/1/check/loci-placement/d3cf63c/ : FAILURE in 27m 21s (non-voting)\n- loci-requirements http://logs.openstack.org/93/660493/1/check/loci-requirements/ffc39d3/ : FAILURE in 27m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4af4c7177db14058b4600a4bebaf2cde494cada8","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2019-05-29 18:48:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7c16bf03f2169214bbb939f787124b056f565bf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 19:46:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/93/660493/1/check/loci-cinder/fca7a38/ : SUCCESS in 35m 00s\n- loci-glance http://logs.openstack.org/93/660493/1/check/loci-glance/6fd2b7e/ : SUCCESS in 38m 37s\n- loci-heat http://logs.openstack.org/93/660493/1/check/loci-heat/dbbe9c6/ : SUCCESS in 36m 53s\n- loci-horizon http://logs.openstack.org/93/660493/1/check/loci-horizon/c4bdaa9/ : SUCCESS in 39m 44s\n- loci-ironic http://logs.openstack.org/93/660493/1/check/loci-ironic/d52d4e7/ : SUCCESS in 34m 08s\n- loci-keystone http://logs.openstack.org/93/660493/1/check/loci-keystone/49a34aa/ : SUCCESS in 35m 02s\n- loci-neutron http://logs.openstack.org/93/660493/1/check/loci-neutron/29f398e/ : SUCCESS in 36m 20s\n- loci-nova http://logs.openstack.org/93/660493/1/check/loci-nova/8095466/ : FAILURE in 33m 33s\n- loci-octavia http://logs.openstack.org/93/660493/1/check/loci-octavia/e37981d/ : SUCCESS in 29m 16s\n- loci-placement http://logs.openstack.org/93/660493/1/check/loci-placement/dbb0584/ : SUCCESS in 37m 59s (non-voting)\n- loci-requirements http://logs.openstack.org/93/660493/1/check/loci-requirements/61bc8cf/ : SUCCESS in 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"72c1416c302fc3e307b1e5cb22b62bbea1733e7f","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-06-07 15:08:40.000000000","message":"Abandoned\n\nRemoving test review.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d2910af2a8f53e75750d77371526c2249c212a4","revisions":{"0d2910af2a8f53e75750d77371526c2249c212a4":{"kind":"REWORK","_number":1,"created":"2019-05-21 18:52:03.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/93/660493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/93/660493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/660493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/660493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/93/660493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/93/660493/1"}}},"commit":{"parents":[{"commit":"b901ac0bed3f3dcc3f3f5d332e966a6608d1091b","subject":"Remove pylxd before 2.2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b901ac0bed3f3dcc3f3f5d332e966a6608d1091b"}]}],"author":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-21 18:52:01.000000000","tz":0},"committer":{"name":"Walter A. Boring IV","email":"waboring@hemna.com","date":"2019-05-21 18:52:30.000000000","tz":0},"subject":"do not merge","message":"do not merge\n\nThis patch is a fake patch to test to see if wheel builds still fails.\n\nChange-Id: I16fbc8ec5fd2e246e15f0b391a2ce6f5ced3ff31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0d2910af2a8f53e75750d77371526c2249c212a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0d2910af2a8f53e75750d77371526c2249c212a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
