)]}'
{"id":"openstack%2Floci~661242","triplet_id":"openstack%2Floci~master~Id001370dc77a4fda50e6a8f724a2d18632f10d17","project":"openstack/loci","branch":"master","topic":"bp/wheels-private-reg-auth-support","hashtags":[],"change_id":"Id001370dc77a4fda50e6a8f724a2d18632f10d17","subject":"Add basic auth support for wheels registry","status":"ABANDONED","created":"2019-05-24 09:07:51.000000000","updated":"2020-08-26 19:26:20.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5da4ec104ab24156c3d5725551cbb1e3df439e97","_number":661242,"virtual_id_number":661242,"owner":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2019-07-01 14:36:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 08:43:22.000000000","updated_by":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"reviewer":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"state":"REVIEWER"},{"updated":"2019-06-03 08:43:22.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2019-06-03 08:43:22.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2019-06-03 08:43:22.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-06-03 08:51:31.000000000","updated_by":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"reviewer":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"state":"REVIEWER"},{"updated":"2019-06-07 19:59:04.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-06-18 08:43:17.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-01 11:02:32.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2019-07-01 14:36:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81f6f28f2878c449d6704de5fd0152ebe01db650","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-05-24 09:07:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29f615b6d0759dfdf916139389b726bdc96ba850","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-05-24 09:28:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"055c3025308141b8214ec4cec9948708dec2a474","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 09:37:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/42/661242/2/check/loci-cinder/1d6f96e/ : FAILURE in 3m 28s\n- loci-glance http://logs.openstack.org/42/661242/2/check/loci-glance/b2fe656/ : FAILURE in 4m 05s\n- loci-heat http://logs.openstack.org/42/661242/2/check/loci-heat/f44c809/ : FAILURE in 3m 36s\n- loci-horizon http://logs.openstack.org/42/661242/2/check/loci-horizon/8df330e/ : FAILURE in 4m 07s\n- loci-ironic http://logs.openstack.org/42/661242/2/check/loci-ironic/f76b1c7/ : FAILURE in 3m 48s\n- loci-keystone http://logs.openstack.org/42/661242/2/check/loci-keystone/db6253f/ : FAILURE in 3m 29s\n- loci-neutron http://logs.openstack.org/42/661242/2/check/loci-neutron/67c02f8/ : FAILURE in 4m 01s\n- loci-nova http://logs.openstack.org/42/661242/2/check/loci-nova/f68dd26/ : FAILURE in 3m 52s\n- loci-octavia http://logs.openstack.org/42/661242/2/check/loci-octavia/c1f2b94/ : FAILURE in 3m 31s\n- loci-placement http://logs.openstack.org/42/661242/2/check/loci-placement/e04495e/ : FAILURE in 3m 52s (non-voting)\n- loci-requirements http://logs.openstack.org/42/661242/2/check/loci-requirements/a1f0a3a/ : FAILURE in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b3da7e2afa8cdeaf5d20a441a8e7a5d2013d9e9b","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-05-24 09:38:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a15cebcc5bb512693496c852d8814e0ed82a0427","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 09:47:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/42/661242/3/check/loci-cinder/a54d51f/ : FAILURE in 3m 52s\n- loci-glance http://logs.openstack.org/42/661242/3/check/loci-glance/7d57fe0/ : FAILURE in 3m 51s\n- loci-heat http://logs.openstack.org/42/661242/3/check/loci-heat/fe164c5/ : FAILURE in 3m 56s\n- loci-horizon http://logs.openstack.org/42/661242/3/check/loci-horizon/02292f0/ : FAILURE in 3m 50s\n- loci-ironic http://logs.openstack.org/42/661242/3/check/loci-ironic/1f8ab18/ : FAILURE in 3m 42s\n- loci-keystone http://logs.openstack.org/42/661242/3/check/loci-keystone/25b2b09/ : FAILURE in 3m 45s\n- loci-neutron http://logs.openstack.org/42/661242/3/check/loci-neutron/288d039/ : FAILURE in 3m 36s\n- loci-nova http://logs.openstack.org/42/661242/3/check/loci-nova/adb2fad/ : FAILURE in 4m 55s\n- loci-octavia http://logs.openstack.org/42/661242/3/check/loci-octavia/d255ce5/ : FAILURE in 3m 47s\n- loci-placement http://logs.openstack.org/42/661242/3/check/loci-placement/d9cebe3/ : FAILURE in 3m 50s (non-voting)\n- loci-requirements http://logs.openstack.org/42/661242/3/check/loci-requirements/8b7af16/ : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":3},{"id":"4adaa448e13704035f3dd222edafb5a0131bdfca","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-05-28 07:39:20.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1755529e8a140b35d7cc02386b531d6258bc78ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 07:59:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/42/661242/4/check/loci-cinder/2e07138/ : FAILURE in 4m 07s\n- loci-glance http://logs.openstack.org/42/661242/4/check/loci-glance/c09f18b/ : FAILURE in 4m 23s\n- loci-heat http://logs.openstack.org/42/661242/4/check/loci-heat/c304618/ : FAILURE in 4m 06s\n- loci-horizon http://logs.openstack.org/42/661242/4/check/loci-horizon/176b51a/ : FAILURE in 4m 48s\n- loci-ironic http://logs.openstack.org/42/661242/4/check/loci-ironic/81e2736/ : FAILURE in 3m 55s\n- loci-keystone http://logs.openstack.org/42/661242/4/check/loci-keystone/187d448/ : FAILURE in 4m 07s\n- loci-neutron http://logs.openstack.org/42/661242/4/check/loci-neutron/839c490/ : FAILURE in 4m 23s\n- loci-nova http://logs.openstack.org/42/661242/4/check/loci-nova/24f881e/ : FAILURE in 4m 21s\n- loci-octavia http://logs.openstack.org/42/661242/4/check/loci-octavia/11cca27/ : FAILURE in 5m 29s\n- loci-placement http://logs.openstack.org/42/661242/4/check/loci-placement/63cc165/ : FAILURE in 4m 06s (non-voting)\n- loci-requirements http://logs.openstack.org/42/661242/4/check/loci-requirements/e5fe714/ : FAILURE in 4m 03s","accounts_in_message":[],"_revision_number":4},{"id":"f8baf1e1f24e966359e8c2006e9750b6bb4fc69d","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-05-28 15:57:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1f520495c4e0e5a867603651a39bae764fe78d7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 16:59:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/42/661242/4/check/loci-cinder/c28359f/ : FAILURE in 29m 42s\n- loci-glance http://logs.openstack.org/42/661242/4/check/loci-glance/67b5c12/ : FAILURE in 31m 16s\n- loci-heat http://logs.openstack.org/42/661242/4/check/loci-heat/567a363/ : FAILURE in 36m 55s\n- loci-horizon http://logs.openstack.org/42/661242/4/check/loci-horizon/cc45e05/ : FAILURE in 34m 17s\n- loci-ironic http://logs.openstack.org/42/661242/4/check/loci-ironic/0912f7d/ : FAILURE in 31m 43s\n- loci-keystone http://logs.openstack.org/42/661242/4/check/loci-keystone/a45a765/ : FAILURE in 33m 16s\n- loci-neutron http://logs.openstack.org/42/661242/4/check/loci-neutron/dac31a7/ : FAILURE in 33m 08s\n- loci-nova http://logs.openstack.org/42/661242/4/check/loci-nova/6806f77/ : FAILURE in 38m 17s\n- loci-octavia http://logs.openstack.org/42/661242/4/check/loci-octavia/a7185d6/ : FAILURE in 33m 43s\n- loci-placement http://logs.openstack.org/42/661242/4/check/loci-placement/dffd405/ : FAILURE in 28m 56s (non-voting)\n- loci-requirements http://logs.openstack.org/42/661242/4/check/loci-requirements/89dac04/ : SUCCESS in 26m 56s","accounts_in_message":[],"_revision_number":4},{"id":"e7754cb0d9c7a9047d2d32aebc1538e7e415bf13","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-05-29 08:08:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"afb7085904bf6cb88dd950611a0601c2da95be08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 09:06:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/42/661242/5/check/loci-cinder/1fcbd91/ : FAILURE in 30m 39s\n- loci-glance http://logs.openstack.org/42/661242/5/check/loci-glance/8869722/ : FAILURE in 28m 19s\n- loci-heat http://logs.openstack.org/42/661242/5/check/loci-heat/3d170ab/ : FAILURE in 24m 57s\n- loci-horizon http://logs.openstack.org/42/661242/5/check/loci-horizon/58a5b3a/ : FAILURE in 25m 23s\n- loci-ironic http://logs.openstack.org/42/661242/5/check/loci-ironic/87ab98c/ : FAILURE in 27m 42s\n- loci-keystone http://logs.openstack.org/42/661242/5/check/loci-keystone/f1c4463/ : FAILURE in 28m 12s\n- loci-neutron http://logs.openstack.org/42/661242/5/check/loci-neutron/03d15e1/ : FAILURE in 26m 58s\n- loci-nova http://logs.openstack.org/42/661242/5/check/loci-nova/2777aa7/ : FAILURE in 27m 04s\n- loci-octavia http://logs.openstack.org/42/661242/5/check/loci-octavia/f5bfc67/ : FAILURE in 27m 37s\n- loci-placement http://logs.openstack.org/42/661242/5/check/loci-placement/17ee0e1/ : FAILURE in 24m 27s (non-voting)\n- loci-requirements http://logs.openstack.org/42/661242/5/check/loci-requirements/8d94ea0/ : FAILURE in 25m 07s","accounts_in_message":[],"_revision_number":5},{"id":"23148dffd8458567ad1c9953dff91dba98910dfb","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-06-01 01:37:33.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"d4521282f59d479e479525f3089603ad8e542e9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 02:21:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder http://logs.openstack.org/42/661242/6/check/loci-cinder/45edc5c/ : SUCCESS in 31m 47s\n- loci-glance http://logs.openstack.org/42/661242/6/check/loci-glance/2f46382/ : SUCCESS in 33m 41s\n- loci-heat http://logs.openstack.org/42/661242/6/check/loci-heat/f81822b/ : SUCCESS in 39m 08s\n- loci-horizon http://logs.openstack.org/42/661242/6/check/loci-horizon/5cf3108/ : SUCCESS in 33m 59s\n- loci-ironic http://logs.openstack.org/42/661242/6/check/loci-ironic/0e66d06/ : SUCCESS in 29m 39s\n- loci-keystone http://logs.openstack.org/42/661242/6/check/loci-keystone/cf5e8e9/ : SUCCESS in 38m 26s\n- loci-neutron http://logs.openstack.org/42/661242/6/check/loci-neutron/032a790/ : SUCCESS in 30m 44s\n- loci-nova http://logs.openstack.org/42/661242/6/check/loci-nova/f0a1970/ : SUCCESS in 38m 32s\n- loci-octavia http://logs.openstack.org/42/661242/6/check/loci-octavia/653b750/ : SUCCESS in 32m 50s\n- loci-placement http://logs.openstack.org/42/661242/6/check/loci-placement/264dec0/ : SUCCESS in 29m 39s (non-voting)\n- loci-requirements http://logs.openstack.org/42/661242/6/check/loci-requirements/407a8d7/ : SUCCESS in 21m 01s","accounts_in_message":[],"_revision_number":6},{"id":"48582bd75981c823a82c77fb43097176cbf21ca2","author":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"date":"2019-06-03 08:51:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"93cd12380c3edfd28db2fd9520c401229e1bec5f","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-06-07 00:28:51.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"d0e1b97784a5fe31dd108ed30d3751e85bf46785","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-07 15:47:55.000000000","message":"Patch Set 6: Code-Review+1\n\nI am fine with this. However I would have preferred if this was tested using, for example, the infra roles which provide a secure buildset registry. Alternatively, setting up a job with a secure registry would be fine too.","accounts_in_message":[],"_revision_number":6},{"id":"219a9535b989948cf64c3aec4d59042058f2df54","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-06-07 19:59:04.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"fcf6c309d6ae18b5afa81d139fe25bac75e67680","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-06-13 08:53:54.000000000","message":"Patch Set 6: Code-Review-1\n\nDNM\n\nusing secrets in docker build args leaks the secret in the image manifest\n\nsee: https://stackoverflow.com/questions/33621242/why-is-arg-in-a-dockerfile-not-recommended-for-passing-secrets#33623649\n\nwe should use the new buildkit \u0027secret\u0027 option instead, which requires a rework of the PR","accounts_in_message":[],"_revision_number":6},{"id":"7699e24dd19e5059177415459475c4251b2f5ac1","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-06-13 08:54:09.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"5dce647cd78c46783da54bc45074e18e99447483","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-13 12:53:28.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"728e6f2f82110da73c443d6c0c63b81de0527ff5","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-06-13 15:02:06.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"5517aafebec3e731958b63664178a38fd664c798","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-06-14 14:54:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2b1ef57120737944b9cb63484c60c0d7691fb273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 16:49:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder http://logs.openstack.org/42/661242/7/check/loci-cinder/3c52912/ : SUCCESS in 32m 05s\n- loci-glance http://logs.openstack.org/42/661242/7/check/loci-glance/288ca82/ : SUCCESS in 30m 14s\n- loci-heat http://logs.openstack.org/42/661242/7/check/loci-heat/ca35487/ : SUCCESS in 35m 06s\n- loci-horizon http://logs.openstack.org/42/661242/7/check/loci-horizon/88e1ea9/ : SUCCESS in 33m 46s\n- loci-ironic http://logs.openstack.org/42/661242/7/check/loci-ironic/87f76a0/ : SUCCESS in 31m 54s\n- loci-keystone http://logs.openstack.org/42/661242/7/check/loci-keystone/906a09b/ : SUCCESS in 34m 33s\n- loci-neutron http://logs.openstack.org/42/661242/7/check/loci-neutron/d6db026/ : SUCCESS in 37m 51s\n- loci-nova http://logs.openstack.org/42/661242/7/check/loci-nova/3fb34b1/ : SUCCESS in 37m 47s\n- loci-octavia http://logs.openstack.org/42/661242/7/check/loci-octavia/22caa63/ : SUCCESS in 35m 09s\n- loci-placement http://logs.openstack.org/42/661242/7/check/loci-placement/a9cc3ef/ : SUCCESS in 27m 28s (non-voting)\n- loci-requirements http://logs.openstack.org/42/661242/7/check/loci-requirements/a70e843/ : SUCCESS in 23m 17s","accounts_in_message":[],"_revision_number":7},{"id":"e76ef2a61352b69b892fbb545a5815b332b13b3b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-18 08:43:17.000000000","message":"Patch Set 7:\n\nI like this better. I think there are a few whitespaces and [WIP] parts of the documentation that needs to be removed, but except that, I like it.","accounts_in_message":[],"_revision_number":7},{"id":"97b2cebb867bfed4fd658a621c84eb27d8319640","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-07-01 11:02:32.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b90b23c5707792971d95ba38372cc6ec001b2016","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-07-01 12:25:18.000000000","message":"Patch Set 7:\n\nhi vlad\n\nthanks for your review\n\nthis is still WIP, we\u0027ve noticed the same issues as you reported\n\n( using a harbor private registry, and not gitlab )\n\nand we\u0027ll come with a new patchset soon","accounts_in_message":[],"_revision_number":7},{"id":"3ebe255f1eada84ad487bab73107fcb2fc478589","author":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"date":"2019-07-01 13:50:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"451af45867064d16916737ae3d20c40167b5a266","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 14:36:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder http://logs.openstack.org/42/661242/8/check/loci-cinder/d57ea8e/ : SUCCESS in 37m 46s\n- loci-glance http://logs.openstack.org/42/661242/8/check/loci-glance/d409439/ : SUCCESS in 36m 24s\n- loci-heat http://logs.openstack.org/42/661242/8/check/loci-heat/78c11f1/ : SUCCESS in 32m 05s\n- loci-horizon http://logs.openstack.org/42/661242/8/check/loci-horizon/5dc68cf/ : SUCCESS in 34m 44s\n- loci-ironic http://logs.openstack.org/42/661242/8/check/loci-ironic/446382b/ : SUCCESS in 36m 01s\n- loci-keystone http://logs.openstack.org/42/661242/8/check/loci-keystone/4017ab1/ : SUCCESS in 35m 13s\n- loci-neutron http://logs.openstack.org/42/661242/8/check/loci-neutron/88311e7/ : SUCCESS in 34m 35s\n- loci-nova http://logs.openstack.org/42/661242/8/check/loci-nova/8f92b11/ : SUCCESS in 39m 11s\n- loci-octavia http://logs.openstack.org/42/661242/8/check/loci-octavia/21c0f81/ : SUCCESS in 36m 45s\n- loci-placement http://logs.openstack.org/42/661242/8/check/loci-placement/73a37e8/ : SUCCESS in 34m 13s (non-voting)\n- loci-requirements http://logs.openstack.org/42/661242/8/check/loci-requirements/d9e0f91/ : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":8},{"id":"717ab97abc97f223bd4c2a840b7c42ba3d636b41","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-08-26 19:26:20.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"063f243a6700a87023615389f402b852e726ad36","revisions":{"4e23e0150295ad1899e3a5b91d8a9e40c2c2de1a":{"kind":"REWORK","_number":1,"created":"2019-05-24 09:07:51.000000000","uploader":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"ref":"refs/changes/42/661242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/42/661242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/42/661242/1"}}},"commit":{"parents":[{"commit":"b901ac0bed3f3dcc3f3f5d332e966a6608d1091b","subject":"Remove pylxd before 2.2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b901ac0bed3f3dcc3f3f5d332e966a6608d1091b"}]}],"author":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:07:32.000000000","tz":0},"committer":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:07:32.000000000","tz":0},"subject":"Add basic auth support for wheels registry","message":"Add basic auth support for wheels registry\n\nChange-Id: Id001370dc77a4fda50e6a8f724a2d18632f10d17\nImplements: blueprint wheels-private-reg-auth-support\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4e23e0150295ad1899e3a5b91d8a9e40c2c2de1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4e23e0150295ad1899e3a5b91d8a9e40c2c2de1a"}]},"branch":"refs/heads/master"},"4aec8c90d8e5efbe6423bc4c35c70a6453bf0573":{"kind":"REWORK","_number":2,"created":"2019-05-24 09:28:04.000000000","uploader":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"ref":"refs/changes/42/661242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/42/661242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/42/661242/2"}}},"commit":{"parents":[{"commit":"b901ac0bed3f3dcc3f3f5d332e966a6608d1091b","subject":"Remove pylxd before 2.2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b901ac0bed3f3dcc3f3f5d332e966a6608d1091b"}]}],"author":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:07:32.000000000","tz":0},"committer":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:27:50.000000000","tz":0},"subject":"Add basic auth support for wheels registry","message":"Add basic auth support for wheels registry\n\nChange-Id: Id001370dc77a4fda50e6a8f724a2d18632f10d17\nImplements: blueprint wheels-private-reg-auth-support\nSigned-off-by: ydegat \u003cyann@2kmail.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4aec8c90d8e5efbe6423bc4c35c70a6453bf0573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4aec8c90d8e5efbe6423bc4c35c70a6453bf0573"}]},"branch":"refs/heads/master"},"b29fb8e56c35e16a94977411b502d25869e4b1b4":{"kind":"REWORK","_number":3,"created":"2019-05-24 09:38:42.000000000","uploader":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"ref":"refs/changes/42/661242/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/42/661242/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/42/661242/3"}}},"commit":{"parents":[{"commit":"b901ac0bed3f3dcc3f3f5d332e966a6608d1091b","subject":"Remove pylxd before 2.2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b901ac0bed3f3dcc3f3f5d332e966a6608d1091b"}]}],"author":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:07:32.000000000","tz":0},"committer":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:38:29.000000000","tz":0},"subject":"Add basic auth support for wheels registry","message":"Add basic auth support for wheels registry\n\nChange-Id: Id001370dc77a4fda50e6a8f724a2d18632f10d17\nImplements: blueprint wheels-private-reg-auth-support\nSigned-off-by: ydegat \u003cyann@2kmail.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b29fb8e56c35e16a94977411b502d25869e4b1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b29fb8e56c35e16a94977411b502d25869e4b1b4"}]},"branch":"refs/heads/master"},"5f2caed666cd9c08f0aee855f8dff3dfe21c2f59":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-28 07:39:20.000000000","uploader":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"ref":"refs/changes/42/661242/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/42/661242/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/42/661242/4"}}},"commit":{"parents":[{"commit":"97ef1b74be8c8fe980a01b504f163b7a15a0d532","subject":"Remove pywbem to blacklist M2Crypto","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/97ef1b74be8c8fe980a01b504f163b7a15a0d532"}]}],"author":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:07:32.000000000","tz":0},"committer":{"name":"yann degat","email":"yann@2kmail.net","date":"2019-05-28 07:39:20.000000000","tz":0},"subject":"Add basic auth support for wheels registry","message":"Add basic auth support for wheels registry\n\nChange-Id: Id001370dc77a4fda50e6a8f724a2d18632f10d17\nImplements: blueprint wheels-private-reg-auth-support\nSigned-off-by: ydegat \u003cyann@2kmail.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5f2caed666cd9c08f0aee855f8dff3dfe21c2f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5f2caed666cd9c08f0aee855f8dff3dfe21c2f59"}]},"branch":"refs/heads/master"},"e5e81c23a3cc57a006f2cac8f1033a93a67aa312":{"kind":"REWORK","_number":5,"created":"2019-05-29 08:08:31.000000000","uploader":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"ref":"refs/changes/42/661242/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/42/661242/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/42/661242/5"}}},"commit":{"parents":[{"commit":"b901ac0bed3f3dcc3f3f5d332e966a6608d1091b","subject":"Remove pylxd before 2.2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b901ac0bed3f3dcc3f3f5d332e966a6608d1091b"}]}],"author":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:07:32.000000000","tz":0},"committer":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-29 08:08:09.000000000","tz":0},"subject":"Add basic auth support for wheels registry","message":"Add basic auth support for wheels registry\n\nChange-Id: Id001370dc77a4fda50e6a8f724a2d18632f10d17\nImplements: blueprint wheels-private-reg-auth-support\nSigned-off-by: ydegat \u003cyann@2kmail.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e5e81c23a3cc57a006f2cac8f1033a93a67aa312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e5e81c23a3cc57a006f2cac8f1033a93a67aa312"}]},"branch":"refs/heads/master"},"2ed8f04a94af40e25b250b3ad9db496da17410cf":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-01 01:37:33.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/42/661242/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/42/661242/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/42/661242/6"}}},"commit":{"parents":[{"commit":"2a2ddb832ef41fd921b1f6d5100559e1914fd031","subject":"Add qemu profile for Glance into the bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2a2ddb832ef41fd921b1f6d5100559e1914fd031"}]}],"author":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:07:32.000000000","tz":0},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-06-01 01:37:33.000000000","tz":0},"subject":"Add basic auth support for wheels registry","message":"Add basic auth support for wheels registry\n\nChange-Id: Id001370dc77a4fda50e6a8f724a2d18632f10d17\nImplements: blueprint wheels-private-reg-auth-support\nSigned-off-by: ydegat \u003cyann@2kmail.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2ed8f04a94af40e25b250b3ad9db496da17410cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2ed8f04a94af40e25b250b3ad9db496da17410cf"}]},"branch":"refs/heads/master"},"6cc62089b3efbc2d2174c57ae9a2add68b241b15":{"kind":"REWORK","_number":7,"created":"2019-06-14 14:54:29.000000000","uploader":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"ref":"refs/changes/42/661242/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/42/661242/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/42/661242/7"}}},"commit":{"parents":[{"commit":"ac2b5a80adf997ba0d398ba40d6eec7de11e4299","subject":"Updated README.md to reflect need to build base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac2b5a80adf997ba0d398ba40d6eec7de11e4299"}]}],"author":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:07:32.000000000","tz":0},"committer":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-06-14 14:54:14.000000000","tz":0},"subject":"Add basic auth support for wheels registry","message":"Add basic auth support for wheels registry\n\nChange-Id: Id001370dc77a4fda50e6a8f724a2d18632f10d17\nImplements: blueprint wheels-private-reg-auth-support\nSigned-off-by: ydegat \u003cyann@2kmail.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6cc62089b3efbc2d2174c57ae9a2add68b241b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6cc62089b3efbc2d2174c57ae9a2add68b241b15"}]},"branch":"refs/heads/master"},"063f243a6700a87023615389f402b852e726ad36":{"kind":"REWORK","_number":8,"created":"2019-07-01 13:50:52.000000000","uploader":{"_account_id":30385,"name":"yann degat","email":"yann@2kmail.net","username":"ydegat"},"ref":"refs/changes/42/661242/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/42/661242/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/42/661242/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/42/661242/8"}}},"commit":{"parents":[{"commit":"ac2b5a80adf997ba0d398ba40d6eec7de11e4299","subject":"Updated README.md to reflect need to build base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac2b5a80adf997ba0d398ba40d6eec7de11e4299"}]}],"author":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-05-24 09:07:32.000000000","tz":0},"committer":{"name":"ydegat","email":"yann@2kmail.net","date":"2019-07-01 13:50:38.000000000","tz":0},"subject":"Add basic auth support for wheels registry","message":"Add basic auth support for wheels registry\n\nChange-Id: Id001370dc77a4fda50e6a8f724a2d18632f10d17\nImplements: blueprint wheels-private-reg-auth-support\nSigned-off-by: ydegat \u003cyann@2kmail.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/063f243a6700a87023615389f402b852e726ad36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/063f243a6700a87023615389f402b852e726ad36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
