)]}'
{"id":"openstack%2Floci~662043","triplet_id":"openstack%2Floci~master~Iaadaf924bf38465ca70f685cb0e80d1e36c29539","project":"openstack/loci","branch":"master","topic":"cleanup","hashtags":[],"change_id":"Iaadaf924bf38465ca70f685cb0e80d1e36c29539","subject":"Move comment to correct location","status":"MERGED","created":"2019-05-29 17:00:54.000000000","updated":"2019-06-07 15:32:44.000000000","submitted":"2019-06-07 15:32:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662043-1559921564945-2d20edbd","meta_rev_id":"ad62777d96fa9c26b7ceabfae445caab1fb7d735","_number":662043,"virtual_id_number":662043,"owner":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-07 15:32:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":2,"date":"2019-05-29 22:13:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 14:46:33.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"all":[{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-07 14:46:33.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 17:21:33.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-07 14:46:33.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-06-07 15:32:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e61c23c4d1d0ea37f3a7e15670bb0716e0b531ce","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2019-05-29 17:00:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5eeccd6730d40e55eaa66d66fe6c273e0e553ed0","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-29 17:21:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"76485271d7a38051c700a06a54c613645379f730","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 17:47:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/43/662043/1/check/loci-cinder/3b1cdb4/ : FAILURE in 26m 25s\n- loci-glance http://logs.openstack.org/43/662043/1/check/loci-glance/54259a1/ : FAILURE in 29m 37s\n- loci-heat http://logs.openstack.org/43/662043/1/check/loci-heat/8253ecc/ : FAILURE in 24m 57s\n- loci-horizon http://logs.openstack.org/43/662043/1/check/loci-horizon/cec503f/ : FAILURE in 30m 24s\n- loci-ironic http://logs.openstack.org/43/662043/1/check/loci-ironic/d6f31de/ : FAILURE in 28m 50s\n- loci-keystone http://logs.openstack.org/43/662043/1/check/loci-keystone/2fa4d27/ : FAILURE in 28m 46s\n- loci-neutron http://logs.openstack.org/43/662043/1/check/loci-neutron/43c7e73/ : FAILURE in 24m 12s\n- loci-nova http://logs.openstack.org/43/662043/1/check/loci-nova/c7c140f/ : FAILURE in 26m 49s\n- loci-octavia http://logs.openstack.org/43/662043/1/check/loci-octavia/0f6e4a4/ : FAILURE in 28m 08s\n- loci-placement http://logs.openstack.org/43/662043/1/check/loci-placement/c13ae09/ : FAILURE in 29m 48s (non-voting)\n- loci-requirements http://logs.openstack.org/43/662043/1/check/loci-requirements/d68b893/ : FAILURE in 26m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5ecdb80b0105a6b2f8807cfd2f0488cc847602c0","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2019-05-29 22:13:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c28dc28077390d13befee0f6b68caace558ba871","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 00:02:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder http://logs.openstack.org/43/662043/2/check/loci-cinder/c340022/ : SUCCESS in 33m 42s\n- loci-glance http://logs.openstack.org/43/662043/2/check/loci-glance/5b6cc9f/ : SUCCESS in 34m 06s\n- loci-heat http://logs.openstack.org/43/662043/2/check/loci-heat/c631c6a/ : SUCCESS in 34m 53s\n- loci-horizon http://logs.openstack.org/43/662043/2/check/loci-horizon/42a377e/ : SUCCESS in 39m 53s\n- loci-ironic http://logs.openstack.org/43/662043/2/check/loci-ironic/d2fbcc7/ : SUCCESS in 37m 05s\n- loci-keystone http://logs.openstack.org/43/662043/2/check/loci-keystone/e940b73/ : SUCCESS in 34m 34s\n- loci-neutron http://logs.openstack.org/43/662043/2/check/loci-neutron/c410323/ : SUCCESS in 38m 28s\n- loci-nova http://logs.openstack.org/43/662043/2/check/loci-nova/e05eaf7/ : SUCCESS in 41m 09s\n- loci-octavia http://logs.openstack.org/43/662043/2/check/loci-octavia/9dab8c0/ : SUCCESS in 33m 49s\n- loci-placement http://logs.openstack.org/43/662043/2/check/loci-placement/74e1fe9/ : SUCCESS in 33m 24s (non-voting)\n- loci-requirements http://logs.openstack.org/43/662043/2/check/loci-requirements/bc74313/ : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a9654cbdecc246603a68e5415b34e3bf9a905b8a","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-06-07 14:46:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3ccfff3e787706460f5fdf156947f2e5fdeb9621","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:46:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"394732bd504a41ffd85b06ef33ac71397a637424","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:32:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder http://logs.openstack.org/43/662043/2/gate/loci-cinder/be0ef4f/ : SUCCESS in 36m 05s\n- loci-glance http://logs.openstack.org/43/662043/2/gate/loci-glance/89eedc5/ : SUCCESS in 34m 18s\n- loci-heat http://logs.openstack.org/43/662043/2/gate/loci-heat/13ec85f/ : SUCCESS in 36m 19s\n- loci-horizon http://logs.openstack.org/43/662043/2/gate/loci-horizon/68e5dcb/ : SUCCESS in 40m 00s\n- loci-ironic http://logs.openstack.org/43/662043/2/gate/loci-ironic/882ef6e/ : SUCCESS in 36m 30s\n- loci-keystone http://logs.openstack.org/43/662043/2/gate/loci-keystone/ed27f43/ : SUCCESS in 34m 09s\n- loci-neutron http://logs.openstack.org/43/662043/2/gate/loci-neutron/36e684f/ : SUCCESS in 34m 12s\n- loci-nova http://logs.openstack.org/43/662043/2/gate/loci-nova/0345e02/ : SUCCESS in 40m 02s\n- loci-octavia http://logs.openstack.org/43/662043/2/gate/loci-octavia/24662cd/ : SUCCESS in 37m 58s\n- loci-requirements http://logs.openstack.org/43/662043/2/gate/loci-requirements/49fbd6d/ : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":2},{"id":"51ab503bc8eb8015abc7e5e5df5378c23b78352b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:32:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"370eecee18ca53e313b0bd09b8c597ce0df3f43a","revisions":{"0a7ab74b93e297880830a57a193a2949debcfa70":{"kind":"REWORK","_number":1,"created":"2019-05-29 17:00:54.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/43/662043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/43/662043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/43/662043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/43/662043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/43/662043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/43/662043/1"}}},"commit":{"parents":[{"commit":"f0ab8faf035f165b4016c24988a5f14eceb5dd10","subject":"Remove unused vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f0ab8faf035f165b4016c24988a5f14eceb5dd10"}]}],"author":{"name":"Sam Yaple","email":"sam+git@yaple.net","date":"2019-05-29 16:57:08.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam+git@yaple.net","date":"2019-05-29 17:00:29.000000000","tz":-240},"subject":"Move comment to correct location","message":"Move comment to correct location\n\nGit blame shows this not shifting over time, moving it to make more\nsense\n\nChange-Id: Iaadaf924bf38465ca70f685cb0e80d1e36c29539\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0a7ab74b93e297880830a57a193a2949debcfa70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0a7ab74b93e297880830a57a193a2949debcfa70"}]},"branch":"refs/heads/master"},"370eecee18ca53e313b0bd09b8c597ce0df3f43a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-29 22:13:30.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/43/662043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/43/662043/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/43/662043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/43/662043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/43/662043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/43/662043/2"}}},"commit":{"parents":[{"commit":"87d96f89b1700f5ef19d325859bf9bc094e4c638","subject":"Remove unused vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/87d96f89b1700f5ef19d325859bf9bc094e4c638"}]}],"author":{"name":"Sam Yaple","email":"sam+git@yaple.net","date":"2019-05-29 16:57:08.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam+git@yaple.net","date":"2019-05-29 22:13:21.000000000","tz":-240},"subject":"Move comment to correct location","message":"Move comment to correct location\n\nGit blame shows this not shifting over time, moving it to make more\nsense\n\nChange-Id: Iaadaf924bf38465ca70f685cb0e80d1e36c29539\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/370eecee18ca53e313b0bd09b8c597ce0df3f43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/370eecee18ca53e313b0bd09b8c597ce0df3f43a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
