)]}'
{"id":"openstack%2Floci~664632","triplet_id":"openstack%2Floci~master~I9bd41f116978e501f0dec9e52f870c81224c0660","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I9bd41f116978e501f0dec9e52f870c81224c0660","subject":"Updated README.md to reflect need to build base image","status":"MERGED","created":"2019-06-11 16:22:05.000000000","updated":"2019-06-12 12:31:00.000000000","submitted":"2019-06-12 12:31:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664632-1560342660915-43f1f006","meta_rev_id":"27c264e3601f5b04da72b91c804b943e7fcf7f9e","_number":664632,"virtual_id_number":664632,"owner":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-12 12:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":2,"date":"2019-06-11 16:33:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2019-06-12 10:15:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-06-12 06:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":1,"date":"2019-06-12 11:34:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 06:31:33.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-12 10:15:17.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-06-12 11:34:17.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-06-12 12:31:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a4b8e285ea82f91242c6379f9127a2ad6408194","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-06-11 16:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1eb2bb8500827b0b5f738472986dbb47a9b3850e","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-06-11 16:33:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e0421af9cf25ea454dcb9130b3261f3540ede1c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 17:12:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/32/664632/1/check/loci-cinder/83a27ac/ : SUCCESS in 36m 28s\n- loci-glance http://logs.openstack.org/32/664632/1/check/loci-glance/8514ed6/ : SUCCESS in 33m 20s\n- loci-heat http://logs.openstack.org/32/664632/1/check/loci-heat/cb09bf2/ : SUCCESS in 35m 31s\n- loci-horizon http://logs.openstack.org/32/664632/1/check/loci-horizon/3a44d4a/ : SUCCESS in 37m 45s\n- loci-ironic http://logs.openstack.org/32/664632/1/check/loci-ironic/19ba1be/ : SUCCESS in 34m 52s\n- loci-keystone http://logs.openstack.org/32/664632/1/check/loci-keystone/a5ab763/ : FAILURE in 24m 09s\n- loci-neutron http://logs.openstack.org/32/664632/1/check/loci-neutron/5d883aa/ : FAILURE in 22m 25s\n- loci-nova http://logs.openstack.org/32/664632/1/check/loci-nova/e720b71/ : FAILURE in 22m 27s\n- loci-octavia http://logs.openstack.org/32/664632/1/check/loci-octavia/dbef829/ : FAILURE in 22m 16s\n- loci-placement http://logs.openstack.org/32/664632/1/check/loci-placement/d02b88d/ : SUCCESS in 33m 11s (non-voting)\n- loci-requirements http://logs.openstack.org/32/664632/1/check/loci-requirements/cb0e4fa/ : SUCCESS in 23m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0a775fbffafad6a310ded511aa5a9497a2dd3813","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-12 06:31:24.000000000","message":"Patch Set 1:\n\nOk for me. I think it would be nice to explain the requirements image later too.","accounts_in_message":[],"_revision_number":1},{"id":"403b8b1b209578f798a1fdf3d21dc5ad19088738","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-12 06:31:33.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f05a07f9aca3731f5f3f0fab182c7def88e946b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:59:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder http://logs.openstack.org/32/664632/1/check/loci-cinder/d29f002/ : SUCCESS in 38m 01s\n- loci-glance http://logs.openstack.org/32/664632/1/check/loci-glance/21f40c7/ : SUCCESS in 42m 28s\n- loci-heat http://logs.openstack.org/32/664632/1/check/loci-heat/e46b1bb/ : SUCCESS in 31m 37s\n- loci-horizon http://logs.openstack.org/32/664632/1/check/loci-horizon/f830e8b/ : SUCCESS in 35m 40s\n- loci-ironic http://logs.openstack.org/32/664632/1/check/loci-ironic/7cbc2b1/ : SUCCESS in 37m 25s\n- loci-keystone http://logs.openstack.org/32/664632/1/check/loci-keystone/dfb4b42/ : SUCCESS in 34m 07s\n- loci-neutron http://logs.openstack.org/32/664632/1/check/loci-neutron/23028cc/ : SUCCESS in 32m 04s\n- loci-nova http://logs.openstack.org/32/664632/1/check/loci-nova/be86525/ : SUCCESS in 38m 20s\n- loci-octavia http://logs.openstack.org/32/664632/1/check/loci-octavia/ffc2800/ : SUCCESS in 34m 10s\n- loci-placement http://logs.openstack.org/32/664632/1/check/loci-placement/3cc5044/ : SUCCESS in 32m 59s (non-voting)\n- loci-requirements http://logs.openstack.org/32/664632/1/check/loci-requirements/9f1b16c/ : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8bc55cc77b619cea373fc31db34b765d7a383588","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-06-12 10:15:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"053e7783faf96903f6d89ab4cf2f9ad5afb30e19","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-06-12 11:34:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a83b86e1f86a141a9bbe39d7a266d3d462a92eac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 11:34:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afbab0fdfe6a6d276948a290d9e7f98e071e3f5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 12:31:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder http://logs.openstack.org/32/664632/1/gate/loci-cinder/3bdd2cf/ : SUCCESS in 36m 55s\n- loci-glance http://logs.openstack.org/32/664632/1/gate/loci-glance/3f04ccb/ : SUCCESS in 36m 30s\n- loci-heat http://logs.openstack.org/32/664632/1/gate/loci-heat/df632ba/ : SUCCESS in 35m 25s\n- loci-horizon http://logs.openstack.org/32/664632/1/gate/loci-horizon/1b256f8/ : SUCCESS in 28m 06s\n- loci-ironic http://logs.openstack.org/32/664632/1/gate/loci-ironic/9a7a838/ : SUCCESS in 34m 56s\n- loci-keystone http://logs.openstack.org/32/664632/1/gate/loci-keystone/23230af/ : SUCCESS in 35m 51s\n- loci-neutron http://logs.openstack.org/32/664632/1/gate/loci-neutron/8f17d2e/ : SUCCESS in 37m 32s\n- loci-nova http://logs.openstack.org/32/664632/1/gate/loci-nova/189a3b1/ : SUCCESS in 48m 10s\n- loci-octavia http://logs.openstack.org/32/664632/1/gate/loci-octavia/32e1da9/ : SUCCESS in 32m 24s\n- loci-requirements http://logs.openstack.org/32/664632/1/gate/loci-requirements/bc5e4d8/ : SUCCESS in 26m 35s","accounts_in_message":[],"_revision_number":1},{"id":"27c264e3601f5b04da72b91c804b943e7fcf7f9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 12:31:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac2b5a80adf997ba0d398ba40d6eec7de11e4299","revisions":{"ac2b5a80adf997ba0d398ba40d6eec7de11e4299":{"kind":"REWORK","_number":1,"created":"2019-06-11 16:22:05.000000000","uploader":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"ref":"refs/changes/32/664632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/664632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/664632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/664632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/664632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/664632/1"}}},"commit":{"parents":[{"commit":"139d072ac6cb4e7c5e1e7aed0b8c5a605da6f7c8","subject":"Merge \"Move comment to correct location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/139d072ac6cb4e7c5e1e7aed0b8c5a605da6f7c8"}]}],"author":{"name":"Chris Hoge","email":"chris@openstack.org","date":"2019-06-11 16:21:08.000000000","tz":-420},"committer":{"name":"Chris Hoge","email":"chris@openstack.org","date":"2019-06-11 16:21:08.000000000","tz":-420},"subject":"Updated README.md to reflect need to build base image","message":"Updated README.md to reflect need to build base image\n\nLoci requires that base images be prepared with some\ncommon tools and repositories. This patch updates the\nREADME to indicate that stock vendor images will not work\nand that the base images must be built.\n\nChange-Id: I9bd41f116978e501f0dec9e52f870c81224c0660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac2b5a80adf997ba0d398ba40d6eec7de11e4299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac2b5a80adf997ba0d398ba40d6eec7de11e4299"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
