)]}'
{"id":"openstack%2Floci~664937","triplet_id":"openstack%2Floci~master~I8aedaaa63d0b5aef6ef9691ee66ea9000ac14cbc","project":"openstack/loci","branch":"master","topic":"python3","hashtags":[],"change_id":"I8aedaaa63d0b5aef6ef9691ee66ea9000ac14cbc","subject":"Install python3-distutils when using python3","status":"MERGED","created":"2019-06-12 14:17:58.000000000","updated":"2019-06-18 09:51:48.000000000","submitted":"2019-06-18 09:51:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664937-1560851508762-e4049d2a","meta_rev_id":"8e3330a8f8b104acef17dc28c5ddb0f481963c76","_number":664937,"virtual_id_number":664937,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-18 09:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":2,"date":"2019-06-18 09:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"value":2,"date":"2019-06-13 15:01:50.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"date":"2019-06-13 13:19:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"date":"2019-06-13 13:19:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"date":"2019-06-13 13:19:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2019-06-18 09:01:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-13 13:19:53.000000000","_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 13:19:53.000000000","updated_by":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"reviewer":{"_account_id":1390,"name":"Paul Bourke","email":"pauldbourke@proton.me","username":"pauldbourke"},"state":"REVIEWER"},{"updated":"2019-06-13 13:19:53.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2019-06-13 13:19:53.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2019-06-13 13:19:53.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-06-13 15:01:50.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-06-18 09:01:48.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-18 09:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb22f8ca44849892093ad51eeafa773cb2d29b50","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2019-06-12 14:17:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eac32d1c6b209bbe06e0e813c365b6b5e578515a","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2019-06-12 14:54:08.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3d88a1657594102b98a9ef8625d3e1726a1ed813","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 16:16:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder http://logs.openstack.org/37/664937/2/check/loci-cinder/da006a0/ : SUCCESS in 36m 22s\n- loci-glance http://logs.openstack.org/37/664937/2/check/loci-glance/1a0f0d2/ : SUCCESS in 39m 49s\n- loci-heat http://logs.openstack.org/37/664937/2/check/loci-heat/cf83f26/ : SUCCESS in 35m 19s\n- loci-horizon http://logs.openstack.org/37/664937/2/check/loci-horizon/cb4e955/ : SUCCESS in 36m 10s\n- loci-ironic http://logs.openstack.org/37/664937/2/check/loci-ironic/542b44e/ : SUCCESS in 35m 48s\n- loci-keystone http://logs.openstack.org/37/664937/2/check/loci-keystone/f67ca49/ : SUCCESS in 28m 20s\n- loci-neutron http://logs.openstack.org/37/664937/2/check/loci-neutron/09e4e38/ : SUCCESS in 38m 03s\n- loci-nova http://logs.openstack.org/37/664937/2/check/loci-nova/2a7218c/ : SUCCESS in 39m 45s\n- loci-octavia http://logs.openstack.org/37/664937/2/check/loci-octavia/3eaea58/ : SUCCESS in 37m 58s\n- loci-placement http://logs.openstack.org/37/664937/2/check/loci-placement/8a5a25e/ : SUCCESS in 36m 39s (non-voting)\n- loci-requirements http://logs.openstack.org/37/664937/2/check/loci-requirements/99a6ae0/ : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":2},{"id":"07d226e8da444a5864a23cd7a79a69d6375c2ce3","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-06-13 15:01:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"797d543f5def188d8f1673db551a07c117db46f3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-18 09:01:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"021f900c2ec1db6ae9ff68cdf08496c2338938d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:02:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84c1de065b90acc9af07455c3556d8d2d9dc20d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:51:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7c88416732c7cd0a53c4614df6b44ee050f5e06e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:51:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder http://logs.openstack.org/37/664937/2/gate/loci-cinder/7536883/ : SUCCESS in 37m 15s\n- loci-glance http://logs.openstack.org/37/664937/2/gate/loci-glance/fb3c033/ : SUCCESS in 28m 41s\n- loci-heat http://logs.openstack.org/37/664937/2/gate/loci-heat/dfeeb91/ : SUCCESS in 37m 44s\n- loci-horizon http://logs.openstack.org/37/664937/2/gate/loci-horizon/404a1b7/ : SUCCESS in 40m 28s\n- loci-ironic http://logs.openstack.org/37/664937/2/gate/loci-ironic/8b1073c/ : SUCCESS in 32m 47s\n- loci-keystone http://logs.openstack.org/37/664937/2/gate/loci-keystone/dc2616b/ : SUCCESS in 34m 11s\n- loci-neutron http://logs.openstack.org/37/664937/2/gate/loci-neutron/9a43ed2/ : SUCCESS in 39m 16s\n- loci-nova http://logs.openstack.org/37/664937/2/gate/loci-nova/8a61222/ : SUCCESS in 36m 45s\n- loci-octavia http://logs.openstack.org/37/664937/2/gate/loci-octavia/0c41b1c/ : SUCCESS in 34m 45s\n- loci-requirements http://logs.openstack.org/37/664937/2/gate/loci-requirements/036118e/ : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1644595201a07519b5f3f94b99273bd1e51a3f31","revisions":{"a9ce2c60b4982281cef79008e6fee23eb0ecf2b3":{"kind":"REWORK","_number":1,"created":"2019-06-12 14:17:58.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/37/664937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/37/664937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/37/664937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/37/664937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/37/664937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/37/664937/1"}}},"commit":{"parents":[{"commit":"ac2b5a80adf997ba0d398ba40d6eec7de11e4299","subject":"Updated README.md to reflect need to build base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac2b5a80adf997ba0d398ba40d6eec7de11e4299"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2019-06-12 14:08:54.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2019-06-12 14:17:47.000000000","tz":180},"subject":"Install python2-distutils when using python3","message":"Install python2-distutils when using python3\n\ndistutils is not included into main python package on ubuntu.\nAdd python3-distutils explicitly when using PYTHON3\u003dyes.\n\nChange-Id: I8aedaaa63d0b5aef6ef9691ee66ea9000ac14cbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9ce2c60b4982281cef79008e6fee23eb0ecf2b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a9ce2c60b4982281cef79008e6fee23eb0ecf2b3"}]},"branch":"refs/heads/master"},"1644595201a07519b5f3f94b99273bd1e51a3f31":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-12 14:54:08.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/37/664937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/37/664937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/37/664937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/37/664937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/37/664937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/37/664937/2"}}},"commit":{"parents":[{"commit":"ac2b5a80adf997ba0d398ba40d6eec7de11e4299","subject":"Updated README.md to reflect need to build base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac2b5a80adf997ba0d398ba40d6eec7de11e4299"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2019-06-12 14:08:54.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2019-06-12 14:54:06.000000000","tz":0},"subject":"Install python3-distutils when using python3","message":"Install python3-distutils when using python3\n\ndistutils is not included into main python package on ubuntu.\nAdd python3-distutils explicitly when using PYTHON3\u003dyes.\n\nChange-Id: I8aedaaa63d0b5aef6ef9691ee66ea9000ac14cbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1644595201a07519b5f3f94b99273bd1e51a3f31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1644595201a07519b5f3f94b99273bd1e51a3f31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
