)]}'
{"id":"openstack%2Floci~665187","triplet_id":"openstack%2Floci~master~I0fcf9587aba02cef0bcb58a82f3b15e884fb6100","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I0fcf9587aba02cef0bcb58a82f3b15e884fb6100","subject":"Make Python3 the default interpreter","status":"ABANDONED","created":"2019-06-13 17:29:47.000000000","updated":"2019-11-07 21:04:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"745bf5a0279b9faed1d77f1b96bd5f000a7fe86b","_number":665187,"virtual_id_number":665187,"owner":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-08-26 11:33:43.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":-1,"date":"2019-07-18 11:47:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-23 20:55:27.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 11:47:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-26 11:33:43.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"}],"messages":[{"id":"39a78288f038b111d2aeff8b104ab8393ee808b3","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-06-13 17:29:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66de7843c06fe629a88d005cffa0f86a1286cca4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 18:07:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/87/665187/1/check/loci-cinder/8aaa672/ : FAILURE in 19m 06s\n- loci-glance http://logs.openstack.org/87/665187/1/check/loci-glance/f5a56b6/ : FAILURE in 24m 21s\n- loci-heat http://logs.openstack.org/87/665187/1/check/loci-heat/1a64ef1/ : FAILURE in 24m 05s\n- loci-horizon http://logs.openstack.org/87/665187/1/check/loci-horizon/b251c85/ : FAILURE in 23m 31s\n- loci-ironic http://logs.openstack.org/87/665187/1/check/loci-ironic/f6fec93/ : FAILURE in 16m 26s\n- loci-keystone http://logs.openstack.org/87/665187/1/check/loci-keystone/c0099f1/ : FAILURE in 24m 36s\n- loci-neutron http://logs.openstack.org/87/665187/1/check/loci-neutron/3ae5954/ : FAILURE in 19m 06s\n- loci-nova http://logs.openstack.org/87/665187/1/check/loci-nova/6e1dcd3/ : FAILURE in 20m 46s\n- loci-octavia http://logs.openstack.org/87/665187/1/check/loci-octavia/abdf7a1/ : FAILURE in 18m 24s\n- loci-placement http://logs.openstack.org/87/665187/1/check/loci-placement/b4e744b/ : FAILURE in 17m 35s (non-voting)\n- loci-requirements http://logs.openstack.org/87/665187/1/check/loci-requirements/b4408c1/ : FAILURE in 20m 49s","accounts_in_message":[],"_revision_number":1},{"id":"72876c29852421f340c78e40b2141374ebd45854","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-18 09:30:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f3228d661e7076913cb1328f411b73acf3656624","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 10:21:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/87/665187/2/check/loci-cinder/1cd1562/ : FAILURE in 15m 38s\n- loci-glance http://logs.openstack.org/87/665187/2/check/loci-glance/9210a07/ : FAILURE in 34m 06s\n- loci-heat http://logs.openstack.org/87/665187/2/check/loci-heat/e23c9a3/ : FAILURE in 14m 26s\n- loci-horizon http://logs.openstack.org/87/665187/2/check/loci-horizon/a20f4e7/ : FAILURE in 15m 39s\n- loci-ironic http://logs.openstack.org/87/665187/2/check/loci-ironic/09b164c/ : FAILURE in 13m 29s\n- loci-keystone http://logs.openstack.org/87/665187/2/check/loci-keystone/36619e1/ : FAILURE in 15m 23s\n- loci-neutron http://logs.openstack.org/87/665187/2/check/loci-neutron/7d43ecb/ : FAILURE in 20m 00s\n- loci-nova http://logs.openstack.org/87/665187/2/check/loci-nova/305a243/ : FAILURE in 14m 17s\n- loci-octavia http://logs.openstack.org/87/665187/2/check/loci-octavia/8a44620/ : FAILURE in 12m 47s\n- loci-placement http://logs.openstack.org/87/665187/2/check/loci-placement/322cbe0/ : FAILURE in 17m 18s (non-voting)\n- loci-requirements http://logs.openstack.org/87/665187/2/check/loci-requirements/84d7e92/ : FAILURE in 13m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d53fb143ffa43a90bc40060dfee01ba3f74687ce","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-07-18 11:14:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"58b1e16530bbebe0cd2b92a85ef3dea217a0c0c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 11:47:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/87/665187/3/check/loci-cinder/e5a326e/ : FAILURE in 15m 55s\n- loci-glance http://logs.openstack.org/87/665187/3/check/loci-glance/f20688d/ : FAILURE in 14m 09s\n- loci-heat http://logs.openstack.org/87/665187/3/check/loci-heat/7486e27/ : FAILURE in 16m 53s\n- loci-horizon http://logs.openstack.org/87/665187/3/check/loci-horizon/ef82325/ : FAILURE in 15m 25s\n- loci-ironic http://logs.openstack.org/87/665187/3/check/loci-ironic/43c02ab/ : FAILURE in 26m 30s\n- loci-keystone http://logs.openstack.org/87/665187/3/check/loci-keystone/24ab70f/ : FAILURE in 17m 44s\n- loci-neutron http://logs.openstack.org/87/665187/3/check/loci-neutron/0ecb5fb/ : FAILURE in 24m 26s\n- loci-nova http://logs.openstack.org/87/665187/3/check/loci-nova/fbd943f/ : FAILURE in 24m 52s\n- loci-octavia http://logs.openstack.org/87/665187/3/check/loci-octavia/5d8b651/ : FAILURE in 15m 21s\n- loci-placement http://logs.openstack.org/87/665187/3/check/loci-placement/aadceb3/ : FAILURE in 16m 42s (non-voting)\n- loci-requirements http://logs.openstack.org/87/665187/3/check/loci-requirements/1e92f70/ : FAILURE in 23m 38s","accounts_in_message":[],"_revision_number":3},{"id":"ae01699cab25d73d4cbb539594015845a19e1124","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-08-23 20:55:27.000000000","message":"Patch Set 3: Workflow-1\n\nPython3 on CentOS 7 is holding this one up. Our options are to wait for CentOS 8 and document the heck out of it (so that older releases can still use Python2) or drop support altogether (which I\u0027m not in favor of).","accounts_in_message":[],"_revision_number":3},{"id":"7f389af5e0d21577ec44eb436f039d55a9955325","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-26 11:08:40.000000000","message":"Patch Set 3:\n\nIn OSH we take the approach per distro: Leap 15 only supports py3, Ubuntu supports py2/3 depending on the version. I think makes more sense than saying \"everyone moves to py3\".\n\nIt\u0027s an approach we have in OpenStack-Ansible too, btw.","accounts_in_message":[],"_revision_number":3},{"id":"9fa0a75452dde17014191e4341cc50dd69ca88d7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-26 11:33:43.000000000","message":"Patch Set 3:\n\nExtra note: I would expect it would be possible to add py3 support to centos7 when 7.7 is out.","accounts_in_message":[],"_revision_number":3},{"id":"745bf5a0279b9faed1d77f1b96bd5f000a7fe86b","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-11-07 21:04:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"56c19fa28911d5c3d87683fd1545354b4e2981f7","revisions":{"ecebbf05e6c16b66ef00b89cb154b5b013a19e32":{"kind":"REWORK","_number":1,"created":"2019-06-13 17:29:47.000000000","uploader":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"ref":"refs/changes/87/665187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/87/665187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/665187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/665187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/665187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/87/665187/1"}}},"commit":{"parents":[{"commit":"ac2b5a80adf997ba0d398ba40d6eec7de11e4299","subject":"Updated README.md to reflect need to build base image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac2b5a80adf997ba0d398ba40d6eec7de11e4299"}]}],"author":{"name":"Chris Hoge","email":"chris@openstack.org","date":"2019-06-13 17:28:39.000000000","tz":-420},"committer":{"name":"Chris Hoge","email":"chris@openstack.org","date":"2019-06-13 17:28:39.000000000","tz":-420},"subject":"Make Python3 the default interpreter","message":"Make Python3 the default interpreter\n\nPython2 is going away. Make Python3 the default build interpreter\nfor Loci.\n\nChange-Id: I0fcf9587aba02cef0bcb58a82f3b15e884fb6100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ecebbf05e6c16b66ef00b89cb154b5b013a19e32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ecebbf05e6c16b66ef00b89cb154b5b013a19e32"}]},"branch":"refs/heads/master"},"59386f3a34e93fae5a07b83abfbaa5dc916b86c8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-18 09:30:25.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/87/665187/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/87/665187/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/665187/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/665187/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/665187/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/87/665187/2"}}},"commit":{"parents":[{"commit":"1644595201a07519b5f3f94b99273bd1e51a3f31","subject":"Install python3-distutils when using python3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1644595201a07519b5f3f94b99273bd1e51a3f31"}]}],"author":{"name":"Chris Hoge","email":"chris@openstack.org","date":"2019-06-13 17:28:39.000000000","tz":-420},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-06-18 09:30:25.000000000","tz":0},"subject":"Make Python3 the default interpreter","message":"Make Python3 the default interpreter\n\nPython2 is going away. Make Python3 the default build interpreter\nfor Loci.\n\nChange-Id: I0fcf9587aba02cef0bcb58a82f3b15e884fb6100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/59386f3a34e93fae5a07b83abfbaa5dc916b86c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/59386f3a34e93fae5a07b83abfbaa5dc916b86c8"}]},"branch":"refs/heads/master"},"56c19fa28911d5c3d87683fd1545354b4e2981f7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-18 11:14:38.000000000","uploader":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"ref":"refs/changes/87/665187/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/87/665187/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/665187/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/665187/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/87/665187/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/87/665187/3"}}},"commit":{"parents":[{"commit":"ed422d6e239ecaf4480d04b3121c8aecd8a3d690","subject":"Add python3-rados for dpkg and glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ed422d6e239ecaf4480d04b3121c8aecd8a3d690"}]}],"author":{"name":"Chris Hoge","email":"chris@openstack.org","date":"2019-06-13 17:28:39.000000000","tz":-420},"committer":{"name":"Pete Birley","email":"petebirley@gmail.com","date":"2019-07-18 11:14:38.000000000","tz":0},"subject":"Make Python3 the default interpreter","message":"Make Python3 the default interpreter\n\nPython2 is going away. Make Python3 the default build interpreter\nfor Loci.\n\nChange-Id: I0fcf9587aba02cef0bcb58a82f3b15e884fb6100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/56c19fa28911d5c3d87683fd1545354b4e2981f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/56c19fa28911d5c3d87683fd1545354b4e2981f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
