)]}'
{"id":"openstack%2Floci~668516","triplet_id":"openstack%2Floci~master~I17354765c896fddc5962ee53d287d5a61abe8e11","project":"openstack/loci","branch":"master","topic":"pike-em","hashtags":[],"change_id":"I17354765c896fddc5962ee53d287d5a61abe8e11","subject":"Move CentOS to use RDO Stein packages","status":"MERGED","created":"2019-07-01 16:18:36.000000000","updated":"2019-07-18 12:10:46.000000000","submitted":"2019-07-18 12:10:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668516-1563451846456-a91269ad","meta_rev_id":"4a15facd63dbb4da0ee5e495dff90a706ef8ddaf","_number":668516,"virtual_id_number":668516,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-02 07:32:51.000000000","_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-18 12:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2019-07-18 11:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":2,"date":"2019-07-09 12:52:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2019-07-18 11:14:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 12:52:37.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-18 11:14:04.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-07-18 12:10:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea2ec0c2417600f950746f7c0484131240b8d987","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-07-01 16:18:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77ecb55add889670790e49bebb7608831d166b08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 17:08:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder http://logs.openstack.org/16/668516/1/check/loci-cinder/b9b0238/ : SUCCESS in 43m 04s\n- loci-glance http://logs.openstack.org/16/668516/1/check/loci-glance/12b6bf0/ : SUCCESS in 37m 19s\n- loci-heat http://logs.openstack.org/16/668516/1/check/loci-heat/770a675/ : SUCCESS in 38m 13s\n- loci-horizon http://logs.openstack.org/16/668516/1/check/loci-horizon/d5e4185/ : SUCCESS in 32m 13s\n- loci-ironic http://logs.openstack.org/16/668516/1/check/loci-ironic/74c375f/ : FAILURE in 26m 01s\n- loci-keystone http://logs.openstack.org/16/668516/1/check/loci-keystone/2ed7302/ : SUCCESS in 33m 29s\n- loci-neutron http://logs.openstack.org/16/668516/1/check/loci-neutron/a14cc6e/ : SUCCESS in 30m 58s\n- loci-nova http://logs.openstack.org/16/668516/1/check/loci-nova/3a4e8b8/ : SUCCESS in 33m 43s\n- loci-octavia http://logs.openstack.org/16/668516/1/check/loci-octavia/e68f182/ : SUCCESS in 35m 12s\n- loci-placement http://logs.openstack.org/16/668516/1/check/loci-placement/b18d565/ : SUCCESS in 34m 41s (non-voting)\n- loci-requirements http://logs.openstack.org/16/668516/1/check/loci-requirements/180de04/ : SUCCESS in 31m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f2066291493fbe7761dc2cbaeaefeac3ee9716d1","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-07-02 07:32:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af889b512f03080e11e0d02e318a04c1106d9592","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 08:34:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder http://logs.openstack.org/16/668516/1/check/loci-cinder/f5f1408/ : SUCCESS in 54m 04s\n- loci-glance http://logs.openstack.org/16/668516/1/check/loci-glance/07d0101/ : SUCCESS in 53m 59s\n- loci-heat http://logs.openstack.org/16/668516/1/check/loci-heat/180c8e5/ : SUCCESS in 53m 28s\n- loci-horizon http://logs.openstack.org/16/668516/1/check/loci-horizon/cb741c9/ : SUCCESS in 57m 04s\n- loci-ironic http://logs.openstack.org/16/668516/1/check/loci-ironic/0d0b9cb/ : SUCCESS in 36m 31s\n- loci-keystone http://logs.openstack.org/16/668516/1/check/loci-keystone/82967d6/ : SUCCESS in 30m 31s\n- loci-neutron http://logs.openstack.org/16/668516/1/check/loci-neutron/1b4d254/ : SUCCESS in 43m 39s\n- loci-nova http://logs.openstack.org/16/668516/1/check/loci-nova/5c84a1f/ : SUCCESS in 55m 08s\n- loci-octavia http://logs.openstack.org/16/668516/1/check/loci-octavia/671a57a/ : SUCCESS in 34m 40s\n- loci-placement http://logs.openstack.org/16/668516/1/check/loci-placement/554af64/ : SUCCESS in 34m 28s (non-voting)\n- loci-requirements http://logs.openstack.org/16/668516/1/check/loci-requirements/40d1b0e/ : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3822dbdbb4efb61d2b856e26f56385bbef3088f4","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-09 12:52:37.000000000","message":"Patch Set 1: Code-Review+2\n\nSeems better than what we have, and we\u0027ve historically focused on master, so that\u0027s the closest we have.","accounts_in_message":[],"_revision_number":1},{"id":"7881643bdf901bf207f8985ad03d55669cbce110","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-07-18 11:14:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b0743eb8db808bdf6ba14a08397fcc16c9dd8da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 11:14:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd0bbc5de9bbb5c1d52068627c69ea06a88c9f05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 12:10:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4b76b946b211bbf39a866f8beade9eda2c8b00b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 12:10:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder http://logs.openstack.org/16/668516/1/gate/loci-cinder/ade3c56/ : SUCCESS in 37m 41s\n- loci-glance http://logs.openstack.org/16/668516/1/gate/loci-glance/7fb8e2e/ : SUCCESS in 43m 52s\n- loci-heat http://logs.openstack.org/16/668516/1/gate/loci-heat/74783a5/ : SUCCESS in 32m 24s\n- loci-horizon http://logs.openstack.org/16/668516/1/gate/loci-horizon/002a380/ : SUCCESS in 37m 02s\n- loci-ironic http://logs.openstack.org/16/668516/1/gate/loci-ironic/299d6e4/ : SUCCESS in 35m 28s\n- loci-keystone http://logs.openstack.org/16/668516/1/gate/loci-keystone/ae397fe/ : SUCCESS in 33m 33s\n- loci-neutron http://logs.openstack.org/16/668516/1/gate/loci-neutron/a53351d/ : SUCCESS in 43m 26s\n- loci-nova http://logs.openstack.org/16/668516/1/gate/loci-nova/085caf5/ : SUCCESS in 43m 31s\n- loci-octavia http://logs.openstack.org/16/668516/1/gate/loci-octavia/89faa99/ : SUCCESS in 41m 14s\n- loci-requirements http://logs.openstack.org/16/668516/1/gate/loci-requirements/a349796/ : SUCCESS in 31m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14ee0f585af414494057d10e5aae473c4dd803a3","revisions":{"14ee0f585af414494057d10e5aae473c4dd803a3":{"kind":"REWORK","_number":1,"created":"2019-07-01 16:18:36.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/16/668516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/16/668516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/16/668516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/16/668516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/16/668516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/16/668516/1"}}},"commit":{"parents":[{"commit":"ed422d6e239ecaf4480d04b3121c8aecd8a3d690","subject":"Add python3-rados for dpkg and glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ed422d6e239ecaf4480d04b3121c8aecd8a3d690"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-07-01 16:14:50.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-07-01 16:18:33.000000000","tz":120},"subject":"Move CentOS to use RDO Stein packages","message":"Move CentOS to use RDO Stein packages\n\nPike is in extended maintenance and is EOLed in terms of RDO.\n\nTo be accurate the version of RDO repo should match with the branch\nbeing used to build the containers. Anyway, IIUC, most python packages\nare installed from source or pip, so i guess few packages should be\ninstalled from distro repos.\n\nChange-Id: I17354765c896fddc5962ee53d287d5a61abe8e11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/14ee0f585af414494057d10e5aae473c4dd803a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/14ee0f585af414494057d10e5aae473c4dd803a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
