)]}'
{"id":"openstack%2Floci~671438","triplet_id":"openstack%2Floci~master~I0551af67292c489b6345d8f60d874dd10905c28a","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I0551af67292c489b6345d8f60d874dd10905c28a","subject":"Add LICENSE file to make it explicit","status":"MERGED","created":"2019-07-18 09:34:25.000000000","updated":"2019-07-22 12:59:37.000000000","submitted":"2019-07-22 12:59:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671438-1563800377691-e6e18664","meta_rev_id":"7a936fc468a88538f9a31532e524ca897e03d514","_number":671438,"virtual_id_number":671438,"owner":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-22 12:59:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":2,"date":"2019-07-18 11:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":2,"date":"2019-07-22 12:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2019-07-22 12:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 11:12:51.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-07-22 12:10:04.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-22 12:59:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8e086fe6658c75c23c6f4a7b2a83d327aed70a6","author":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"date":"2019-07-18 09:34:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3eeafa90ba775d96e567937f2910aaee854edbb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 10:36:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder http://logs.openstack.org/38/671438/1/check/loci-cinder/7d9f523/ : SUCCESS in 49m 38s\n- loci-glance http://logs.openstack.org/38/671438/1/check/loci-glance/f13144b/ : SUCCESS in 38m 39s\n- loci-heat http://logs.openstack.org/38/671438/1/check/loci-heat/bdb2c47/ : SUCCESS in 38m 16s\n- loci-horizon http://logs.openstack.org/38/671438/1/check/loci-horizon/055df95/ : SUCCESS in 52m 34s\n- loci-ironic http://logs.openstack.org/38/671438/1/check/loci-ironic/a1a0f58/ : SUCCESS in 50m 00s\n- loci-keystone http://logs.openstack.org/38/671438/1/check/loci-keystone/ec57aba/ : SUCCESS in 47m 56s\n- loci-neutron http://logs.openstack.org/38/671438/1/check/loci-neutron/802ef38/ : SUCCESS in 54m 30s\n- loci-nova http://logs.openstack.org/38/671438/1/check/loci-nova/79f957f/ : SUCCESS in 42m 35s\n- loci-octavia http://logs.openstack.org/38/671438/1/check/loci-octavia/9f0f0c1/ : SUCCESS in 35m 56s\n- loci-placement http://logs.openstack.org/38/671438/1/check/loci-placement/099dc74/ : SUCCESS in 36m 31s (non-voting)\n- loci-requirements http://logs.openstack.org/38/671438/1/check/loci-requirements/7c94880/ : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":1},{"id":"bda562136d2497a1740c315f6acc8e68c5f2d8f8","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-07-18 11:12:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d05cfc58d83ab72b48322d9a466505661a96ece","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-22 12:10:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nChecked if license text is correct, and it is. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"2e6980aeb1398641fb18f21c406eea577095c2a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:10:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d22ebb8a10f719b77c500057d436dfa8677f8eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:59:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a936fc468a88538f9a31532e524ca897e03d514","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:59:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder http://logs.openstack.org/38/671438/1/gate/loci-cinder/1eb2af8/ : SUCCESS in 41m 28s\n- loci-glance http://logs.openstack.org/38/671438/1/gate/loci-glance/9ed76a6/ : SUCCESS in 39m 51s\n- loci-heat http://logs.openstack.org/38/671438/1/gate/loci-heat/5310609/ : SUCCESS in 37m 33s\n- loci-horizon http://logs.openstack.org/38/671438/1/gate/loci-horizon/2c9dc58/ : SUCCESS in 38m 20s\n- loci-ironic http://logs.openstack.org/38/671438/1/gate/loci-ironic/13371c8/ : SUCCESS in 41m 42s\n- loci-keystone http://logs.openstack.org/38/671438/1/gate/loci-keystone/36a6437/ : SUCCESS in 43m 20s\n- loci-neutron http://logs.openstack.org/38/671438/1/gate/loci-neutron/99151a8/ : SUCCESS in 39m 53s\n- loci-nova http://logs.openstack.org/38/671438/1/gate/loci-nova/ee5d314/ : SUCCESS in 39m 24s\n- loci-octavia http://logs.openstack.org/38/671438/1/gate/loci-octavia/8a436d2/ : SUCCESS in 36m 12s\n- loci-requirements http://logs.openstack.org/38/671438/1/gate/loci-requirements/1e88ec8/ : SUCCESS in 24m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3759f7050bed7684c0a09ef469c925c935a33150","revisions":{"3759f7050bed7684c0a09ef469c925c935a33150":{"kind":"REWORK","_number":1,"created":"2019-07-18 09:34:25.000000000","uploader":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"ref":"refs/changes/38/671438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/38/671438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/38/671438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/38/671438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/38/671438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/38/671438/1"}}},"commit":{"parents":[{"commit":"ed422d6e239ecaf4480d04b3121c8aecd8a3d690","subject":"Add python3-rados for dpkg and glance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ed422d6e239ecaf4480d04b3121c8aecd8a3d690"}]}],"author":{"name":"Jan Zerebecki","email":"jzerebecki@suse.com","date":"2019-07-18 09:25:09.000000000","tz":120},"committer":{"name":"Jan Zerebecki","email":"jzerebecki@suse.com","date":"2019-07-18 09:25:09.000000000","tz":120},"subject":"Add LICENSE file to make it explicit","message":"Add LICENSE file to make it explicit\n\nChange-Id: I0551af67292c489b6345d8f60d874dd10905c28a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3759f7050bed7684c0a09ef469c925c935a33150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3759f7050bed7684c0a09ef469c925c935a33150"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
