)]}'
{"id":"openstack%2Floci~675832","triplet_id":"openstack%2Floci~master~I57655313e1a53b7f8c91a9e03f4749273665037a","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I57655313e1a53b7f8c91a9e03f4749273665037a","subject":"Add job to build Manila images","status":"MERGED","created":"2019-08-12 07:22:31.000000000","updated":"2019-08-23 15:45:22.000000000","submitted":"2019-08-23 15:45:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675832-1566575122543-91ed003a","meta_rev_id":"c6c50ee307fa183e591bcb5e193ec02147faacbd","_number":675832,"virtual_id_number":675832,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":2,"date":"2019-08-23 15:45:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":2,"date":"2019-08-23 14:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-23 14:05:45.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":1,"date":"2019-08-23 14:28:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 14:05:45.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-08-23 14:28:57.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-08-23 15:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e00cc500d3e93749a5e6cb94deed94fa7854f2dc","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-08-12 07:22:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4d9ed166385571d35a4268e79a7ac2e687aca9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 08:19:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://logs.opendev.org/32/675832/1/check/loci-cinder/14de8e8/ : SUCCESS in 37m 47s\n- loci-glance https://logs.opendev.org/32/675832/1/check/loci-glance/83eb071/ : SUCCESS in 50m 55s\n- loci-heat https://logs.opendev.org/32/675832/1/check/loci-heat/a1624ad/ : SUCCESS in 36m 26s\n- loci-horizon https://logs.opendev.org/32/675832/1/check/loci-horizon/ff0a5a1/ : SUCCESS in 36m 34s\n- loci-ironic https://logs.opendev.org/32/675832/1/check/loci-ironic/46c518e/ : SUCCESS in 33m 13s\n- loci-keystone https://logs.opendev.org/32/675832/1/check/loci-keystone/e9c38ab/ : SUCCESS in 40m 44s\n- loci-manila https://logs.opendev.org/32/675832/1/check/loci-manila/49e8a40/ : SUCCESS in 34m 58s\n- loci-neutron https://logs.opendev.org/32/675832/1/check/loci-neutron/988f41a/ : SUCCESS in 34m 40s\n- loci-nova https://logs.opendev.org/32/675832/1/check/loci-nova/6b6db01/ : SUCCESS in 34m 21s\n- loci-octavia https://logs.opendev.org/32/675832/1/check/loci-octavia/5591bb5/ : SUCCESS in 40m 40s\n- loci-placement https://logs.opendev.org/32/675832/1/check/loci-placement/4d1a351/ : SUCCESS in 31m 52s (non-voting)\n- loci-requirements https://logs.opendev.org/32/675832/1/check/loci-requirements/f18196b/ : SUCCESS in 30m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e0c1748263035104e3d6ab61217a4b0ae8bc691d","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-08-23 14:05:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e508208798c20e3b89726bc18bda8dca0ff391c9","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-08-23 14:28:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"49e1c8f235531f8324ad37e313f23ef3c1862d8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 14:29:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87b6c654285ee57aa12a383b8f61d2c024b6c7aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:45:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c6c50ee307fa183e591bcb5e193ec02147faacbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:45:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/bf3d36b6016045dc956c45debd05ec0f : SUCCESS in 36m 57s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c537adbf42af49b1bb89064a2e9ace3f : SUCCESS in 35m 03s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5a920dd0cb24444891209b60959589a9 : SUCCESS in 38m 27s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ac898d97da9846308bbe84dc9c707948 : SUCCESS in 36m 19s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8fff7ce79ce94af3bb32a2d5a3c6db31 : SUCCESS in 36m 35s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2f4b75c1762e41ce9d9c4c9b57761a55 : SUCCESS in 39m 17s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9f42d8a2f62f4d5db3e2caf18a08178c : SUCCESS in 30m 48s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/81ba5be1edec44cbaa36e12cad2a16e1 : SUCCESS in 32m 06s\n- loci-nova https://zuul.opendev.org/t/openstack/build/68e1b92619ac415aa165b3cc71f0ea03 : SUCCESS in 35m 10s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/417b212db92846cf86b522385898708f : SUCCESS in 31m 03s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e9ef38ddbb014af0b7d77c5204a9d697 : SUCCESS in 24m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01705a2b5dabbd3dbfa21f82125dd7d1e2106295","revisions":{"01705a2b5dabbd3dbfa21f82125dd7d1e2106295":{"kind":"REWORK","_number":1,"created":"2019-08-12 07:22:31.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/32/675832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/675832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/675832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/675832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/675832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/675832/1"}}},"commit":{"parents":[{"commit":"c64ea7529b4152d7628c5cc841ccd4dd6dba1615","subject":"Bypass trollius building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c64ea7529b4152d7628c5cc841ccd4dd6dba1615"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-12 07:21:49.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-08-12 07:21:49.000000000","tz":120},"subject":"Add job to build Manila images","message":"Add job to build Manila images\n\nThis is useful for people who want to use Manila (shared filesystem as\na service).\n\nChange-Id: I57655313e1a53b7f8c91a9e03f4749273665037a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/01705a2b5dabbd3dbfa21f82125dd7d1e2106295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/01705a2b5dabbd3dbfa21f82125dd7d1e2106295"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
