)]}'
{"id":"openstack%2Floci~676398","triplet_id":"openstack%2Floci~master~I83fe0c46cad24d9a1c64c4047429bb3491791fe5","project":"openstack/loci","branch":"master","topic":"cleanup","hashtags":[],"change_id":"I83fe0c46cad24d9a1c64c4047429bb3491791fe5","subject":"Revert \"Remove explict enum-compat\"","status":"MERGED","created":"2019-08-14 12:06:27.000000000","updated":"2019-08-16 15:58:05.000000000","submitted":"2019-08-16 15:58:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676398-1565971085761-8c476a95","meta_rev_id":"e7693cbad1701b133d09a7da7f5e91dbb2a5ec3f","_number":676398,"virtual_id_number":676398,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":2,"date":"2019-08-16 15:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"all":[{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-16 15:01:40.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"all":[{"value":0,"date":"2019-08-14 12:06:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-16 15:01:40.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 12:06:27.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2019-08-16 15:01:40.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-08-16 15:58:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bd22260891aefbdcb360a439c3e865174ef5524","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-14 12:06:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f5657f4b1f69f64a403cac7c91633e7e7d9268e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 13:03:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/294d944cf60f4430b0218c578b1d6e5c : SUCCESS in 32m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b34320ce70df4ae4910dc7d08ad8b234 : SUCCESS in 37m 52s\n- loci-heat https://zuul.opendev.org/t/openstack/build/09224059250c4dfcaf087fb9f395e569 : SUCCESS in 50m 37s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b6110c89171a42388118f9a8c14e7c98 : SUCCESS in 38m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/13fac8772b004b0a85dddb98baef2fa5 : SUCCESS in 47m 58s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6e605626878d47f4a9bacdc16b5165fd : SUCCESS in 37m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/973c9c473c87425099b3c06f528d36ad : SUCCESS in 49m 42s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a2aa3b282919404fa4ed5645a6596214 : SUCCESS in 39m 52s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/daeee59e140b457bb4ea2fd879f9f307 : SUCCESS in 47m 53s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0592805afd4b4af880d1b1b3e71403b3 : SUCCESS in 36m 21s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4bd59f9e12a24f8794c5f98a94b3a320 : SUCCESS in 40m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f225a105df7490fca2160efb0002de8ef048ded3","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-08-16 15:01:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1403e1cdefe1bafc1cff6b3789cfbc6853219218","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:01:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2c0d6d2e34c1fee18a59c51565eee8968c44aa5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:58:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5c50194aa647426c82755ffbfd3e263b : SUCCESS in 40m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/3f301d0ea83d4548ab86a99d61ff2dd9 : SUCCESS in 41m 37s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d84bd366e0cc432bbaa6bef5a34ccaff : SUCCESS in 38m 10s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/739384e711cd4a289b59e40ae7615dc9 : SUCCESS in 39m 37s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8a395d25b7a7403dbe2deb94153f9f2e : SUCCESS in 43m 11s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c061cd9cd9a14399a2d26b9725ff9b02 : SUCCESS in 42m 04s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/13f37ddadc604eafa3687d6c9aae52d4 : SUCCESS in 43m 23s\n- loci-nova https://zuul.opendev.org/t/openstack/build/1cab9decc48b4f66985cd8ec565c64e9 : SUCCESS in 50m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/53fe833f357b4900ae43f08894b02c76 : SUCCESS in 36m 40s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6c57763a545e493fadfa97c02b899a3f : SUCCESS in 36m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5fbdfba4f75bef5c071c9eaad8c1c0c51398a567","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:58:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"435bd3731d28a6a1bb57d02f5d7895ce748668d0","revisions":{"435bd3731d28a6a1bb57d02f5d7895ce748668d0":{"kind":"REWORK","_number":1,"created":"2019-08-14 12:06:27.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/98/676398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/98/676398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/676398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/676398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/676398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/98/676398/1"}}},"commit":{"parents":[{"commit":"acfd446c03324e9fb22b2598629d79d8c0bc026a","subject":"Remove explict enum-compat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/acfd446c03324e9fb22b2598629d79d8c0bc026a"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-14 12:06:27.000000000","tz":0},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-08-14 12:06:27.000000000","tz":0},"subject":"Revert \"Remove explict enum-compat\"","message":"Revert \"Remove explict enum-compat\"\n\nThis reverts commit acfd446c03324e9fb22b2598629d79d8c0bc026a.\nThe workarounds discussed here are implemented in ocata/pike,\nStein and above doesn\u0027t need enum-compat. However Rocky\nwould still fail without this.\n\nChange-Id: I83fe0c46cad24d9a1c64c4047429bb3491791fe5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/435bd3731d28a6a1bb57d02f5d7895ce748668d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/435bd3731d28a6a1bb57d02f5d7895ce748668d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
