)]}'
{"id":"openstack%2Floci~676952","triplet_id":"openstack%2Floci~master~I4ff7dbd28220da07f4616970be12abcb8de58ceb","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I4ff7dbd28220da07f4616970be12abcb8de58ceb","subject":"Add dhclient package for Octavia","status":"MERGED","created":"2019-08-16 13:54:57.000000000","updated":"2019-08-23 15:52:34.000000000","submitted":"2019-08-23 15:52:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676952-1566575554054-91ed003a","meta_rev_id":"c76d3f94a36077698291176de69a06fdd2c2d1db","_number":676952,"virtual_id_number":676952,"owner":{"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":2,"date":"2019-08-23 15:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},{"value":2,"date":"2019-08-23 14:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-08-20 14:53:39.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},{"value":1,"date":"2019-08-23 14:29:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 07:22:33.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-08-20 14:53:39.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-08-23 14:29:10.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-08-23 15:52:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c22154e8e2f957e07c3c8fd35adf7431df4ee77","author":{"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},"date":"2019-08-16 13:54:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"458af4fdb9e97a6980f31cfe5cfb2aa9a5e73f83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 14:52:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/613f4d96115545a4bd36b3e9efb641e0 : SUCCESS in 36m 11s\n- loci-glance https://zuul.opendev.org/t/openstack/build/da37e09c11754a94a99cfa0a6534ba27 : SUCCESS in 36m 00s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5dd573b4f9f540168a8741aed3f2db5c : SUCCESS in 36m 45s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2b76a6bcf63e4d72bba9a0b065cb190e : SUCCESS in 43m 57s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c9ce50d803c2445093e89e39656cb279 : SUCCESS in 32m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f7c98d48cf0e468e86c48439a8a7ddc1 : SUCCESS in 46m 05s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/61de9a002b8342d7b60b6ffc2622eac9 : SUCCESS in 52m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/6d64aaa351bc4ca786bb3e5a65f12080 : SUCCESS in 41m 23s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/53ce6784be5841ef8ca394068ef5df6e : FAILURE in 37m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/64a15cb840944789820da0d8ddcbaefc : SUCCESS in 32m 34s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/14040d783e3c42edaadc20667983ece9 : SUCCESS in 26m 24s","accounts_in_message":[],"_revision_number":1},{"id":"008a26b468a2cb348bb9094ca7543366568ce055","author":{"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},"date":"2019-08-16 15:06:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6535c339e7fd00d6f89bfe8632ee51b14032ad5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:59:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/836f1cff78f64ee8a8f1c8436c212a94 : SUCCESS in 41m 34s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2499b91919bf4c53b3e50a3758440042 : SUCCESS in 37m 27s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b24328e471704c168e98d06ab48a855b : SUCCESS in 38m 44s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/8df699ba36dd4124a098da7d3eaad60b : SUCCESS in 37m 40s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/bdd8b323a1324654bf5d5c27ad06bdae : SUCCESS in 38m 43s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d09f7015320e4a09bab3dc46bcb848fb : SUCCESS in 40m 45s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5c7512e2909145af8bd5c427d13f8ea0 : SUCCESS in 42m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/af91e2cda983451199bf165ad60ce4f5 : SUCCESS in 44m 33s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/53c3969ea38148019539c1202b1cf7a0 : SUCCESS in 36m 31s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2b791a07799a4e45a5f9718778963698 : SUCCESS in 37m 55s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7e59b1e01f964b9f99a842683f5e7fae : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":2},{"id":"50b750caba725c927faf4f26ea164c37893afc46","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-08-20 07:22:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3e8f60171ec109b6d9c8e7ae6908fd36e22d10c","author":{"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},"date":"2019-08-20 07:29:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"76f69274195723949580cb80f8f30b32d46afbd3","author":{"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},"date":"2019-08-20 07:29:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d50ac66f13f66777976b654e08b482e5f718c714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 08:28:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/63d31183d92a4381a67f8f70d5f662e8 : SUCCESS in 36m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a18e566dcb9248db849bbc7fd150cb2e : SUCCESS in 37m 21s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f5e75b8209524adbbbfa42d04f8a43c4 : SUCCESS in 34m 32s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/732aaf6104a24d5a961696dbee538065 : SUCCESS in 39m 16s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8ac59c8d8f6449bb8cdae991096434b5 : SUCCESS in 41m 52s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/cf8cfa0c3d044b1bb16baf305a390664 : SUCCESS in 47m 30s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/794c620815ea4c37ad385af0cc43ac57 : SUCCESS in 53m 48s\n- loci-nova https://zuul.opendev.org/t/openstack/build/af8c692c9c6d49389ad2a2a0d014aec7 : SUCCESS in 52m 32s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/840f98ac11f74fc1a3d9269e252c41b5 : SUCCESS in 50m 31s\n- loci-placement https://zuul.opendev.org/t/openstack/build/701121506ec44640ac2558bdfb0fd3d4 : SUCCESS in 31m 43s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/373ad10138de4d28887e0e6ff86883ff : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":3},{"id":"bb9d61c40f12624898869dfcc23d902f3f18cbbd","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-08-20 14:53:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"16c4c1594f266364e0340b41aed14ed2e91b173d","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-08-23 14:29:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e31d5730a8d7c557372d9b6ca7a9d0caeae386a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 14:29:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"20fd01420fdea139bd3713cb2c5a1f5ba88ac738","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:52:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a118b58dc05d4fab99dc18446eef3f08 : SUCCESS in 33m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7222c2a1cb1e460d829493f1332e3ed5 : SUCCESS in 35m 35s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2b5a5f8f9a354df983df6c33eae2d628 : SUCCESS in 40m 36s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f93dfd01f34a4248b47bbcbafa2adbf9 : SUCCESS in 36m 29s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/dc16c0ec5dce43bdb7084e85db035ca9 : SUCCESS in 36m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/99069b4373ea457ba7f6bbeb796945fc : SUCCESS in 33m 09s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4a9498a5e0ba4d049f5ef96215aa9d4f : SUCCESS in 40m 25s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/fe49e0b2428d457ca54e0dbd550059cc : SUCCESS in 41m 28s\n- loci-nova https://zuul.opendev.org/t/openstack/build/de3900a3c56b4e428922b94e3e339642 : SUCCESS in 33m 41s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/24f239e3ef5245f1870d3bac58b6a718 : SUCCESS in 40m 18s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/db6cda8f4ded4d9288afe55d455f7e4a : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":3},{"id":"c76d3f94a36077698291176de69a06fdd2c2d1db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:52:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b4bdec997baa7bcef6ad403de4ac7e22cf60409f","revisions":{"86dbb3174dce723940b02f79d548222777273164":{"kind":"REWORK","_number":1,"created":"2019-08-16 13:54:57.000000000","uploader":{"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},"ref":"refs/changes/52/676952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/52/676952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/676952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/676952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/676952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/52/676952/1"}}},"commit":{"parents":[{"commit":"c7a68ed0d53de3687b11f76bd7d85e318c3e3bce","subject":"Merge \"Remove explict enum-compat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c7a68ed0d53de3687b11f76bd7d85e318c3e3bce"}]}],"author":{"name":"hagun.kim","email":"hagun.kim@samsung.com","date":"2019-08-16 13:48:47.000000000","tz":-420},"committer":{"name":"hagun.kim","email":"hagun.kim@samsung.com","date":"2019-08-16 13:54:54.000000000","tz":-420},"subject":"Add dhclient package for Octavia","message":"Add dhclient package for Octavia\n\nTo support openstack-helm Octavia chart,\n\ndhclient package is added to Octavia loci.\n\nRefer to https://review.opendev.org/630821\n\nChange-Id: I4ff7dbd28220da07f4616970be12abcb8de58ceb\nSigned-off-by: hagun.kim \u003chagun.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/86dbb3174dce723940b02f79d548222777273164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/86dbb3174dce723940b02f79d548222777273164"}]},"branch":"refs/heads/master"},"e755768a7f68fbe795b2037c243be2ea97c42467":{"kind":"REWORK","_number":2,"created":"2019-08-16 15:06:46.000000000","uploader":{"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},"ref":"refs/changes/52/676952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/52/676952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/676952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/676952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/676952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/52/676952/2"}}},"commit":{"parents":[{"commit":"c7a68ed0d53de3687b11f76bd7d85e318c3e3bce","subject":"Merge \"Remove explict enum-compat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c7a68ed0d53de3687b11f76bd7d85e318c3e3bce"}]}],"author":{"name":"hagun.kim","email":"hagun.kim@samsung.com","date":"2019-08-16 13:48:47.000000000","tz":-420},"committer":{"name":"hagun.kim","email":"hagun.kim@samsung.com","date":"2019-08-16 15:06:58.000000000","tz":-420},"subject":"Add dhclient package for Octavia","message":"Add dhclient package for Octavia\n\nTo support openstack-helm Octavia chart,\n\ndhclient package is added to Octavia loci.\n\nRefer to https://review.opendev.org/630821\n\nChange-Id: I4ff7dbd28220da07f4616970be12abcb8de58ceb\nSigned-off-by: hagun.kim \u003chagun.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e755768a7f68fbe795b2037c243be2ea97c42467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e755768a7f68fbe795b2037c243be2ea97c42467"}]},"branch":"refs/heads/master"},"b4bdec997baa7bcef6ad403de4ac7e22cf60409f":{"kind":"REWORK","_number":3,"created":"2019-08-20 07:29:22.000000000","uploader":{"_account_id":29796,"name":"Kim Ha-gun","email":"hagun.kim@samsung.com","username":"hagun-kim"},"ref":"refs/changes/52/676952/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/52/676952/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/676952/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/676952/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/676952/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/52/676952/3"}}},"commit":{"parents":[{"commit":"c7a68ed0d53de3687b11f76bd7d85e318c3e3bce","subject":"Merge \"Remove explict enum-compat\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c7a68ed0d53de3687b11f76bd7d85e318c3e3bce"}]}],"author":{"name":"hagun.kim","email":"hagun.kim@samsung.com","date":"2019-08-16 13:48:47.000000000","tz":-420},"committer":{"name":"hagun.kim","email":"hagun.kim@samsung.com","date":"2019-08-20 07:32:19.000000000","tz":540},"subject":"Add dhclient package for Octavia","message":"Add dhclient package for Octavia\n\nTo support openstack-helm Octavia chart,\n\ndhclient package is added to Octavia loci.\n\nRefer to https://review.opendev.org/630821\n\nChange-Id: I4ff7dbd28220da07f4616970be12abcb8de58ceb\nSigned-off-by: hagun.kim \u003chagun.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b4bdec997baa7bcef6ad403de4ac7e22cf60409f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b4bdec997baa7bcef6ad403de4ac7e22cf60409f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
