)]}'
{"id":"openstack%2Floci~677033","triplet_id":"openstack%2Floci~master~Ibf88bd2113029d750f9a4f32e288e3d8be3a1bfd","project":"openstack/loci","branch":"master","topic":"ldaplongterm","hashtags":[],"change_id":"Ibf88bd2113029d750f9a4f32e288e3d8be3a1bfd","subject":"python-ldap fix","status":"MERGED","created":"2019-08-16 21:14:22.000000000","updated":"2019-08-23 15:52:34.000000000","submitted":"2019-08-23 15:52:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677033-1566575554971-456e8807","meta_rev_id":"4678030d9bde64de26e021edfae0542e5073b213","_number":677033,"virtual_id_number":677033,"owner":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-16 21:16:48.000000000","_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":2,"date":"2019-08-23 15:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":2,"date":"2019-08-23 14:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-20 14:54:33.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":1,"date":"2019-08-23 14:29:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 21:15:27.000000000","updated_by":{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},"reviewer":{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},"state":"REVIEWER"},{"updated":"2019-08-20 14:54:33.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-08-23 14:29:29.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-08-23 15:52:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b6497e60dbd16d8bdde77e4798aefff1fb66631","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2019-08-16 21:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"126aaf94ca502a1c84a6ff1fe8ce95f30a02ebc8","author":{"_account_id":7769,"name":"Pentheus","display_name":"Alan Meadows","email":"alan.meadows@gmail.com","username":"alanmeadows"},"date":"2019-08-16 21:15:27.000000000","message":"Patch Set 1:\n\nHey little guy!","accounts_in_message":[],"_revision_number":1},{"id":"091b8d2100b39e1475df51f97d09bf41b4322eed","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2019-08-16 21:16:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d6819f229d6b9ede16d87de75362c305976737b","author":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"date":"2019-08-16 21:16:48.000000000","message":"Patch Set 2:\n\n\u003e Hey little guy!\n\nhey bud! its been a minute","accounts_in_message":[],"_revision_number":2},{"id":"1c87eacb1451b3ac1ab24f1cc5903e349a57f565","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 22:11:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9b2c398683a64820a8971697f7661b80 : SUCCESS in 39m 26s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d20e0d53621542c4821b141ed9b2b98c : SUCCESS in 34m 48s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e5bcbe9c5f5c42838da164f7d3fc39b6 : SUCCESS in 36m 27s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c5a6937fc6574bca94a383b2a9f0e361 : SUCCESS in 49m 47s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/42854b5d84c848018f42f8ac2ae86caf : SUCCESS in 36m 52s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/35d3bfda0e0a435b88b9c467adae6859 : SUCCESS in 36m 27s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/54db9fa16e9a4c1db1c1a834d5703fa7 : SUCCESS in 44m 42s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8f31ac2a048b478d887e5adf64ba019a : SUCCESS in 38m 43s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/9aabfbd046284874a67e196f176d9cb5 : SUCCESS in 36m 18s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ee5436df47e54fa991e34fc25f74d195 : SUCCESS in 47m 14s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/06e656e6ddc549858fa2192cc4d95026 : SUCCESS in 27m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e861be7abcfb45223827b75f4ecd4af1b259acb4","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-08-20 14:54:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dda7b350d7c4ba481c6832360939b1359f9b2467","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-08-23 14:29:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9478679bdce6b524900d968701b141cb3bab6869","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 14:29:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ebd56717bdb5f999735653b797383b7924c16666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:52:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/436eea1462814d379a37b8fa8121e7b7 : SUCCESS in 39m 29s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d5439ee2fdae403bb57c5e446236d349 : SUCCESS in 46m 20s\n- loci-heat https://zuul.opendev.org/t/openstack/build/93c0681c0f464313bd8579236557d3b1 : SUCCESS in 35m 46s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a8b226a27f434b48a620b7c9c7a2ab7d : SUCCESS in 36m 31s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/37cbdb35cf0f447c8dc7c75e35872c66 : SUCCESS in 38m 20s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/742f3d4730bb4df6a414e5e4387facc3 : SUCCESS in 37m 18s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3cc04c563dfc40daa9c856980b0b89e6 : SUCCESS in 36m 37s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/1f4fdbdf29fc4979b64005723d44b33a : SUCCESS in 50m 56s\n- loci-nova https://zuul.opendev.org/t/openstack/build/aa763b6f21a441b8bbda9a2f47ef8e85 : SUCCESS in 51m 39s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3a13b725babe49299e872fe7ba246ada : SUCCESS in 38m 47s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c466a93ba61f43d094494cd2a0c80ad2 : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4fb752bf5d80116422eba73afc042afdd19c818c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:52:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfec5ade4c323a01b7b50edd604bd3379c49b376","revisions":{"bf32f636b2cbef68f525172bce86dfb24cebce92":{"kind":"REWORK","_number":1,"created":"2019-08-16 21:14:22.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/33/677033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/33/677033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/677033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/677033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/677033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/33/677033/1"}}},"commit":{"parents":[{"commit":"3c3caa0765781d76c852d336101b055e0d972058","subject":"Merge \"Revert \"Remove explict enum-compat\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3c3caa0765781d76c852d336101b055e0d972058"}]}],"author":{"name":"Sam Yaple","email":"sam+git@yaple.net","date":"2019-08-16 21:12:20.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam+git@yaple.net","date":"2019-08-16 21:12:20.000000000","tz":-240},"subject":"python-ldap fix","message":"python-ldap fix\n\nIn rocky and beyond pyldap becomes python-ldap. The structure of the\nprevious patch to fix them means we must always update the pydep.txt\nlist with a new release of openstack.\n\nReordering this we can explictly install only when the version is older\nthan rocky. Added openstack releases back to the Newton release for\ncompatibility.\n\nChange-Id: Ibf88bd2113029d750f9a4f32e288e3d8be3a1bfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bf32f636b2cbef68f525172bce86dfb24cebce92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bf32f636b2cbef68f525172bce86dfb24cebce92"}]},"branch":"refs/heads/master"},"dfec5ade4c323a01b7b50edd604bd3379c49b376":{"kind":"REWORK","_number":2,"created":"2019-08-16 21:16:19.000000000","uploader":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"ref":"refs/changes/33/677033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/33/677033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/677033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/677033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/33/677033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/33/677033/2"}}},"commit":{"parents":[{"commit":"3c3caa0765781d76c852d336101b055e0d972058","subject":"Merge \"Revert \"Remove explict enum-compat\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3c3caa0765781d76c852d336101b055e0d972058"}]}],"author":{"name":"Sam Yaple","email":"sam+git@yaple.net","date":"2019-08-16 21:12:20.000000000","tz":-240},"committer":{"name":"Sam Yaple","email":"sam+git@yaple.net","date":"2019-08-16 21:16:08.000000000","tz":-240},"subject":"python-ldap fix","message":"python-ldap fix\n\nIn rocky and beyond pyldap becomes python-ldap. The structure of the\nprevious patch to fix them means we must always update the pydep.txt\nlist with a new release of openstack.\n\nReordering this we can explictly install only when the version is older\nthan rocky. Added openstack releases back to the Newton release for\ncompatibility.\n\nChange-Id: Ibf88bd2113029d750f9a4f32e288e3d8be3a1bfd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfec5ade4c323a01b7b50edd604bd3379c49b376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfec5ade4c323a01b7b50edd604bd3379c49b376"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
