)]}'
{"id":"openstack%2Floci~678120","triplet_id":"openstack%2Floci~master~I06d50b690a2425336a8b81ef3c5fb58e5830f86b","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I06d50b690a2425336a8b81ef3c5fb58e5830f86b","subject":"Fix error when build keystone image on Ubuntu 18.04","status":"ABANDONED","created":"2019-08-23 01:48:11.000000000","updated":"2020-04-22 00:40:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"065941cc0f25ff495788999d633d8007619fe455","_number":678120,"virtual_id_number":678120,"owner":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-23 02:34:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-23 14:14:04.000000000","_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 02:34:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72405798d41f844784a1736f5e3da861b969fe07","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2019-08-23 01:48:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"820b5b1d5370326ec3b44a3ade8c8cdddc390121","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 02:34:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/11b87c2fc4d945ab81a5b3d7405f206a : SUCCESS in 33m 26s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a7a82b811c72433ab9615820b36912cd : SUCCESS in 37m 54s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0a7ffc4be5b042ab8ad7cddd6a54df7a : SUCCESS in 38m 51s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ffd0355a9a84472dafa90501b9a8c7ac : SUCCESS in 43m 06s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1cf29079f8ca452cb0abd748bb4e6faf : SUCCESS in 33m 18s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/70319bf474b74beea56bd31b5b1d005a : SUCCESS in 40m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/96f5cf70d5b64e7abefa8768263a955b : SUCCESS in 38m 56s\n- loci-nova https://zuul.opendev.org/t/openstack/build/db0ee89c5de04bc3ad0518ec11d5d8d9 : SUCCESS in 42m 51s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/efec508f73c640759fc8bf27645d207f : SUCCESS in 35m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0c167f6cf2ee4fff9fd7ad924988ebc5 : SUCCESS in 37m 34s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b24db32170cf4fe094d304846822c642 : SUCCESS in 26m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e992cd6d2f06916c6f42c5e14b33aa372baa7c34","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-08-23 14:04:32.000000000","message":"Patch Set 1: Code-Review-1\n\nI don\u0027t think that the base images are the place for e the dev packages. This is going to pull in a lot of extra stuff. bindep.txt is the place to handle this","accounts_in_message":[],"_revision_number":1},{"id":"3e3b7bc16c66ad1b9652b0471f31cf66f5f5c393","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2019-08-23 14:14:04.000000000","message":"Patch Set 1:\n\n@Chris, if ${PLUGIN} is \"no\", the packages can not be installed before installing ${PYDEP_PACKAGES}.","accounts_in_message":[],"_revision_number":1},{"id":"f7146e088359e56e388e30942fd718a6ead955fd","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2019-11-07 21:05:59.000000000","message":"Removed reviewer Chris Hoge with the following votes:\n\n* Code-Review-1 by Chris Hoge \u003cchris@hogepodge.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"065941cc0f25ff495788999d633d8007619fe455","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2020-04-22 00:40:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aea13d42869054db7ae6d763b5065bce1587abd2","revisions":{"aea13d42869054db7ae6d763b5065bce1587abd2":{"kind":"REWORK","_number":1,"created":"2019-08-23 01:48:11.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/20/678120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/20/678120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/678120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/678120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/20/678120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/20/678120/1"}}},"commit":{"parents":[{"commit":"3c3caa0765781d76c852d336101b055e0d972058","subject":"Merge \"Revert \"Remove explict enum-compat\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3c3caa0765781d76c852d336101b055e0d972058"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2019-08-23 01:44:53.000000000","tz":480},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2019-08-23 01:44:53.000000000","tz":480},"subject":"Fix error when build keystone image on Ubuntu 18.04","message":"Fix error when build keystone image on Ubuntu 18.04\n\nUse ubuntu 18.04 base image to build keystone, missing\nsome packages including libldap2-dev/libsasl2-dev/\nlibffi-dev/libssl-dev.\n\nChange-Id: I06d50b690a2425336a8b81ef3c5fb58e5830f86b\nCloses-Bug: #2006432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/aea13d42869054db7ae6d763b5065bce1587abd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/aea13d42869054db7ae6d763b5065bce1587abd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
