)]}'
{"id":"openstack%2Floci~685870","triplet_id":"openstack%2Floci~master~I278f2eb264668becdf9eba479d3f1ddf8da563e9","project":"openstack/loci","branch":"master","topic":"allow_custom_service","hashtags":[],"change_id":"I278f2eb264668becdf9eba479d3f1ddf8da563e9","subject":"allow pass custom service name","status":"ABANDONED","created":"2019-10-01 02:42:14.000000000","updated":"2020-07-24 04:42:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d54e16fd4ba90a1fe4850d2dbbd7d9536e906c90","_number":685870,"virtual_id_number":685870,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-01 06:49:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-01 05:51:27.000000000","_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 06:49:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6370889fee43d0c3204009a77f8607c7b595ad9a","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2019-10-01 02:42:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"646c101fe816ba12a857a0de03686daf96e1fe5c","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2019-10-01 03:28:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1b7f1a945bd1d7696adfbb02a0a038ebbf343b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 04:25:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/72f666e6ff7c44ad8f38e13549fa97a0 : FAILURE in 32m 31s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ab3cffb0ec7a47259602edf67d22ba7a : SUCCESS in 51m 23s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3a2f2f873c9f41359165b80250c4b2aa : SUCCESS in 31m 35s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e67711e8be7e4b7eb61618b0280b0e1e : SUCCESS in 47m 29s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3254f9cd3f19481d91f3dcbf06ccc057 : SUCCESS in 39m 50s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/81e82c044c774658baecf00bf1e418c3 : SUCCESS in 51m 15s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e62f5b6609eb47c9966364fb999b9c42 : SUCCESS in 50m 59s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/58e46574ff284ecf87828ad452cc16e1 : SUCCESS in 53m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ed91a42b9c184dcb95377ee04e936191 : SUCCESS in 49m 15s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/80ba039584bf47888d38c4b55227959c : SUCCESS in 47m 24s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2a75d99f6e064010b6e48e576b90dd7f : FAILURE in 27m 34s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c9731048a33d48d9b16626df9ef76dd2 : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0a268a7ef8990c17dd8b7776bd3dd088cb7df27d","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2019-10-01 05:51:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"88d4f4aa8be7a85c57127c2d3816124e567606bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 06:49:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f5aec1545bb44540b50657e6cfc17cb6 : SUCCESS in 47m 22s\n- loci-glance https://zuul.opendev.org/t/openstack/build/46d213ee338b4bacb2be2ffbeec3ce0d : SUCCESS in 43m 07s\n- loci-heat https://zuul.opendev.org/t/openstack/build/80f364a1796940d4bbda9caf2055f3fe : SUCCESS in 37m 58s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3a66031568f741519d9f2d77503ad400 : SUCCESS in 38m 13s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f4dbc5a3847543888c30284ea4f096b9 : SUCCESS in 37m 30s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5ed16811791f47e39c618a8c02d6586b : SUCCESS in 41m 30s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8d1b11f05a564b6493c9918f071b8c20 : SUCCESS in 36m 44s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/0f0ce2d0b7db40fc8881610f25fd3ea2 : SUCCESS in 38m 16s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f86d5f5dbda94f9c92581b0d127e3513 : SUCCESS in 52m 25s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f2ae93c903e446e489fba0c59e6af243 : SUCCESS in 38m 23s\n- loci-placement https://zuul.opendev.org/t/openstack/build/988b71b2bb7446f8983ad9651f0b6e5c : SUCCESS in 38m 20s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0e8fc677407844a58d3b7645f7086fc3 : SUCCESS in 34m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f1738521fdff98d6e51d002f82d8956925cc62a4","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-07-24 04:42:16.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"788221e52968a5d551473ea2d4d803850a3d99dd","revisions":{"b126c56105551b88ba4af795c2cd246e754921e6":{"kind":"REWORK","_number":1,"created":"2019-10-01 02:42:14.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/70/685870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/70/685870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/70/685870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/70/685870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/70/685870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/70/685870/1"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2019-10-01 01:57:31.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2019-10-01 02:42:08.000000000","tz":480},"subject":"allow pass custom service name","message":"allow pass custom service name\n\nfor harbor registry, domain name may not equal\nto service name\n\nChange-Id: I278f2eb264668becdf9eba479d3f1ddf8da563e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b126c56105551b88ba4af795c2cd246e754921e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b126c56105551b88ba4af795c2cd246e754921e6"}]},"branch":"refs/heads/master"},"788221e52968a5d551473ea2d4d803850a3d99dd":{"kind":"REWORK","_number":2,"created":"2019-10-01 03:28:36.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/70/685870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/70/685870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/70/685870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/70/685870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/70/685870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/70/685870/2"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2019-10-01 01:57:31.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2019-10-01 03:10:52.000000000","tz":480},"subject":"allow pass custom service name","message":"allow pass custom service name\n\nfor harbor registry, domain name may not equal\nto service name\n\nChange-Id: I278f2eb264668becdf9eba479d3f1ddf8da563e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/788221e52968a5d551473ea2d4d803850a3d99dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/788221e52968a5d551473ea2d4d803850a3d99dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
