)]}'
{"id":"openstack%2Floci~687221","triplet_id":"openstack%2Floci~master~I781457ceeb821a0cad7edc9c0921736b989e9ec7","project":"openstack/loci","branch":"master","topic":"rm","hashtags":[],"change_id":"I781457ceeb821a0cad7edc9c0921736b989e9ec7","subject":"Add collect_info.sh stage","status":"MERGED","created":"2019-10-08 07:13:15.000000000","updated":"2020-07-20 20:49:00.000000000","submitted":"2020-07-20 20:49:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687221-1595278140301-dfe663bb","meta_rev_id":"31f70bae2c734fcee0da94afdbe93dd4981ea9d7","_number":687221,"virtual_id_number":687221,"owner":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 20:49:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"date":"2020-07-20 09:02:17.000000000","_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2020-07-20 19:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":1,"date":"2020-07-13 18:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":2,"date":"2020-07-13 22:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2020-07-20 19:47:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 12:32:13.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2019-10-08 12:32:13.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2019-10-08 12:32:13.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-11-13 10:39:38.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-12-12 20:20:21.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2020-05-15 10:54:14.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-07-13 22:15:27.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2020-07-20 19:47:37.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-07-20 20:49:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1557269e5518362eee7d0d3586773c3683b2516","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2019-10-08 07:13:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd109e635f03cce2428f130fdd4c0412fc0ee288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 08:17:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/2d5f39d97957429fa49c26459608e1cc : SUCCESS in 40m 09s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9d02a78f1cc44381876aeafceb43f0e9 : SUCCESS in 38m 29s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7aefb746b96f42fca717de0f3af2e3ed : SUCCESS in 37m 11s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4d2d280498ab42c8b2e4418c2a2d0f96 : SUCCESS in 43m 25s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e807cc23e7144fc5ad762683035bc46c : SUCCESS in 38m 00s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9956e4d247214c68b184298d2c8daf9b : SUCCESS in 39m 12s\n- loci-manila https://zuul.opendev.org/t/openstack/build/dba883c028b54f06a34681bff22c1faf : SUCCESS in 37m 58s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/68e8c78e32124bb6ac227c05e510a887 : SUCCESS in 41m 30s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a2bb54e00fa84b9bbb606de58064bae6 : SUCCESS in 1h 00m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/107393dfc2c84ae892a78db80bef8469 : SUCCESS in 42m 09s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c8a5f43199904cf8b448bc52cf74778f : SUCCESS in 37m 19s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4f196b4859eb40658c2addfa333a2856 : SUCCESS in 29m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c6fbfc9e58ea54c9be239145cc2ed53e96c77f8c","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2019-11-07 10:32:31.000000000","message":"Patch Set 1:\n\nbump, please review","accounts_in_message":[],"_revision_number":1},{"id":"04968bac9c856886109ec43136de7f1d58ba4a55","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-13 10:39:38.000000000","message":"Patch Set 1:\n\nCan we make this optional?","accounts_in_message":[],"_revision_number":1},{"id":"370870b52a36618222dcda255c85a2ecc8fbcc87","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2019-12-12 19:15:17.000000000","message":"Patch Set 1:\n\nbump, please review","accounts_in_message":[],"_revision_number":1},{"id":"836ac8c2b0ed84d1873e121ec3a045f0926475f9","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-12-12 20:20:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"508b304b861c78c8b7814c36a1d59d200bf52161","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2020-02-03 15:14:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6448e1bc1ae7db69d362c1f57acea0f3701e9319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 16:24:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5684a7d914764a0089b579ad274d1c53 : SUCCESS in 42m 49s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c1f40ff558fb4d32927f460c15c451a6 : SUCCESS in 40m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/6cfa5d91b0794e0a881d17a2253dd234 : TIMED_OUT in 1h 01m 10s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a70f7bf1fa6246c496cc1c8b34cb53e2 : SUCCESS in 42m 14s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/4a3e6b2a9d394fa9ae89c541b76ffe7c : SUCCESS in 40m 05s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1da42b9370d444099e5e564a2cbc2177 : SUCCESS in 42m 03s\n- loci-manila https://zuul.opendev.org/t/openstack/build/cc1024d8ae9648f2ad837b2fa6b7246b : SUCCESS in 43m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8d1aa50e2ad444788e9a7a2483e370ba : SUCCESS in 45m 54s\n- loci-nova https://zuul.opendev.org/t/openstack/build/711a93ccbb734669a8b47ddc5ebbaca7 : TIMED_OUT in 1h 01m 31s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/59f10d5377d0472abf215a78505d3487 : SUCCESS in 43m 35s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4de1741b429a4202877205e1ce01d5a4 : SUCCESS in 35m 19s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/283e692bd8fa4324bcfe72c2205aecca : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6987775394d8a591be8b7065c99370f9e0261170","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-15 10:54:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"96e463de07eb4df7257c091af7fe1c5db7cf88ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 11:58:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3404d615ad784ee49880439aca2d6919 : SUCCESS in 42m 04s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ea80b813492242ba8ef50c55254bd0f9 : SUCCESS in 53m 28s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d411a702700641b6b9faa2e3e9f25a67 : SUCCESS in 42m 28s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/1f3ef06b00094586a4387e93ebfb98fe : SUCCESS in 50m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e68138420b18405284039fccabb25790 : SUCCESS in 50m 04s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/40a9f874983647a2b51ee09ebada8a9e : SUCCESS in 50m 35s\n- loci-manila https://zuul.opendev.org/t/openstack/build/12701d052ffb4240942525a691915f04 : SUCCESS in 46m 54s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ec10ef69c5e84b839a14da7f01a128a9 : SUCCESS in 52m 45s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5c3dbb15360a4a73af76e920839f4bc7 : SUCCESS in 58m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/83136f036959490d8be2a93c96392d7f : SUCCESS in 56m 59s\n- loci-placement https://zuul.opendev.org/t/openstack/build/60b747af4e9e4eccaf9f19439a3722b0 : SUCCESS in 43m 35s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/78b2eea3cddd40cc9c8d8170b8c696bf : TIMED_OUT in 41m 12s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/01f4096d19404896aeb957c0fcac22f7 : SUCCESS in 45m 50s","accounts_in_message":[],"_revision_number":2},{"id":"0a21a0ea8f60f2c6c7aa2c814a8387eb31850d87","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-05-20 18:01:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"741d1ae24ad76b9d1cfdc767c2c5f128a3371e92","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-05-20 18:01:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"37097d275a02b150f09b5f19c215ec628c04d9d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 18:58:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5faa0aef42ca48b7a84c06e4187dbab9 : SUCCESS in 37m 54s\n- loci-glance https://zuul.opendev.org/t/openstack/build/012a4b600c0d4f7e83d1c21e83fa4a6f : SUCCESS in 33m 45s\n- loci-heat https://zuul.opendev.org/t/openstack/build/34333b191de44692818e9b8e0f32b54b : SUCCESS in 32m 41s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6982f5d6e633407889ac60976644f0bd : SUCCESS in 53m 35s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5be5816407a1404592edf8678c7b6a78 : SUCCESS in 39m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d1017f1522794ad7b662158c8b720398 : SUCCESS in 34m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/01a6bb70764348c6bdc9b2bf0a9aa967 : SUCCESS in 35m 36s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b36e1af198a74716be9fe1c9989aa678 : SUCCESS in 39m 50s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5e17d7233a2c4011a4287bb96538329b : SUCCESS in 38m 15s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5eae9c163e8b4495ad1e269a1962f12d : SUCCESS in 38m 48s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a307fbdf670c4197b41d48dffa074d44 : SUCCESS in 26m 42s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e634c4fd0e564df7ba589eae2b068195 : SUCCESS in 27m 59s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/47c370e0d66b47d3b2ca79605e4b26f3 : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":3},{"id":"09d02d1004665832174daf958d87746ad2bb5d73","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-07-13 18:34:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f3ad172debee28a56577f096edc2069ad9b465b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 19:25:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f18f4fdb97ef4541a10c52c61733b940 : SUCCESS in 39m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2c78d4993bdf4f03bd9ce7ecddab0d67 : SUCCESS in 41m 45s\n- loci-heat https://zuul.opendev.org/t/openstack/build/539d0c7bf775435abc72661dd13cba0e : SUCCESS in 40m 48s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2e7d3e3337194951a6439a7ce0725286 : SUCCESS in 39m 38s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5dc287e6ec844664b58b5605310cc1fb : SUCCESS in 38m 54s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9b445d758064463e98f94b2b86cf8cce : SUCCESS in 29m 49s\n- loci-manila https://zuul.opendev.org/t/openstack/build/02d4ac1ccc3f4aac8c9730fc325ad9ad : SUCCESS in 44m 41s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8636851223534be8b915e51ab5a60342 : SUCCESS in 41m 38s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7f26352cfd3e43fea0d40c835bcb9801 : SUCCESS in 42m 45s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4ce565ad4d9746d9bc888930dc534f0e : SUCCESS in 45m 35s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b585e503970b4a2faf9c77df8fa820e1 : SUCCESS in 32m 23s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1b8053e439744242b8281000b22a2ad4 : SUCCESS in 35m 25s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2a888b9531694bb694f2c4e000edace4 : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":4},{"id":"475e731c08c46feb64af746a2225030608257d8c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-07-13 22:15:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"13698cde012e87d9ef3f3f6099a7d84c4b5f76d0","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2020-07-20 09:02:17.000000000","message":"Patch Set 4:\n\nbump plz review","accounts_in_message":[],"_revision_number":4},{"id":"f312c723f878ca1eb66109d55dbf11da3d3fff43","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-07-20 19:47:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"351d897eebb3d688cb9dcf1bc8fcd0d415464873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 19:47:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cb36714d2c8d7ea1a091ac1687a388b14846d1fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 20:49:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"cb32ac1648b9a6eb93aa3c66c305d7d867da5ff9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 20:49:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/2e684bd189a941a68586292fe8859f6a : SUCCESS in 21m 36s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d43bec75684c4b9ebbe418d1d5b2bc61 : SUCCESS in 30m 45s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e7f2f8e302ce4c158e037d1bd5f6426a : SUCCESS in 41m 20s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/face70bd856b49438fa15440e9c3f0f4 : SUCCESS in 43m 45s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9d7178421bd4479186f25f9dd8f205d6 : SUCCESS in 41m 25s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/bed44e24ef3e47f1b2ffda2570f605f5 : SUCCESS in 46m 15s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c6b13d8486e44c9ab10c7a22bf49d1c4 : SUCCESS in 33m 19s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/272eba228c424c70baebae4bcade01fb : SUCCESS in 53m 24s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b36c934cee784694b0548656255429d8 : SUCCESS in 33m 04s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/02ead90f0256460e8b6dc123d35b8875 : SUCCESS in 42m 31s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ca0e31db569446e796f642f7b0627f08 : SUCCESS in 29m 39s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/37bb4056aaa8436aa16f9db32a3b8eaa : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ec429cb082ef1c1dc5e781a131cbd61aecf3acd3","revisions":{"967f1aa1a2eab235382d2e77372c02737142462f":{"kind":"REWORK","_number":1,"created":"2019-10-08 07:13:15.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/21/687221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/21/687221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/21/687221/1"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2019-10-08 06:24:38.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2019-10-08 07:13:04.000000000","tz":180},"subject":"Add collect_info.sh stage","message":"Add collect_info.sh stage\n\nCollect info about installed packages versions and\npip packages into /etc/image_info folder\n\nChange-Id: I781457ceeb821a0cad7edc9c0921736b989e9ec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/967f1aa1a2eab235382d2e77372c02737142462f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/967f1aa1a2eab235382d2e77372c02737142462f"}]},"branch":"refs/heads/master"},"81ad39fbec5236ff8fc863311a7bce7c882be3ca":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-03 15:14:19.000000000","uploader":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"ref":"refs/changes/21/687221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/21/687221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/21/687221/2"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2019-10-08 06:24:38.000000000","tz":180},"committer":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2020-02-03 15:14:19.000000000","tz":0},"subject":"Add collect_info.sh stage","message":"Add collect_info.sh stage\n\nCollect info about installed packages versions and\npip packages into /etc/image_info folder\n\nChange-Id: I781457ceeb821a0cad7edc9c0921736b989e9ec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/81ad39fbec5236ff8fc863311a7bce7c882be3ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/81ad39fbec5236ff8fc863311a7bce7c882be3ca"}]},"branch":"refs/heads/master"},"e9e8d855a1dfeb249a18923b0c0cf6e54bf99ff3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-20 18:01:18.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/21/687221/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/21/687221/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/21/687221/3"}}},"commit":{"parents":[{"commit":"f022ecba553903df3df72d3668e143e9eb9ceded","subject":"Merge \"openstackhelm/nova:rocky-ubuntu_bionic missing python3-rbd/rados lib.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f022ecba553903df3df72d3668e143e9eb9ceded"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2019-10-08 06:24:38.000000000","tz":180},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-05-20 18:01:18.000000000","tz":0},"subject":"Add collect_info.sh stage","message":"Add collect_info.sh stage\n\nCollect info about installed packages versions and\npip packages into /etc/image_info folder\n\nChange-Id: I781457ceeb821a0cad7edc9c0921736b989e9ec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e9e8d855a1dfeb249a18923b0c0cf6e54bf99ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e9e8d855a1dfeb249a18923b0c0cf6e54bf99ff3"}]},"branch":"refs/heads/master"},"ec429cb082ef1c1dc5e781a131cbd61aecf3acd3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-13 18:34:19.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/21/687221/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/21/687221/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/21/687221/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/21/687221/4"}}},"commit":{"parents":[{"commit":"938a110f307d82400e2d83c09fa72ed24931a6b5","subject":"Merge \"Add delimiter for service image build example\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/938a110f307d82400e2d83c09fa72ed24931a6b5"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2019-10-08 06:24:38.000000000","tz":180},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-07-13 18:34:19.000000000","tz":0},"subject":"Add collect_info.sh stage","message":"Add collect_info.sh stage\n\nCollect info about installed packages versions and\npip packages into /etc/image_info folder\n\nChange-Id: I781457ceeb821a0cad7edc9c0921736b989e9ec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ec429cb082ef1c1dc5e781a131cbd61aecf3acd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ec429cb082ef1c1dc5e781a131cbd61aecf3acd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
