)]}'
{"id":"openstack%2Floci~689028","triplet_id":"openstack%2Floci~master~I86051fc48af907be05cfdd4fbb86d35aae704c4d","project":"openstack/loci","branch":"master","topic":"keystone_ldap_fix","attention_set":{},"removed_from_attention_set":{"9373":{"account":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"last_update":"2022-12-19 20:15:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I86051fc48af907be05cfdd4fbb86d35aae704c4d","subject":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile","status":"MERGED","created":"2019-10-16 19:53:56.000000000","updated":"2022-12-19 20:15:03.000000000","submitted":"2022-12-19 20:15:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"689028-keystone_ldap_fix","meta_rev_id":"6301b92dbf888e6d52acdf2c33996ede535dbde7","_number":689028,"virtual_id_number":689028,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-19 20:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":2,"date":"2022-12-18 18:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-19 19:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-19 19:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 20:28:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 16:11:20.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-12-18 18:33:29.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"}],"messages":[{"id":"cddf223e133d44c0d6bf33e2aa8725b80ab636a4","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-10-16 19:53:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebc67b36fe969e2b37125ab6a532ea96ebdbf5ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 20:40:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5efbd1280e4741e79080a32b8e26e281 : SUCCESS in 32m 24s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c0276a344830404497afd17489826244 : FAILURE in 32m 59s\n- loci-heat https://zuul.opendev.org/t/openstack/build/75923a8fe7c14380beefd066e6667a6d : SUCCESS in 33m 06s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5f36810330134fdb81d0aa805a629336 : SUCCESS in 36m 08s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/96d26dec484f461fb5907762331c7db2 : FAILURE in 33m 10s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2c3160db881e4372a6cad798a2251dc4 : SUCCESS in 37m 47s\n- loci-manila https://zuul.opendev.org/t/openstack/build/d7dbcb0ba9064de98c083231b9e68d8f : SUCCESS in 31m 44s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ae2b03ff8a174033b7a3ebed0c21cb76 : SUCCESS in 42m 28s\n- loci-nova https://zuul.opendev.org/t/openstack/build/96e88643fdb44fe3ae0b4bc1215d4818 : SUCCESS in 45m 12s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/576917a9210640cda21123fdb8cc606d : FAILURE in 32m 56s\n- loci-placement https://zuul.opendev.org/t/openstack/build/9d50ca7c25b94d69ae4db615be2634dd : SUCCESS in 37m 24s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/081f6133ad9c4725b6ebdda9e4d75cfa : SUCCESS in 24m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e6a83d25ec67dd8048e9e924a24cf56f4d59fe8f","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-10-16 20:47:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"38727ae7c433eea594ab6b69fdfd3e251935d5ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 21:48:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7c8d7fae16164c329307cc311ac82a65 : SUCCESS in 54m 04s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1488b9a06dff424cbc794444ea942f2b : SUCCESS in 53m 29s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e7acc574994544c6a8073aad315f14f2 : SUCCESS in 50m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6823c481ae3a47f99e1561050c62355f : SUCCESS in 53m 17s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/504cf0a72d4d44fd9805f86111996596 : FAILURE in 29m 50s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f7cb5c31e1824d168b7d9b349d969c7e : FAILURE in 25m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/cebdaf82013c45e091f77c37b072931a : SUCCESS in 55m 25s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/abde34880f6f4edf99765f92ffa8b4e6 : FAILURE in 33m 20s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d5bee7b6cd76465c89d097068b7f7980 : SUCCESS in 42m 51s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/42b266df7f3c4ef488695f92ca2d3b59 : SUCCESS in 41m 26s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cef6ab41862b4cd2b3f904e78b14c196 : SUCCESS in 35m 22s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a47bdab9315646509b2cdde8dac92121 : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b5778e142e8bde801c849b58c9ada15612f32828","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-10-18 09:15:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e4b539599c275962b10a57bc461b0cb8fab46d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 10:06:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0dfa7c66787e48e882824969ce071dbf : SUCCESS in 43m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b4bab73678a7490c9ad4c1d929a8f664 : SUCCESS in 41m 15s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b5e5391c23a84a078febb06bc36372c4 : SUCCESS in 40m 27s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/021e069995ed473fa2c7c46b0669151d : SUCCESS in 40m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/096349044d1f4379b772772cf261b5a5 : SUCCESS in 34m 26s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/43eade9d142140c5891a0521862e58fe : SUCCESS in 32m 26s\n- loci-manila https://zuul.opendev.org/t/openstack/build/922e63b2e8104ea0a7a7dd1d43c01714 : SUCCESS in 44m 38s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b0dece2684d1490a9758c22ac592370c : SUCCESS in 43m 41s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b6c5104c37434a33b5c07c088942d90d : SUCCESS in 40m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/52f57ea3d3774dcd938c71c2a07805e1 : SUCCESS in 31m 34s\n- loci-placement https://zuul.opendev.org/t/openstack/build/7252b158b87d4ce8b73c5da9a180d991 : SUCCESS in 36m 42s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d6458ff64d1d48d88f8a19976e101a23 : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":1},{"id":"20314ad52d0ceff20954243d4fa377d83231ca9a","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-08-26 19:24:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9e7b3267b584aa70313c67ad3e996dca03a8d281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:28:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/853af2782d5d4951a9662bff18c8e87e : SUCCESS in 19m 17s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b8e217001f584d2e9854a6bb8920ef7f : SUCCESS in 35m 26s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7764e87efcbc4b91ab4896a183b93181 : SUCCESS in 40m 20s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/714325cf75cc491fab55eb158c33bc1b : SUCCESS in 43m 12s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f048ea9bf06c48168d00ea36e2e26cda : SUCCESS in 32m 31s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/fcc27dfb16cd44e2b27c7160b776a7b4 : SUCCESS in 35m 48s\n- loci-manila https://zuul.opendev.org/t/openstack/build/16b91b4023604a058b2b8c00ebe45854 : SUCCESS in 38m 11s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a74ea5ca5ce3438ab86130d2d687bb22 : SUCCESS in 42m 38s\n- loci-nova https://zuul.opendev.org/t/openstack/build/dfd31229555a4494a5fad09c5f2e5199 : SUCCESS in 44m 04s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/be0e33597b664815ad66b3798f7ba271 : SUCCESS in 36m 21s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0bd09a4537a7417d9a7be68f9e10685c : SUCCESS in 28m 51s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/3b2fe7e8e523443a861b776ce7a87998 : SUCCESS in 25m 46s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f6728beb66c94c8a89a0a2a83c8ed6b8 : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4978c27cb074dbef41f54dff78831739b842900c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-30 15:57:46.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this or create a new change.","accounts_in_message":[],"_revision_number":2},{"id":"b865e3485a1e27f7c95472e2248820d2ed4d6bf2","tag":"autogenerated:gerrit:restore","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2021-09-30 16:08:58.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"5adfae049aa9a22783a4be4d6fa624140e442118","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2021-09-30 16:09:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"81c9e26f161297d711e6a78f4e9ea427d1b1c379","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-30 16:11:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5448571e2ed00c0342f3a95a97eb14f8fe4c0b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 16:50:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/b0efedd437ae48cd8cb07b54fb1c50c4 : FAILURE in 34m 58s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0e0bcc356bb749c8b0afd56cc682c6f8 : FAILURE in 35m 31s\n- loci-heat https://zuul.opendev.org/t/openstack/build/350c104fd29541a69602c4c15f99ee40 : FAILURE in 35m 01s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d4d15c0e97d44fe4ac5c5ffc28a69c5e : FAILURE in 35m 12s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d90257be9fbd41a69ebea744cc6c83db : FAILURE in 35m 06s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/df1aaa182be84f0087c92577893dc958 : FAILURE in 34m 45s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2820da9d603942fba81ca704778c1edf : FAILURE in 34m 56s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/dce0303167094a24a9d82621f0f74744 : FAILURE in 35m 05s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/08ddcc71c5314ec78f607eaee335c3de : FAILURE in 34m 42s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d609db2175f54e5e98d520539796fbdc : FAILURE in 34m 35s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/849efe2af16a442f8d986a8c4c7fda2b : FAILURE in 35m 16s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f25fcce79d2b4cd9953ab42b2e0c6496 : FAILURE in 35m 01s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b90b8af1f41f45eb9e5e88c2b2492ae9 : FAILURE in 34m 28s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c9b09663101445bfa16481d128a43fec : FAILURE in 34m 26s","accounts_in_message":[],"_revision_number":3},{"id":"f7295a7e62bd7b338e3f440fe63864fb341585a9","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2021-10-12 10:09:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8386b86dc9219f64d4317b8f622ab156784058f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 10:55:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a04bf59af53c46a5a49924d7b5eb3ce1 : FAILURE in 36m 05s\n- loci-glance https://zuul.opendev.org/t/openstack/build/93d3b1e610d449ab95155ea2e7ab7df6 : FAILURE in 41m 15s\n- loci-heat https://zuul.opendev.org/t/openstack/build/abf8bd19011a45f1b1341b2ef0bd8634 : FAILURE in 33m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/31ca941e7a044f7394024493aa205b00 : FAILURE in 37m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/62fb39e78f8e4f55b4187c4618466f51 : FAILURE in 39m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/0217f1b4a5bf40c2b7f8bff7f2a2238b : FAILURE in 43m 08s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3aa82ab312f54df08136ffd072fa9c9b : FAILURE in 32m 15s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2929a7d5b6854a02811679cf060c797b : FAILURE in 34m 14s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/65a4022ceb0c458d8d7eb83574b945a1 : FAILURE in 31m 39s\n- loci-nova https://zuul.opendev.org/t/openstack/build/69a6916d552d470d8ef989e754390443 : FAILURE in 39m 05s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4a9772a2b71848818e66f813275672b7 : FAILURE in 36m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4a3f99ad13ad43aeaf34d17e502a1f8a : FAILURE in 36m 56s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/95f66442619247ffa0aed16c04854f94 : FAILURE in 39m 29s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/fb81651206624cc4a2aeb6a8c3be1114 : FAILURE in 38m 58s","accounts_in_message":[],"_revision_number":3},{"id":"42566a07cfac424898a3fcbc3b1c9203a3afb3d8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-13 13:18:32.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this or create a new change.","accounts_in_message":[],"_revision_number":3},{"id":"3ea69202bd75fa8a7588fdab09de3eb5bdf2fc69","tag":"autogenerated:gerrit:restore","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2022-07-13 13:24:11.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"d8e89bd7f0dbf7b05e0482ec60138d8d254e50ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 13:24:32.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/loci for 689028,3","accounts_in_message":[],"_revision_number":3},{"id":"8227b5a43d1772084d71efde1fc7670936545f69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2022-07-13 13:29:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0e71b014b2a5fad97fe4ea37ef9815d428e605b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 13:45:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96b4cbe27d8a4b92be7b684748965e17\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/827b85ccc8824811b766b8211734cce5 : FAILURE in 4m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/dfc29d9db4d44b4e896a2535169aca4b : FAILURE in 6m 58s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a617cfba5b49473cafd01b36bb2ed980 : FAILURE in 4m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/01f21f9568b8451ebff336130c9141c8 : FAILURE in 4m 42s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/adf23745d31e47129d026dc21f563e46 : FAILURE in 4m 05s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/058dcf2fc91840578e125fd4e82dcfec : FAILURE in 4m 28s\n- loci-manila https://zuul.opendev.org/t/openstack/build/573263750b8f48cebc8aa12f4cd2d392 : FAILURE in 4m 33s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/9dcc58eaa1e14bf9b9d67efae87d5dbe : FAILURE in 4m 29s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8604398c6cac488c863e3c22194cdc56 : FAILURE in 4m 40s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fb89e991933b4cfbaa62eb2fb2551bae : FAILURE in 4m 25s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/bb554e30586440ceba5f0af62d3d24ce : FAILURE in 4m 33s\n- loci-placement https://zuul.opendev.org/t/openstack/build/101bd1a6a3414a48aaa1ffe081891439 : FAILURE in 4m 12s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f0c7d1019bf9427b967221d75622ec50 : FAILURE in 4m 17s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/431e8e105a614340a6c9b24048ce649b : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":4},{"id":"ab0e5a4c66c24bb10f981267f857db3adf5c436a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-07-25 22:52:09.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"62225992f06e84c3463df873b3262753e2c94482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 23:20:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/413826e8813f4d6482aa17589fe382d0\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d3c097a44b164a5da294c82e9c76eda6 : SUCCESS in 16m 39s\n- loci-glance https://zuul.opendev.org/t/openstack/build/260fd5a3572240e9af548bd0bc2599d2 : SUCCESS in 20m 12s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2214f70030664cb8aa71207696c503bf : SUCCESS in 18m 00s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/dbaccfe8e1dc40cbbc8343a043f66779 : SUCCESS in 21m 34s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/07a0a8c1b82d49ec991a48a2de4e7d9a : SUCCESS in 20m 29s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7a6bd8d7f0934333a333b434644a66aa : SUCCESS in 20m 32s\n- loci-manila https://zuul.opendev.org/t/openstack/build/11c830eaffd64534b34d46e09091a6e0 : SUCCESS in 20m 58s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/3a90377adba7455a98df4a5d07a68953 : SUCCESS in 19m 13s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/749c3765747b412aa2c4b12e5eb47872 : SUCCESS in 21m 59s\n- loci-nova https://zuul.opendev.org/t/openstack/build/dfdd4eae46c14544a5fc6973d9dce54d : SUCCESS in 21m 26s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3e524502dd1b422f86ea6dcc5262a04a : SUCCESS in 22m 16s\n- loci-placement https://zuul.opendev.org/t/openstack/build/908f45df1ea44bf9bce2ab254fd1150e : SUCCESS in 21m 41s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/821b8226eccc47fd9abbf5001cc8dd25 : SUCCESS in 17m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c49c13d80b204d889489f2b6e6c03ddb : SUCCESS in 21m 18s","accounts_in_message":[],"_revision_number":5},{"id":"3a1f813d5aed7af1b61822a3ff2d61477711f34a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-11-10 17:51:53.000000000","message":"Abandoned\n\nAuto-abandon: unmergable change is 107 days old.\n\nPlease restore if needed again.","accounts_in_message":[],"_revision_number":5},{"id":"e3b23170adbd848ae72abf73b4b02c4830db6f7f","tag":"autogenerated:gerrit:restore","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2022-12-06 16:33:07.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"e82da7479a7d9a826bb7fa073a2535ac7350405d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2022-12-06 16:33:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ce9c218a108f2561cf85c257bd74e2034cb9e021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 17:09:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dabd457150c74419ac15c52492ee4907\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3d2ac4af2cb740168a715ce03e818a35 : SUCCESS in 29m 41s\n- loci-glance https://zuul.opendev.org/t/openstack/build/138bd6867ad2416cb7d162c1273178c4 : SUCCESS in 18m 09s\n- loci-heat https://zuul.opendev.org/t/openstack/build/18e73ee51c834c1c968d125c750950c2 : SUCCESS in 17m 09s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/45ebfa857bf24778965fb104e8a0ef02 : SUCCESS in 17m 08s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e2be6b7601004d2a916450c855e57f64 : SUCCESS in 18m 41s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/fad631822d5b4067886526d5d6ea46c0 : SUCCESS in 18m 49s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2840f5b959e14750aa64ea9570563871 : SUCCESS in 16m 37s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/4e9a207821334ce68b2fc34841083b6f : SUCCESS in 18m 53s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a220dcb95b9e462191cc668df4f73c05 : SUCCESS in 18m 25s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b6001a781707404db7f0bd557a9727bf : SUCCESS in 20m 32s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a0e61bac7e164660ba1c91e0ea959bd2 : SUCCESS in 20m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/25de86ec8309435480b0ba7aaca3f2bb : SUCCESS in 22m 08s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8a0a56c4c8414fea8ed97e754a8bc77b : SUCCESS in 14m 52s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/661292278d5e442bba54198af4c29be2 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":6},{"id":"e374113f6bc223a04af4fae0b5f0bd25c0f4c083","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-12-18 18:33:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"fbe49d14b4398cf6a90a5a2fc435b9785893ca8d","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-12-19 19:48:33.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d9e12c5b377822048f23cd1f4a17c6e7e50edb10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 19:49:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"296a5551738d457a20f9743708470af092b9b1cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 20:15:03.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1959c63560eb402cbd18e6f8a4301c2f\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/54e6455095a845e2ab2047b154fe15b5 : SUCCESS in 21m 56s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9ca0a7c63f1a47769bf1ad5c476e0166 : SUCCESS in 19m 35s\n- loci-heat https://zuul.opendev.org/t/openstack/build/50bdc3be8d4b4cbf920a3ae604c24907 : SUCCESS in 23m 11s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/703da225b9394ca5a9aafa4139a0f133 : SUCCESS in 23m 38s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c2b9ef417b6a4894bb6ca798934006f1 : SUCCESS in 22m 33s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e7a3c013e85d4d2cbafba1fe05d3baa7 : SUCCESS in 22m 28s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f59a2ecd097e468ca1a848719883fbef : SUCCESS in 22m 19s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/08775edf95c14e939d5dba6c09b2eac1 : SUCCESS in 18m 47s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/093495e4f7af45419b6d8a02677afb7c : SUCCESS in 19m 48s\n- loci-nova https://zuul.opendev.org/t/openstack/build/9d9879d8e0af479c8afb0d2de77ff6f4 : SUCCESS in 23m 16s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/aa83eca0626c4204b908af724877e1e7 : SUCCESS in 22m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c4ba617ba6884d9f89f3e936b6c4d2fc : SUCCESS in 20m 38s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/eca9ed09c55248008b49798ead1cbc3f : SUCCESS in 15m 24s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/4420f47f9db646348215cb2d18751e2e : SUCCESS in 21m 42s","accounts_in_message":[],"_revision_number":6},{"id":"6301b92dbf888e6d52acdf2c33996ede535dbde7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 20:15:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f725a1256a2413fdada29b4d9d88acae63970ae9","revisions":{"90a338bd61ff51e4059b6f47acc2d87bdfc8702d":{"kind":"REWORK","_number":1,"created":"2019-10-16 19:53:56.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/28/689028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/689028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/689028/1"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-10-16 19:53:36.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-10-16 19:53:36.000000000","tz":180},"subject":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile","message":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile\n\nChange-Id: I86051fc48af907be05cfdd4fbb86d35aae704c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/90a338bd61ff51e4059b6f47acc2d87bdfc8702d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/90a338bd61ff51e4059b6f47acc2d87bdfc8702d"}]},"branch":"refs/heads/master"},"33b4f3af284b851a72a6925ed9be88919f4bebb0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-26 19:24:14.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/28/689028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/689028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/689028/2"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-10-16 19:53:36.000000000","tz":180},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-08-26 19:24:14.000000000","tz":0},"subject":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile","message":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile\n\nChange-Id: I86051fc48af907be05cfdd4fbb86d35aae704c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/33b4f3af284b851a72a6925ed9be88919f4bebb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/33b4f3af284b851a72a6925ed9be88919f4bebb0"}]},"branch":"refs/heads/master"},"d75ac5930ec6c71836ef37532b789eafd2b5a0cb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-30 16:09:09.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/28/689028/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/689028/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/689028/3"}}},"commit":{"parents":[{"commit":"68c3d9177f83501c0d44dcb36bae0c286fa66d43","subject":"Pin virtualenv to 20.7.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/68c3d9177f83501c0d44dcb36bae0c286fa66d43"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-10-16 19:53:36.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2021-09-30 16:09:09.000000000","tz":0},"subject":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile","message":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile\n\nChange-Id: I86051fc48af907be05cfdd4fbb86d35aae704c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d75ac5930ec6c71836ef37532b789eafd2b5a0cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d75ac5930ec6c71836ef37532b789eafd2b5a0cb"}]},"branch":"refs/heads/master","description":"Rebase"},"e511b07953361943fbf786d160d7e9abefc70129":{"kind":"REWORK","_number":4,"created":"2022-07-13 13:29:31.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/28/689028/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/689028/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/689028/4"}}},"commit":{"parents":[{"commit":"9a3d79d4a40184d260e9708531eb18b74eb6042f","subject":"Merge \"Reduce build targets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9a3d79d4a40184d260e9708531eb18b74eb6042f"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2022-07-13 13:27:41.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2022-07-13 13:29:17.000000000","tz":180},"subject":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile","message":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile\n\nChange-Id: I86051fc48af907be05cfdd4fbb86d35aae704c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e511b07953361943fbf786d160d7e9abefc70129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e511b07953361943fbf786d160d7e9abefc70129"}]},"branch":"refs/heads/master"},"72342c3e2d7e1e93688a2bbc236c5621b8f52313":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-25 22:52:09.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/28/689028/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/689028/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/689028/5"}}},"commit":{"parents":[{"commit":"2537db07aa3a3836cd215281e2fe2aa7923706b0","subject":"Refactor docker image task values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2537db07aa3a3836cd215281e2fe2aa7923706b0"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2022-07-13 13:27:41.000000000","tz":180},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-07-25 22:52:09.000000000","tz":0},"subject":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile","message":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile\n\nChange-Id: I86051fc48af907be05cfdd4fbb86d35aae704c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/72342c3e2d7e1e93688a2bbc236c5621b8f52313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/72342c3e2d7e1e93688a2bbc236c5621b8f52313"}]},"branch":"refs/heads/master","description":"Rebase"},"f725a1256a2413fdada29b4d9d88acae63970ae9":{"kind":"REWORK","_number":6,"created":"2022-12-06 16:33:15.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/28/689028/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/28/689028/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/28/689028/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/28/689028/6"}}},"commit":{"parents":[{"commit":"5806885d68c6e15251168183245976a26c0da81a","subject":"Merge \"Drop UWSGI from always being installed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5806885d68c6e15251168183245976a26c0da81a"}]}],"author":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2022-07-13 13:27:41.000000000","tz":180},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2022-12-06 16:18:46.000000000","tz":120},"subject":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile","message":"Install libldap-2.4-2 bindep library for the py3 ldap keystone profile\n\nChange-Id: I86051fc48af907be05cfdd4fbb86d35aae704c4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f725a1256a2413fdada29b4d9d88acae63970ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f725a1256a2413fdada29b4d9d88acae63970ae9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[]}
