)]}'
{"id":"openstack%2Floci~693754","triplet_id":"openstack%2Floci~master~I9562acc20b1f08051b2645febe993cbc7cbb1b19","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I9562acc20b1f08051b2645febe993cbc7cbb1b19","subject":"Install libpython[23].x after python installation","status":"MERGED","created":"2019-11-11 22:50:57.000000000","updated":"2020-01-27 18:47:54.000000000","submitted":"2020-01-27 18:47:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693754-1580150874208-728c2603","meta_rev_id":"6b6087535462ca4b7e474fb6a38ac4c21369742d","_number":693754,"virtual_id_number":693754,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 18:47:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":0,"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"value":0,"date":"2020-01-27 17:01:37.000000000","_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"recommended":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":1,"date":"2020-01-27 12:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":2,"date":"2020-01-27 12:15:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2020-01-27 12:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2020-01-27 12:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":1,"date":"2020-01-27 12:15:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":1,"date":"2020-01-27 12:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":1,"date":"2020-01-27 14:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":0,"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 23:24:02.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2019-11-12 16:44:01.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-11-12 16:44:11.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-11-15 19:27:48.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2019-11-27 16:20:27.000000000","updated_by":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"reviewer":{"_account_id":2243,"name":"John Griffith","email":"john.griffith8@gmail.com","username":"john-griffith"},"state":"REVIEWER"},{"updated":"2019-12-12 20:15:45.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2020-01-09 08:10:03.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"state":"REVIEWER"},{"updated":"2020-01-13 02:04:30.000000000","updated_by":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"reviewer":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"state":"REVIEWER"},{"updated":"2020-01-21 22:03:09.000000000","updated_by":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"reviewer":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"state":"REVIEWER"},{"updated":"2020-01-27 14:50:34.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-01-27 18:47:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68d9074d646d6d3bb11cecd91bab2573a1574b0e","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-11 22:50:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"807b6910133855c41408022dc2abc282c097c240","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-11 22:55:09.000000000","message":"Patch Set 1:\n\nalternative change https://review.opendev.org/#/c/693753/1","accounts_in_message":[],"_revision_number":1},{"id":"0ce31afd28b75a9fb5e7c5a02ecd8c7dce1db44c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 23:30:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/25b203ae315a4a03acbbc050dd294265 : SUCCESS in 39m 35s\n- loci-glance https://zuul.opendev.org/t/openstack/build/083a6a058dc94185adcabf5eff1a6d33 : SUCCESS in 31m 40s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c7cfee001aa24406b772dcb765ba973b : SUCCESS in 32m 02s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5fd1f9b873b34c858f3a1e43c00431dd : SUCCESS in 32m 21s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c2e31d9555294fcc90745ef43e46c743 : SUCCESS in 31m 36s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/713d89b134114632a5ef764a9a3c50d4 : SUCCESS in 31m 13s\n- loci-manila https://zuul.opendev.org/t/openstack/build/184f147545e34397ad4221e167f05f49 : SUCCESS in 30m 36s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6ef8391ca617441faaf1f84e18517828 : SUCCESS in 33m 02s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c0b48767cf044fa9ba801a85512cd3aa : SUCCESS in 33m 39s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3424e6f10a7b44e189243d7688aca884 : SUCCESS in 31m 43s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3b04101a88b041318a76bd0c895c5d03 : SUCCESS in 31m 02s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/68e206468dce4bc086ae48f654ac7eb1 : SUCCESS in 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0e518ac6d326c7554cead2fd0befaf1e1f5a38d1","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-12 00:43:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43617ddf3baa6f75b00626b253cd9124685c8fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 01:41:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/6721e85ad191434181f8f27408143005 : SUCCESS in 36m 21s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b7f21b0a07874286ba59b462c76a2d2f : SUCCESS in 52m 00s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3e586ea86f0549f395c0e79a100db2c5 : SUCCESS in 49m 18s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/0504f87b28f644af86c3497906241577 : SUCCESS in 52m 28s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/69e0aeaa00b0481b974ede51bb06a3ac : SUCCESS in 48m 25s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7aea8446730245498864a90ba3b69367 : SUCCESS in 50m 01s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7686c305dddc4863bdc53e4293602c8d : SUCCESS in 48m 08s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/40ab69f1fca9447b9fe885cb61655a7c : SUCCESS in 55m 44s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e7caea74a1a84020a1006d8988943b1f : SUCCESS in 51m 36s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/fd5ee512eaa744c2a449dda758b310c6 : SUCCESS in 52m 35s\n- loci-placement https://zuul.opendev.org/t/openstack/build/844120e950c147cfbcf5c7d08abe8f53 : SUCCESS in 45m 48s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6a06b02d533e4d4588089c55cf25a25e : SUCCESS in 21m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2b6141aa9cbb2ef7cc7938e0c5c7a18f8e091d8e","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-11-12 14:42:52.000000000","message":"Patch Set 2:\n\nAfter looking at the differing methods proposed here, I think this one is the better of the two, considering the options we have.","accounts_in_message":[],"_revision_number":2},{"id":"80bb0907acf65a503d4d0c8db02fab9b5fab804a","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-11-12 14:43:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c6f913c8f5cc4d28ba609ffefd2b49f152a5b54","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-12 16:38:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a656e3432d2a20c51eff4a393c94163633440f4f","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-12 16:39:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f1655ddb9e1e2c92cc3830e3ff535e425e081d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 16:48:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0a6bb3f0b84a4cb78d2ba931eed4a113 : FAILURE in 4m 05s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9aa85ac91a3743129518be889b57bb0a : FAILURE in 4m 32s\n- loci-heat https://zuul.opendev.org/t/openstack/build/32181f21bd7b4f6bb98a71d8512ea9d9 : FAILURE in 4m 07s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5b4f4dc0128a4de2b45e0de0a5f2c23f : FAILURE in 3m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/20fcd7fdfce24391be7e8a73b64d7164 : FAILURE in 4m 18s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1a9b6c78a85749098573908d8d3f0f2c : FAILURE in 4m 16s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7500baede33949d694effb2ddc4878ce : FAILURE in 3m 18s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/38ea7040e96f439b8cb11f4dcfc90542 : FAILURE in 4m 05s\n- loci-nova https://zuul.opendev.org/t/openstack/build/171ebca61a1740a39c439750be2e31ae : FAILURE in 3m 14s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4865b137631f44c98059177df6021e69 : FAILURE in 2m 54s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f22c4072a6b3490d9ebbfeafdf9b6586 : FAILURE in 4m 20s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c6fe21ef782648e49e651ea4a25819ff : FAILURE in 2m 39s","accounts_in_message":[],"_revision_number":3},{"id":"2027644a476ade7836e717be4fdc43e3ad6a0bcf","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-12 16:57:21.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"9ae27af8806c4571478e9032b8933b780fe58b63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 17:06:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/05c36f3329de40a7b114f2256944c3fd : FAILURE in 5m 59s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cc4ee5bc5e65481099a9f3dabcde34c7 : FAILURE in 4m 46s\n- loci-heat https://zuul.opendev.org/t/openstack/build/930910d4eb5e4bf388a36ec5caedbd1a : FAILURE in 4m 45s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/478c3c1eea7a41bfb0b3830099a47d0c : FAILURE in 6m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c80f09abc93740dcaf4e5d0c15169f00 : FAILURE in 3m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d944ffb6f9bf4c20ae7dce59ef2b2ea3 : FAILURE in 3m 35s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4dadb3bdefe8402fbe55879eca49c185 : FAILURE in 5m 28s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b7e634e901d2476f848bde83aa352743 : FAILURE in 3m 48s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f00c39bfbaac4b578e75dd8b469ad910 : FAILURE in 5m 12s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/bb56ebdd4f524140b407b2153fc6ff8c : FAILURE in 3m 02s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2ad4646784cd4d90b0e50b007d300c10 : FAILURE in 2m 39s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/343cf0a769eb46b1a8344148e2e4b7cc : FAILURE in 2m 30s","accounts_in_message":[],"_revision_number":4},{"id":"9571cbdde2f02ec0a204c45c169d8730e020cde1","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-12 17:16:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"db08a696845dc63a11fc0c9470dd9c06545154b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 17:26:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9d3a890b8885425f8216d0ad65fd297c : FAILURE in 4m 07s\n- loci-glance https://zuul.opendev.org/t/openstack/build/dc6ec08cc7674b73abe58eac3572239f : FAILURE in 4m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/bbcad159cc8647e5a322ab01257175d2 : FAILURE in 3m 01s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/b12427d15a504bbfa1a1c1ab57a35ad0 : FAILURE in 3m 22s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5f39020872ae411e99092da6a172949a : FAILURE in 2m 47s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2cdeea355a6f49c99300f88df87643d0 : FAILURE in 3m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/bcc5f528a51044afbb496898d9f9c095 : FAILURE in 3m 34s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ed65e93d1dcf4a9aa33b158da0078ffb : FAILURE in 3m 21s\n- loci-nova https://zuul.opendev.org/t/openstack/build/22a890bed6db4474807c60021d75de62 : FAILURE in 4m 31s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f7630de35eec4a228a032f42245fc9c4 : FAILURE in 3m 22s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3c3cb07351ec4a778d81677f43638823 : FAILURE in 2m 28s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8833287a620743c9912286290a28294c : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":5},{"id":"db6b8e4c2d6936691ff620f54be1d2d431fe7cb3","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-12 17:28:08.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"21af0ca0662dc5e49012657054c97a7556eb6c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 18:19:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5fd74d9159994771bfacea45c5cec7e8 : SUCCESS in 37m 20s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c98b8ff7934a4ce8be36434857753a35 : SUCCESS in 38m 00s\n- loci-heat https://zuul.opendev.org/t/openstack/build/22102a786bdf4372bda1101797b9375f : SUCCESS in 40m 05s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/139bc29b8ac44142954ba3b95000d597 : SUCCESS in 37m 12s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/7c7735297879437c9646edd48c25f5db : SUCCESS in 39m 27s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/74bb9c2272764f71b4bd7594f6f9c399 : SUCCESS in 39m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5a81fd5da09c43c6a6751613746375fb : SUCCESS in 35m 45s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7377ebd8f05941c98005dd2308244fd7 : SUCCESS in 45m 32s\n- loci-nova https://zuul.opendev.org/t/openstack/build/101d98d17f6e47aab07ad4bc2fc4ec2b : SUCCESS in 40m 33s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5f1cedddb430438b992026f7f35593ab : SUCCESS in 44m 12s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cf67a5fee16f4b648f0c68c4afee2576 : SUCCESS in 42m 34s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/80971980bd7547b18a21ed913ad7b1a8 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":6},{"id":"408525a9d090551248e8145ed827af773cc1a085","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2019-11-13 15:57:41.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"884941b927ddf34b10bfbb8f6361f48fe39b94ca","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-13 16:18:56.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"589bec42c3559eba3acbb58de81311f602c46fd9","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-13 16:19:41.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"151c96f180421e25c40c3b335e26ef6c5b9c03b4","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-13 16:20:06.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cc3f959ab47d233b0f00f6a3f894234ef0d27dd6","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-13 16:55:17.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"1159b99f6d0b1c637961a0acc952ea1fb2f23d0e","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-13 16:58:06.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"3a29d0bf919e48fcfaac8e362c255f653ee4dff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 18:27:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c8f418f4662e43e295253ad054e004f9 : SUCCESS in 38m 46s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6b78616486274f2ebf7805651075ba3e : SUCCESS in 37m 23s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a30291ab43ea4a5fbde8228752c2320f : SUCCESS in 37m 07s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a630c045c0274affb600534cb163a9de : SUCCESS in 39m 01s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a976b122b0dc42199a5b316c86fc712b : SUCCESS in 32m 56s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1299dd58d27f4626ad2bda0f5a275b85 : SUCCESS in 32m 35s\n- loci-manila https://zuul.opendev.org/t/openstack/build/863a9c18b7d04e89a6a19c5f1795d110 : SUCCESS in 31m 39s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/0c77492d7db84e5ca16d35f56e0bb18f : SUCCESS in 34m 19s\n- loci-nova https://zuul.opendev.org/t/openstack/build/134f9bea313c4d80981949d0f3800a67 : SUCCESS in 43m 03s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/45c2b47b28b84c2cbbaa8772611ed689 : SUCCESS in 37m 54s\n- loci-placement https://zuul.opendev.org/t/openstack/build/d2bee70481514ae7ac06fa948aebef81 : SUCCESS in 31m 02s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4c3c176da382407c8659a1ac1c535b88 : SUCCESS in 30m 58s","accounts_in_message":[],"_revision_number":10},{"id":"401884dc15e655baa9ade4faf0b8a87a5b08d655","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2019-11-15 19:27:48.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"8272075c9aded114a511375c27cdac3679d6f3e9","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-18 22:33:08.000000000","message":"Patch Set 10:\n\nplease review. Thank you!","accounts_in_message":[],"_revision_number":10},{"id":"2c6c43916478c065e924441629ec00d4367493d1","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-26 19:31:02.000000000","message":"Patch Set 10:\n\nplease review","accounts_in_message":[],"_revision_number":10},{"id":"55222dd3433a5e27d0c2633b7ebd28c5f61ce751","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-12-09 20:13:23.000000000","message":"Patch Set 10:\n\nplease review","accounts_in_message":[],"_revision_number":10},{"id":"46a5347328df35740d752af67c80a402af283644","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-12-10 06:05:22.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"decc02badb7a50341ab7b94735e021cebc441f8e","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-12-12 20:15:45.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"1e3dabbf0b3a247c85a2f20f352f8872db3685e2","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2020-01-09 08:10:03.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"a317f5651a3349afb44725a4c37c3db57f287d9e","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2020-01-10 14:38:22.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"95433f55c9b720d888692c336953aa587a15a451","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-10 15:50:26.000000000","message":"Patch Set 10:\n\nI think we\u0027ll need wf here as well https://review.opendev.org/#/c/693886","accounts_in_message":[],"_revision_number":10},{"id":"88004f38ec3d2f84b0d5ea169ab6e1fdd5d69706","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2020-01-13 02:04:30.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"586fe64835296792c9de53a8abbad8c8bf7247cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 23:04:58.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"18643c340dc97c358d43552a697855affb69b1ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 23:59:47.000000000","message":"Patch Set 10: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"a5d741e2b80e06f0b9d5f52e7601b29688ca7223","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 06:04:03.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"de8155d2bc79af77138bf43ea7dd6c03d5bdea84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 07:45:07.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ff45741e74f14f77936e8ebd8ba9d9a6 : SUCCESS in 40m 59s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4ee4f71276ac42569d0dc07326128f44 : SUCCESS in 51m 42s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0a0461df26ad437081d04d41937ce35d : SUCCESS in 34m 40s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/44dabde0a04a44c98b1a7bbf69200137 : TIMED_OUT in 1h 02m 15s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3ec46d95bd574005988cd5c3457d8572 : SUCCESS in 40m 46s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e2ad1e97851a42dab2060c69bcfc8773 : SUCCESS in 50m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c914afa42c184a839348ce303cc27ac6 : SUCCESS in 52m 15s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7f42cd3002e0442190182c707b70ca55 : SUCCESS in 51m 11s\n- loci-nova https://zuul.opendev.org/t/openstack/build/961557a386964fa9b9ac390abfd1dd44 : SUCCESS in 41m 05s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/84db9aa0702f45d7a00ab75e606d25fc : TIMED_OUT in 1h 01m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/81ce9f7842e64dca83a05e2a987ea6ba : SUCCESS in 38m 43s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/044260521ef24ef698c17df65b41fc48 : TIMED_OUT in 41m 07s","accounts_in_message":[],"_revision_number":11},{"id":"da7b2d542b343a5b8073e767fb5700908e10dd2d","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 14:03:47.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"ca756222dfa7a87b7359dc4ffe88edff92e658aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 14:51:49.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5dd5a18dcfbd416eb2cd2edd12cc4fa7 : SUCCESS in 42m 24s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d638b2dc64074d31834f8eb92b6e6da3 : SUCCESS in 40m 34s\n- loci-heat https://zuul.opendev.org/t/openstack/build/dc9d1efb55cc49129ab13b9ad66d80c1 : SUCCESS in 39m 05s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/8cee591bc2624f618dac187265345113 : SUCCESS in 40m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2ab0911b8b7d425f82d1c2e8ab31dc7a : SUCCESS in 36m 40s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/daec6f86ad9f4e059f325e9e453cb520 : SUCCESS in 36m 44s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e78710d9266548388279627092557382 : SUCCESS in 43m 01s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a6d8308a326a47faa4e2323d2955b236 : SUCCESS in 41m 23s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0ed54c84760044b59221bd5f9bfd9c5c : SUCCESS in 43m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b5705c119dac421d9a3922b8251ea7af : SUCCESS in 42m 06s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f0f06b5e8ac74569ab3db98ee09d8bcb : SUCCESS in 37m 13s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/884e6f21697041ea9852f1967f72103e : SUCCESS in 31m 36s","accounts_in_message":[],"_revision_number":11},{"id":"926a0c7bdcdeae70de28ff5fab0df3077e7e48e5","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 11:43:49.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"cd5af5d4439d6a80c2a75db85db210214a21a06c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 12:41:00.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/6249683075d8426fbb9a3fd61e8ed052 : SUCCESS in 44m 37s\n- loci-glance https://zuul.opendev.org/t/openstack/build/843fc950e73a4b3489ffb541d02ae698 : SUCCESS in 44m 44s\n- loci-heat https://zuul.opendev.org/t/openstack/build/40dd65e5710a4cffbfc5de98a71f1602 : SUCCESS in 49m 43s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6498ac11e1b24ce3841387681b1fd7b8 : SUCCESS in 46m 46s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0bf01aca3f4845b1bb8b63448edb1515 : SUCCESS in 42m 26s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d4840aa422ff41c28b1a73b06744a175 : SUCCESS in 49m 19s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3aeaf92d17d746dda12e8a89cd74d994 : SUCCESS in 48m 41s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/157ca8501240406c931dfd68197eeee2 : SUCCESS in 52m 01s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7b9f09ebfb5c470eb3c0f2a7cbb75ee1 : SUCCESS in 47m 51s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/50c6eb35f0f8479d8bac7f590727927c : SUCCESS in 42m 54s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f9917f6a62104ffd8f501da9766c2a54 : FAILURE in 36m 18s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/19c39a8cd9f345acbb5d86db0f6984b7 : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":12},{"id":"d98f153b83b99a2c85dfd41881b66c9c965c5653","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 15:59:07.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"7a690dee46d86b4d668ea12ae5667550ff4ad08d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 16:53:25.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4d08882a6ab24a7e8d86d9e74d81008a : FAILURE in 41m 03s\n- loci-glance https://zuul.opendev.org/t/openstack/build/62e82054b050484e95e34a317e2bcdc7 : FAILURE in 23m 29s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b927d59582574d199fc8d5890cf57553 : FAILURE in 39m 29s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d475fcc2040b4d3b981183de632991e6 : FAILURE in 40m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d109778830fb4aed8be25547b23db765 : FAILURE in 44m 53s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6e56ca5789ff48048167e7dbb8ba1fa7 : FAILURE in 37m 32s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a18553f32e4e4dcab38813465a6af3f6 : FAILURE in 35m 17s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5b20c5f381c04c0d96355eec32f7a67c : FAILURE in 49m 26s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b0e6215532a9454c9555dd9c5e9400f7 : FAILURE in 47m 04s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ae0a738cdb7c4d74a306e1583a840f98 : FAILURE in 45m 59s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b81131d851b14ffaa672b41f8c044ba0 : FAILURE in 28m 40s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b9484b946fd54ec89437c02133852b05 : SUCCESS in 37m 11s","accounts_in_message":[],"_revision_number":13},{"id":"ae42c3f4b4e6fd903609b0d37aa74629d1e10605","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 14:28:50.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"9d19fbbf229aaf829c71516d0c29495506116ca4","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 14:29:14.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"e4f9b3790441fc5feb815693a93864126c1a0fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 15:51:35.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/aceca984a4ca4e7fa7a4280e7d1233a6 : SUCCESS in 1h 11m 43s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f3fc762f8235491d9587eb441b6c121c : SUCCESS in 57m 33s\n- loci-heat https://zuul.opendev.org/t/openstack/build/105b32c384214ab295eecf43a8d958e9 : SUCCESS in 1h 06m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/27bf612e02124022bdc73dc16126ff69 : FAILURE in 1h 14m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9d7f2bd459d943f0bc6222171cd215d8 : SUCCESS in 1h 05m 21s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b5fc5f3ed15242a99810e904a47e97cf : SUCCESS in 52m 22s\n- loci-manila https://zuul.opendev.org/t/openstack/build/29a246e8a6d94932b0368b65ce7379ca : SUCCESS in 52m 08s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c954d920dfee4875ad1cb4c036f0793d : FAILURE in 1h 15m 08s\n- loci-nova https://zuul.opendev.org/t/openstack/build/441afbf1d47f4ae1afcd788eb25ee4a7 : SUCCESS in 1h 15m 27s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/d500d77b7f21491dac89299850f7831a : SUCCESS in 1h 08m 48s\n- loci-placement https://zuul.opendev.org/t/openstack/build/938e0a235f524e3e8b1cc78d02142933 : SUCCESS in 53m 28s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/103ea36fdc5b42518860a1b0e5dc8950 : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":15},{"id":"38cd343ab08c36fce6170880746394ebc6943aa5","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 19:29:45.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"285f25af86c8b33dc20f25c97ff663005794b92c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 20:47:31.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/13f872d8d6e240bea3695159664fdc76 : SUCCESS in 48m 10s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f7beaeef67624ea7a6685a5105b7f890 : SUCCESS in 45m 53s\n- loci-heat https://zuul.opendev.org/t/openstack/build/43f12cf608b647258c7cda2f4e35c0e1 : SUCCESS in 47m 22s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/81c291b736694fcfa7d136949db6679a : SUCCESS in 1h 04m 39s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1652b7c8f481424b907002cd029190ec : SUCCESS in 49m 58s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/53250a9ec81b4f339085a6a66b342aa6 : SUCCESS in 54m 08s\n- loci-manila https://zuul.opendev.org/t/openstack/build/088fcb61dfe14a4eaa593a97e92f3ada : SUCCESS in 49m 54s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2f9051b8752b4660ac6443bfefb888e9 : SUCCESS in 54m 56s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b77b3b640741421483849133b6a11e5e : SUCCESS in 1h 09m 44s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4d29389d677e462084ef965ac381ed5e : SUCCESS in 48m 19s\n- loci-placement https://zuul.opendev.org/t/openstack/build/7f6c4a6d95384947949dc53c9a2447d9 : SUCCESS in 49m 03s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/941191853cfe4f33998d3bdafe56fa79 : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":16},{"id":"770c039d02590efbe36bdd311d71eaaa5f45431a","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2020-01-21 22:03:09.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"3ae43eaa378897c5a4bb0651557f79f8d8b6caf0","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 12:15:14.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"b596bd245d0862b4f44961c13c3734e87d1dab10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 13:05:59.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/cbbdb391dbd94698ad157775bb3d5e01 : SUCCESS in 35m 36s\n- loci-glance https://zuul.opendev.org/t/openstack/build/feed1953600247588fe03d2564018137 : SUCCESS in 32m 57s\n- loci-heat https://zuul.opendev.org/t/openstack/build/72628d4409c6443484a2dbce7b47f6d4 : SUCCESS in 34m 12s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/cac83f5401364f36b5b626b3cd72786d : SUCCESS in 38m 00s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ded6f1ae2b1448f98c6f9e983f1a06b2 : SUCCESS in 34m 52s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c40262e9e5b34a7a9a69cf0292f73511 : SUCCESS in 35m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a42a69f857214642af293a8c9438d2d9 : SUCCESS in 33m 03s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/92de017cd9b14af68740b885990503eb : SUCCESS in 41m 59s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ab21902d48c44225ba62619cbcfaebef : SUCCESS in 45m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/bdc1e1f4dddd4016b2bef24f8587eb54 : SUCCESS in 36m 54s\n- loci-placement https://zuul.opendev.org/t/openstack/build/fe489ae2444341b2b3b62b8d1447b358 : SUCCESS in 35m 24s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/efd8223ae9154618906e0bba2422faac : SUCCESS in 25m 34s","accounts_in_message":[],"_revision_number":17},{"id":"282f690659ef66c71cf4b6619dd5f54ba66777e2","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 14:47:41.000000000","message":"Patch Set 17:\n\nWF please. Thank you!","accounts_in_message":[],"_revision_number":17},{"id":"63a17368a62660c14a927fc44d0f3d709dd0dded","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2020-01-27 14:50:34.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"014d9df53d4ab115ee76a07ab7b750353d70a22e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:51:32.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"258927d313f483eebb3818fa1314104e78cf4b5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 15:57:36.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/89aa858fc5e749d0a7e084c0720b484e : SUCCESS in 44m 39s\n- loci-glance https://zuul.opendev.org/t/openstack/build/dcbe1dd7c7644b8bbb6693f5c1cfb800 : SUCCESS in 41m 57s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2f1363b2bb67457eb39b58fcc5b1c2e9 : SUCCESS in 42m 25s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/d2febd78a1da4c31b0356c1199eae8a0 : SUCCESS in 48m 06s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e85d84cec4434cb3aed4cf385bc326d1 : SUCCESS in 49m 54s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/637b05f0800e4928b99cb3448f96a19d : SUCCESS in 39m 11s\n- loci-manila https://zuul.opendev.org/t/openstack/build/97d9f42ef4c74bff853561896c8e5947 : SUCCESS in 35m 44s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a414afd8f8924586a30f2e23d6756ea2 : TIMED_OUT in 1h 01m 01s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fd28b21b5f7a4dca86d233d41febf232 : TIMED_OUT in 1h 01m 11s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8d709c018fac4e028955d0a029d0a97b : SUCCESS in 44m 03s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/cc9566d54e294169941a23103616a7f5 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":17},{"id":"3dc5b507022583fe2ffe845eba0f31805818d269","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 15:57:46.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"c58cab187d9387c502b902ecd9f0b6badae75a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 17:01:24.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e960e21802eb44f8ba07a6a3dd69bf3e : SUCCESS in 47m 09s\n- loci-glance https://zuul.opendev.org/t/openstack/build/239e9091887545fa83ebae981d9e108c : TIMED_OUT in 1h 01m 52s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d94f08c5ad554e5dbe11ed8de6f72cd3 : FAILURE in 36m 07s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/35982e91b7724525a58002270f5c26f7 : TIMED_OUT in 1h 00m 55s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c34de023eda94671b406191c6f9822a9 : SUCCESS in 44m 03s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/24c14c7e4a8d4b6596b3b1c2ba3e02fa : TIMED_OUT in 1h 01m 20s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e9077f88f5ac489c9454313299cbd006 : SUCCESS in 39m 25s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5f306c87744b42c0bea8815aeba946fa : SUCCESS in 45m 52s\n- loci-nova https://zuul.opendev.org/t/openstack/build/afb8773898354f458e01da61ff0e90fd : SUCCESS in 50m 04s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e9a9efdec07c47ec9590f789e5d2db9c : SUCCESS in 48m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a3d3015299cb41f98dc0e654e74b12a7 : SUCCESS in 36m 29s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0d5b126f890e4206a6ba516d17aa111f : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":17},{"id":"2cf5141fbd7a1008c3e5fb3127cb2d324bce4007","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 17:01:37.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"f2afecacd50b9147d1cfd6d5528856056cf7f6c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 17:52:12.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f14e47a6d05643c8b77b2ef87cbc69b5 : SUCCESS in 49m 22s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9811cd02fa2c4f39b4d8478355e8db64 : SUCCESS in 48m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2da138172d974e17bff13090cf98c29e : SUCCESS in 43m 52s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/0fd772be9e6149079361f4ee22dbca07 : SUCCESS in 45m 11s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c23a5459adf04ff6a9f87cc260f02b3b : SUCCESS in 47m 02s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3c736ac68fad484488de3b47e20c47d7 : SUCCESS in 37m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/bdf225c460874ea697263a8faea7e3e0 : SUCCESS in 38m 13s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/edc573356bdf4fc2970bdb98076de73b : SUCCESS in 40m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/23b4f8902fad49019f576a31056eef31 : SUCCESS in 43m 11s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7f6dadd6c0024b0fbedae696a0d3b425 : SUCCESS in 48m 07s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cdea3446959e4dc191006bd50f7b54bd : SUCCESS in 40m 22s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6cc21e99b5a047b1a5c92118b09b09ef : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":17},{"id":"81db32457992129e6e95ba04b294b29ae86ecdc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 17:52:23.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"9c30bd23a924ca716c68602f2c8a2a05d03ff714","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 18:47:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"6b6087535462ca4b7e474fb6a38ac4c21369742d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 18:47:54.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/70bb663d6a2e4814bc7a8fc6553aa1e9 : SUCCESS in 42m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/80b2de9cd52f4326b1dfa976427351bc : SUCCESS in 46m 42s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f35de5c4f1f3463db96123c3f864ae08 : SUCCESS in 43m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9e8d2274c26649e8aecd46fefbf2e03b : SUCCESS in 46m 44s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5ff7e62e167c472cb6f3607d8aa46877 : SUCCESS in 45m 05s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/96e865db9ed0413fa942fc625cafc810 : SUCCESS in 46m 12s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6b770fc8ca734006be5097d472e598ac : SUCCESS in 37m 11s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d1fe546df4a84cc98d3cedad92014fce : SUCCESS in 38m 27s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e21e78925b8f40ccac32fcf821e41359 : SUCCESS in 53m 52s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/907008be8fce4594a51da164ff7aeb97 : SUCCESS in 47m 33s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ca72ff04df464a60a9bf695cf1d62b17 : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"497691cd2c1ceb14ad261353772966c55d6ef10b","revisions":{"bbab668d2d985ee2ea846b800ba2c87bb8ee0ef0":{"kind":"REWORK","_number":1,"created":"2019-11-11 22:50:57.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/1"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"subject":"Install libpython3.x after python3 installation","message":"Install libpython3.x after python3 installation\n\nMove libpython3.x installation logic from bindep to scripts\nto allow dynamic identification of python3 version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bbab668d2d985ee2ea846b800ba2c87bb8ee0ef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bbab668d2d985ee2ea846b800ba2c87bb8ee0ef0"}]},"branch":"refs/heads/master"},"8c66ce186c97c0a004a62bc372379a4bc28cc3e8":{"kind":"REWORK","_number":2,"created":"2019-11-12 00:43:06.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/2"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 00:43:02.000000000","tz":-360},"subject":"Install libpython3.x after python3 installation","message":"Install libpython3.x after python3 installation\n\nMove libpython3.x installation logic from bindep to scripts\nto allow dynamic identification of python3 version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8c66ce186c97c0a004a62bc372379a4bc28cc3e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8c66ce186c97c0a004a62bc372379a4bc28cc3e8"}]},"branch":"refs/heads/master"},"5e0b55c35b3a51ca21f2e5122537091890808113":{"kind":"REWORK","_number":3,"created":"2019-11-12 16:38:51.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/3"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 16:38:41.000000000","tz":-360},"subject":"Install libpython3.x after python3 installation","message":"Install libpython3.x after python3 installation\n\nMove libpython3.x installation logic from bindep to scripts\nto allow dynamic identification of python3 version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e0b55c35b3a51ca21f2e5122537091890808113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e0b55c35b3a51ca21f2e5122537091890808113"}]},"branch":"refs/heads/master"},"7e817270a79bffb6dc727c5df2b3713304c514c9":{"kind":"REWORK","_number":4,"created":"2019-11-12 16:57:21.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/4"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 16:50:56.000000000","tz":0},"subject":"Install libpython3.x after python3 installation","message":"Install libpython3.x after python3 installation\n\nMove libpython3.x installation logic from bindep to scripts\nto allow dynamic identification of python3 version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7e817270a79bffb6dc727c5df2b3713304c514c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7e817270a79bffb6dc727c5df2b3713304c514c9"}]},"branch":"refs/heads/master"},"c36bf3e60ad5e2e4089b5227c3dd4e04fb66849f":{"kind":"REWORK","_number":5,"created":"2019-11-12 17:16:26.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/5"}}},"commit":{"parents":[{"commit":"da6e3ca04d6b8e8596427094db69b13c165114f7","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/da6e3ca04d6b8e8596427094db69b13c165114f7"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:16:16.000000000","tz":-360},"subject":"Install libpython3.x after python3 installation","message":"Install libpython3.x after python3 installation\n\nMove libpython3.x installation logic from bindep to scripts\nto allow dynamic identification of python3 version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c36bf3e60ad5e2e4089b5227c3dd4e04fb66849f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c36bf3e60ad5e2e4089b5227c3dd4e04fb66849f"}]},"branch":"refs/heads/master"},"98772e63f574270a30779d5048b06882dd9b0d30":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-11-12 17:28:08.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/6"}}},"commit":{"parents":[{"commit":"76f542f796cbd5708583bd86278a45c1046ff65f","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76f542f796cbd5708583bd86278a45c1046ff65f"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:28:08.000000000","tz":0},"subject":"Install libpython3.x after python3 installation","message":"Install libpython3.x after python3 installation\n\nMove libpython3.x installation logic from bindep to scripts\nto allow dynamic identification of python3 version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/98772e63f574270a30779d5048b06882dd9b0d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/98772e63f574270a30779d5048b06882dd9b0d30"}]},"branch":"refs/heads/master"},"a97d97e437299ab6e6db8c145f469201a7c91232":{"kind":"REWORK","_number":7,"created":"2019-11-13 16:18:56.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/7"}}},"commit":{"parents":[{"commit":"76f542f796cbd5708583bd86278a45c1046ff65f","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76f542f796cbd5708583bd86278a45c1046ff65f"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-13 16:18:48.000000000","tz":0},"subject":"Install libpython3.x after python3 installation","message":"Install libpython3.x after python3 installation\n\nMove libpython3.x installation logic from bindep to scripts\nto allow dynamic identification of python3 version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a97d97e437299ab6e6db8c145f469201a7c91232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a97d97e437299ab6e6db8c145f469201a7c91232"}]},"branch":"refs/heads/master"},"67e20dabfe9f78b4edd72400e8b7b9f7d2920cc5":{"kind":"REWORK","_number":8,"created":"2019-11-13 16:19:41.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/8"}}},"commit":{"parents":[{"commit":"76f542f796cbd5708583bd86278a45c1046ff65f","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76f542f796cbd5708583bd86278a45c1046ff65f"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-13 16:19:40.000000000","tz":0},"subject":"Install libpython3.x after python3 installation","message":"Install libpython3.x after python3 installation\n\nMove libpython3.x installation logic from bindep to scripts\nto allow dynamic identification of python3 version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67e20dabfe9f78b4edd72400e8b7b9f7d2920cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67e20dabfe9f78b4edd72400e8b7b9f7d2920cc5"}]},"branch":"refs/heads/master"},"d9a9045291b49ae435772d74723b32b8b12ece5e":{"kind":"REWORK","_number":9,"created":"2019-11-13 16:55:17.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/9"}}},"commit":{"parents":[{"commit":"76f542f796cbd5708583bd86278a45c1046ff65f","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76f542f796cbd5708583bd86278a45c1046ff65f"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-13 16:55:15.000000000","tz":0},"subject":"Install libpython[23].x after python installation","message":"Install libpython[23].x after python installation\n\nMove libpython[23].x installation logic from bindep to scripts\nto allow dynamic identification of python version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d9a9045291b49ae435772d74723b32b8b12ece5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d9a9045291b49ae435772d74723b32b8b12ece5e"}]},"branch":"refs/heads/master"},"be534434310e8751711d6c4b4265663a26611376":{"kind":"REWORK","_number":10,"created":"2019-11-13 16:58:06.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/10"}}},"commit":{"parents":[{"commit":"76f542f796cbd5708583bd86278a45c1046ff65f","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76f542f796cbd5708583bd86278a45c1046ff65f"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-13 16:58:04.000000000","tz":0},"subject":"Install libpython[23].x after python installation","message":"Install libpython[23].x after python installation\n\nMove libpython[23].x installation logic from bindep to scripts\nto allow dynamic identification of python version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/be534434310e8751711d6c4b4265663a26611376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/be534434310e8751711d6c4b4265663a26611376"}]},"branch":"refs/heads/master"},"22a7817d232d12238947b4e88a8c60cf8d2ae48f":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-01-17 06:04:03.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/11"}}},"commit":{"parents":[{"commit":"57d210e45aa5087fc0111b06b7d03baffb6341e8","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/57d210e45aa5087fc0111b06b7d03baffb6341e8"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 06:04:03.000000000","tz":0},"subject":"Install libpython[23].x after python installation","message":"Install libpython[23].x after python installation\n\nMove libpython[23].x installation logic from bindep to scripts\nto allow dynamic identification of python version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/22a7817d232d12238947b4e88a8c60cf8d2ae48f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/22a7817d232d12238947b4e88a8c60cf8d2ae48f"}]},"branch":"refs/heads/master"},"4a948abf4457e102293f64a56472165de6f7689b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-01-18 11:43:49.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/12"}}},"commit":{"parents":[{"commit":"2c3c3d4a98075e0fdea9f4700d0927ee1a968f45","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2c3c3d4a98075e0fdea9f4700d0927ee1a968f45"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 11:43:49.000000000","tz":0},"subject":"Install libpython[23].x after python installation","message":"Install libpython[23].x after python installation\n\nMove libpython[23].x installation logic from bindep to scripts\nto allow dynamic identification of python version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4a948abf4457e102293f64a56472165de6f7689b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4a948abf4457e102293f64a56472165de6f7689b"}]},"branch":"refs/heads/master"},"3987c421a2914cbc7e6dc20af57ce2679ad10026":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-01-20 15:59:07.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/13"}}},"commit":{"parents":[{"commit":"26f9eb530c92f99c4856cf6fc35328fc3d3f082d","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/26f9eb530c92f99c4856cf6fc35328fc3d3f082d"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 15:59:07.000000000","tz":0},"subject":"Install libpython[23].x after python installation","message":"Install libpython[23].x after python installation\n\nMove libpython[23].x installation logic from bindep to scripts\nto allow dynamic identification of python version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3987c421a2914cbc7e6dc20af57ce2679ad10026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3987c421a2914cbc7e6dc20af57ce2679ad10026"}]},"branch":"refs/heads/master"},"17f25699f7cf3f5e6ddc568c255017156b46c880":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2020-01-21 14:28:50.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/14"}}},"commit":{"parents":[{"commit":"82fbed8c851a89952a8cd3b5293d73b755adc474","subject":"Enable Debian gate build and fix mirrors related issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/82fbed8c851a89952a8cd3b5293d73b755adc474"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 14:28:50.000000000","tz":0},"subject":"Install libpython[23].x after python installation","message":"Install libpython[23].x after python installation\n\nMove libpython[23].x installation logic from bindep to scripts\nto allow dynamic identification of python version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/17f25699f7cf3f5e6ddc568c255017156b46c880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/17f25699f7cf3f5e6ddc568c255017156b46c880"}]},"branch":"refs/heads/master"},"272e9eb58166518301e2b463b0a15fe74834b4fa":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-01-21 14:29:14.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/15"}}},"commit":{"parents":[{"commit":"6447387257934c3fb700d4b7e86fc12a23d94fd6","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6447387257934c3fb700d4b7e86fc12a23d94fd6"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 14:29:14.000000000","tz":0},"subject":"Install libpython[23].x after python installation","message":"Install libpython[23].x after python installation\n\nMove libpython[23].x installation logic from bindep to scripts\nto allow dynamic identification of python version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/272e9eb58166518301e2b463b0a15fe74834b4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/272e9eb58166518301e2b463b0a15fe74834b4fa"}]},"branch":"refs/heads/master"},"69a372d314cb3b5e2bfabe29a57f8e161041fac0":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-01-21 19:29:45.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/16"}}},"commit":{"parents":[{"commit":"5245cfc64fb347ac4b8f129c0ed3788069eefcb2","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5245cfc64fb347ac4b8f129c0ed3788069eefcb2"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 19:29:45.000000000","tz":0},"subject":"Install libpython[23].x after python installation","message":"Install libpython[23].x after python installation\n\nMove libpython[23].x installation logic from bindep to scripts\nto allow dynamic identification of python version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/69a372d314cb3b5e2bfabe29a57f8e161041fac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/69a372d314cb3b5e2bfabe29a57f8e161041fac0"}]},"branch":"refs/heads/master"},"497691cd2c1ceb14ad261353772966c55d6ef10b":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2020-01-27 12:15:14.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/54/693754/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/54/693754/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/54/693754/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/54/693754/17"}}},"commit":{"parents":[{"commit":"8ed58b10f6bed12b2a74296e1571b457b577594a","subject":"Allow comments in dep files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ed58b10f6bed12b2a74296e1571b457b577594a"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-11 22:31:50.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-27 12:15:14.000000000","tz":0},"subject":"Install libpython[23].x after python installation","message":"Install libpython[23].x after python installation\n\nMove libpython[23].x installation logic from bindep to scripts\nto allow dynamic identification of python version.\nCurrently for bionic images shared libraries for default python3.6 is\nnot installed at all, causing, at least, uwsgi to fail.\n\nChange-Id: I9562acc20b1f08051b2645febe993cbc7cbb1b19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/497691cd2c1ceb14ad261353772966c55d6ef10b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/497691cd2c1ceb14ad261353772966c55d6ef10b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
