)]}'
{"id":"openstack%2Floci~693886","triplet_id":"openstack%2Floci~master~I2d4c65268cec70c1d03ed1556c5ab5dbab94c398","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I2d4c65268cec70c1d03ed1556c5ab5dbab94c398","subject":"Allow comments in dep files","status":"MERGED","created":"2019-11-12 17:16:26.000000000","updated":"2020-01-27 15:53:48.000000000","submitted":"2020-01-27 15:53:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693886-1580140428629-5f1f21a0","meta_rev_id":"d7e3aa8165277141625f41841c4ba1efd28ee44b","_number":693886,"virtual_id_number":693886,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 15:53:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":17886,"name":"Andrew Leasck","email":"al7689@att.com","username":"leascka"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":0,"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"value":0,"date":"2020-01-27 14:47:25.000000000","_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"recommended":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"all":[{"value":1,"date":"2020-01-27 12:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":2,"date":"2020-01-27 12:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-27 12:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":1,"date":"2020-01-27 12:15:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":1,"date":"2020-01-27 12:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17886,"name":"Andrew Leasck","email":"al7689@att.com","username":"leascka"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":1,"date":"2020-01-27 12:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":1,"date":"2020-01-27 14:51:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"value":0,"_account_id":17886,"name":"Andrew Leasck","email":"al7689@att.com","username":"leascka"},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":0,"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":17886,"name":"Andrew Leasck","email":"al7689@att.com","username":"leascka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 18:37:08.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2019-11-12 18:37:16.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2019-11-15 19:28:17.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2019-12-12 20:15:37.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2020-01-09 08:10:53.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"state":"REVIEWER"},{"updated":"2020-01-16 22:53:12.000000000","updated_by":{"_account_id":17886,"name":"Andrew Leasck","email":"al7689@att.com","username":"leascka"},"reviewer":{"_account_id":17886,"name":"Andrew Leasck","email":"al7689@att.com","username":"leascka"},"state":"REVIEWER"},{"updated":"2020-01-16 22:53:32.000000000","updated_by":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"reviewer":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"state":"REVIEWER"},{"updated":"2020-01-27 14:51:19.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-01-27 15:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01a0ccf5faabe7469d9e3c2f5f15ecdf575845a1","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-12 17:16:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b43fd9a013218a94ffa4125869b5e23580973cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 17:26:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/17b8b4dec38b4be6a5a64b438cec9060 : FAILURE in 4m 33s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2447f3d2ce394f4cbd51c7a936961cf9 : FAILURE in 2m 36s\n- loci-heat https://zuul.opendev.org/t/openstack/build/98afe6fcecc64b53abe4bfae71191fd7 : FAILURE in 2m 38s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a81acda66b4e42d89e29eca8524c7991 : FAILURE in 3m 31s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b11d1fddbcc14f9e96722ed56dbc857b : FAILURE in 3m 09s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4765a16034ee4a7e9da528081216ab1d : FAILURE in 2m 37s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c7ac3d35055d48199f2f804f444da0e5 : FAILURE in 3m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2239eaed459244619575effc574a9b2b : FAILURE in 2m 59s\n- loci-nova https://zuul.opendev.org/t/openstack/build/dcf637bef0784c948a90f5a7febeb054 : FAILURE in 4m 12s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b9054b4992bb4ca8b0e0da5f745526f1 : FAILURE in 3m 10s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2dbba2028d044fba85e6f3d77af01bcb : FAILURE in 3m 01s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/513f74a8cc574422beded7e06d8b373c : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"00ba1de82cda97e65ca956e382a9cfd602f380ac","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-12 17:28:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6c377de6c27264993d1d63d5eab44658c4b061dc","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-12 17:42:45.000000000","message":"Patch Set 2:\n\nnegative scenario tested here https://review.opendev.org/#/c/693891","accounts_in_message":[],"_revision_number":2},{"id":"97efb237ba4d99542ef426fb67465c39014b6156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 18:14:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/553704c528eb43f085fc5cc7d38fc5cd : SUCCESS in 35m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2341b567bf4449ea9bc3b0c04fbb4ee7 : SUCCESS in 37m 25s\n- loci-heat https://zuul.opendev.org/t/openstack/build/4b5a97c7441b4594bde6b6abb09486fa : SUCCESS in 35m 54s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/988ed100d6c94a5e9e1f84eb9a7bcaa2 : SUCCESS in 37m 35s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ba2900e255f4425189522b2ff55c181a : SUCCESS in 34m 19s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/bbf915595c8e4f2ba1aea40b158a77d8 : SUCCESS in 39m 48s\n- loci-manila https://zuul.opendev.org/t/openstack/build/39ee92f46b9d4917a53ba1dbd91c1ad1 : SUCCESS in 30m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e8738b127f824617b5ce18d44bf83165 : SUCCESS in 38m 26s\n- loci-nova https://zuul.opendev.org/t/openstack/build/86550cd2541345908739137f03deee44 : SUCCESS in 34m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/19e848d94be94f5c84222df4515c71e4 : SUCCESS in 36m 18s\n- loci-placement https://zuul.opendev.org/t/openstack/build/eaa4f6a64a514acb9e13cf9e650a1344 : SUCCESS in 35m 27s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/27e91b790f524806821bb13c2df8b3d9 : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":2},{"id":"56fecb3888d2402e02abe38eab11c746bbb8fa64","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-11-12 18:30:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"db967886a35975fc24b49fd3659fb5737b49fa2a","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2019-11-15 19:28:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1ec9a58db0861b55a8d2bac8c4cf590baf655d1b","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-11-18 22:32:56.000000000","message":"Patch Set 2:\n\nplease review. Thank you!","accounts_in_message":[],"_revision_number":2},{"id":"7e2c19a806b6c61e9685960ea6dc1f8a5c8558e5","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2019-12-09 20:13:15.000000000","message":"Patch Set 2:\n\nplease review","accounts_in_message":[],"_revision_number":2},{"id":"f92444864340846d4055c8355dad032907466b6a","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-12-12 20:15:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"131b7329a4ecb9a6eacebc5576b56d70781bfe22","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2020-01-09 08:10:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c8632aea0414e10536edd2d137e79546b241a60c","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2020-01-13 08:42:52.000000000","message":"Patch Set 2:\n\nThe other patch needs this patch to have workflow as well: https://review.opendev.org/#/c/693754/","accounts_in_message":[],"_revision_number":2},{"id":"ff81d321373010be214d02af2eeb35110e5fc5d3","author":{"_account_id":17886,"name":"Andrew Leasck","email":"al7689@att.com","username":"leascka"},"date":"2020-01-16 22:53:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cfc90ca93fc5937dca7151329e795052004dc626","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2020-01-16 22:53:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"236f4fcb50dec16793f5c9c4f8ef50d07f06a677","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2020-01-16 23:04:46.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"602ab41d95f54f187cd73f97d26bfc7bef64a3f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 23:04:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0bd4e86c2187889f098706b4721e6103a2ef4514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 23:59:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c9fce2ca749649ea987b175c2f7c460a : FAILURE in 52m 11s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bd867cb513474d9089faa4aef28af0c5 : FAILURE in 49m 51s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e014b65aa0d54f81a04f820583f27048 : FAILURE in 31m 04s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/cf4275778ad74b13b39cae76a9f0ed71 : FAILURE in 45m 15s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ec177acba53c47869061293ceda6c98d : FAILURE in 36m 54s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b0b2a869095c49169b44ac3e674a45f9 : FAILURE in 30m 41s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a94d1a1874c14afba618a06304589c40 : FAILURE in 51m 02s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cbe1a36cbbfb4a9dbe7b9e6b32b26100 : FAILURE in 34m 28s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ac89669092324c47a1dad1d79e53fddd : FAILURE in 30m 29s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/89233239d2c4436fbafab1bae3cbbbcc : FAILURE in 51m 50s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e0d44bf589664a2ab7744f986e80fe03 : FAILURE in 30m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b7ea555ae2e96f7ebb055535385bd11a7b7d40d0","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 00:06:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8f14f1c8f138df71d41d240506a52d5b128e1004","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 00:15:37.000000000","message":"Patch Set 2:\n\nrelated to building wheel for newly updated neutron-lib\u003d\u003d2.0.0","accounts_in_message":[],"_revision_number":2},{"id":"369c1377d72746654d0ee612ac61cb307b2e2453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 00:40:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/747eb7f11ff04290ace20bb252ba2a07 : FAILURE in 25m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/dc89a404855d4d32b4cd793e02f26866 : FAILURE in 24m 08s\n- loci-heat https://zuul.opendev.org/t/openstack/build/afead0a1d0ff4790bdc4ce3eeb2a98c8 : FAILURE in 25m 00s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ce3077eb15dc45a8adf9e091f70db63d : FAILURE in 25m 03s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/914cedb20d954eed924e2b88ce614ce9 : FAILURE in 24m 52s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5cb688046cfc4332a2b966e8558cee95 : FAILURE in 24m 49s\n- loci-manila https://zuul.opendev.org/t/openstack/build/ce45dbc3c6f341ebb72e2098a7041792 : FAILURE in 33m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5cc4e49d34bb44789bc230f57c5c5ca1 : FAILURE in 25m 18s\n- loci-nova https://zuul.opendev.org/t/openstack/build/4e5c5e1467354c408519dc0d3e71dbff : FAILURE in 25m 33s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7ef9deb6fed54b9fb90b04ba96cca780 : FAILURE in 24m 41s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a0fe3d2f5f9d4e199db8b37040b53ebe : FAILURE in 24m 14s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/67524f93fe904977843b4b16c851d773 : FAILURE in 24m 52s","accounts_in_message":[],"_revision_number":2},{"id":"306c0b4d23d62471ecc335b23754d79f2e722302","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 01:11:33.000000000","message":"Patch Set 2:\n\nneutron-lib 2.0.0 wheel is absent in pip index. Also reviewing gates i see a necessity to bring some life into them","accounts_in_message":[],"_revision_number":2},{"id":"facf938a62029087750a84b89d385d0f693e87e7","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 06:03:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fc0fb65998433062102de5f92ba4b36ee21bb97a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 07:30:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c02a748e90ee476e91a660f34cea6400 : TIMED_OUT in 1h 01m 10s\n- loci-glance https://zuul.opendev.org/t/openstack/build/13f38f39ce3d41748b9b2c9c8354d0bf : TIMED_OUT in 1h 01m 16s\n- loci-heat https://zuul.opendev.org/t/openstack/build/18c5f11fede441fe92f6fd1ab3f0444c : SUCCESS in 40m 46s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3ae3e36e9a8d485dbe5e19847481dff4 : SUCCESS in 42m 10s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5e45d8f56a9b474b86fc91266f270061 : SUCCESS in 43m 41s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f7be2762f61f4cd7bcef1766ce09368c : SUCCESS in 45m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f067b9220cbd414e8293053938cb10d0 : SUCCESS in 38m 38s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/3d634f118da7495494e9239bc3dbc2e6 : TIMED_OUT in 1h 01m 01s\n- loci-nova https://zuul.opendev.org/t/openstack/build/dec5bbad7d184ad485811f837ce466fe : TIMED_OUT in 1h 00m 54s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2c836b50145a4b8989e25945bd5c74fe : TIMED_OUT in 1h 01m 08s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cc0f7a95de154731a64c65b290d79872 : TIMED_OUT in 1h 01m 44s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4a943918a32642479feea853328e3d43 : TIMED_OUT in 41m 24s","accounts_in_message":[],"_revision_number":3},{"id":"d08c9d5494944e4d1ba61794ce0c4ce3112a507a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 14:03:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e7e39d7cebd2d3ce473743844e4ba0918d7baf89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 14:55:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0c5b6893dd8847d889b85042a86def3d : SUCCESS in 41m 48s\n- loci-glance https://zuul.opendev.org/t/openstack/build/92e09c0f1d6a44c6b79ebb53e70827bc : SUCCESS in 34m 31s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fce74a8381604b7a9efa8e55d79e18b8 : SUCCESS in 38m 49s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e4e659c9ffe345c6912b063ad6e827f2 : SUCCESS in 40m 43s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e4111f4d54a041fc87027f037b5674d2 : SUCCESS in 41m 35s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/cb579d0e6e6a468ba7e73af78185d7d0 : SUCCESS in 35m 44s\n- loci-manila https://zuul.opendev.org/t/openstack/build/89b18ae419af4dbc83f4ea27e16f51d3 : SUCCESS in 36m 31s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/74e6ebb10d6b43ef977efd79424877be : SUCCESS in 40m 30s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a9a3f2ffb22f402b8a6b5adc08bb3bc8 : SUCCESS in 45m 52s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2f8cd45b81f845f3adae3581d5bf8ef9 : SUCCESS in 40m 49s\n- loci-placement https://zuul.opendev.org/t/openstack/build/20c8ab0b6f514c8eb9c8603555e6898d : SUCCESS in 38m 19s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/52b382d4a73a41f9b4220f45f8326a98 : SUCCESS in 30m 36s","accounts_in_message":[],"_revision_number":3},{"id":"6f3c3cc5c0846f9496103025cc6313373c2e5442","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 11:43:41.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ba1e76f0fbbdd6ee8d628f418c2b13578ac8d788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 12:38:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/95b29f2258f443a9a5eb33597b120210 : SUCCESS in 47m 02s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d261a8d7916247a0870138c9cb15b120 : SUCCESS in 44m 09s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2a79a4f079e646fa9ea58d7d20981ffc : SUCCESS in 44m 16s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/301c404f46094d9f8b7a4c4468b396e6 : SUCCESS in 45m 24s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/7eea7ca3b42b4fda9f97acc7b60147cd : SUCCESS in 43m 57s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/000789d97ee44c48a3d35dd4702ca22a : SUCCESS in 43m 12s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e28bd615361c4f0b96075179e551590d : SUCCESS in 44m 22s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b3501fdce1354532ba551c060a18ce1c : SUCCESS in 48m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c16240a0956249ae82425f25d0fea797 : SUCCESS in 50m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/10ed0720ced44d6d9f8f8075f9f7681b : SUCCESS in 46m 33s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f40901f00e704575baabea5d0d6a97dd : FAILURE in 48m 41s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a9d8c3b057064d508678bddf7805f452 : SUCCESS in 35m 55s","accounts_in_message":[],"_revision_number":4},{"id":"fa13f3f822575ca0fa2309e3a764c6d28125a751","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 15:58:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"44b648b3c4eab5f23f1ddd687c7a25f90add7842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 16:54:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/711fee0e10d04486b465c90fe4a17307 : FAILURE in 39m 07s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a1fd5cbf8c0c44c6862bbc9c9d30a5f7 : FAILURE in 32m 17s\n- loci-heat https://zuul.opendev.org/t/openstack/build/90aaa3aa0cf54c60b121070ffb9b749e : FAILURE in 37m 59s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9dfe799cdb9942ada8c354e2732201c5 : FAILURE in 29m 13s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c0ef44a6112e4686b2b3a4a72e7a11c0 : FAILURE in 37m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/013a3d4cc7984129bae9d1795104385f : FAILURE in 22m 04s\n- loci-manila https://zuul.opendev.org/t/openstack/build/ca1156fe0c6e44d282605f9f66386075 : FAILURE in 37m 53s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2d120bec4cf049fda83db2428e6459be : FAILURE in 35m 58s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f95274cd609140a7ab3a1f08fa98f18a : FAILURE in 43m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7cceaa522b34403caa1df5d110b48251 : FAILURE in 48m 30s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e3fca452085b4bf8a6b53845e971de07 : FAILURE in 38m 08s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/9c6c8fbd878c454fa9194d2416d52b97 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":5},{"id":"881c5d1abae0924f62be0899142b770fc373661e","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 21:14:48.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"cdeafe6499a311da0b3f23ab3ea6c7d6befedadc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 21:34:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/6146ef6ced36419985a32934d6466c68 : FAILURE in 7m 32s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f025531b3e2742b6bdb24dd2cb16c608 : FAILURE in 6m 08s\n- loci-heat https://zuul.opendev.org/t/openstack/build/db678d9214b2404aacd3fd98e00908e7 : FAILURE in 6m 41s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ae66744ffb64438cad29f196c1ef55c1 : FAILURE in 4m 21s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3abe7be8ab1a4ab48af12f99a31aaf1d : FAILURE in 5m 52s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9cab4f45bef04e42929d84a81c82644e : FAILURE in 6m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/afe7ed85f3014943b89f38940867e589 : FAILURE in 7m 34s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/493689b8d8774ce69cee4afd909e0f62 : FAILURE in 6m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bb2a2daf3a64499883561fa5ed36e7b8 : FAILURE in 12m 35s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5ab904e2c795457c948800d9ccbb51ec : FAILURE in 12m 46s\n- loci-placement https://zuul.opendev.org/t/openstack/build/50e2765c25c2499ba5ad0e29ad1395d5 : FAILURE in 11m 09s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8ae5d905488642cb992a4aca3d1d091a : FAILURE in 5m 15s","accounts_in_message":[],"_revision_number":6},{"id":"9610f8418f65e2606e5f367c367b0e0e1852a2b4","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 14:28:40.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"69ebd51a2b4f9c4ad545efd8558887c12128d6ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 15:41:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/576bacf3b060453ca4c44e9b6235014b : SUCCESS in 47m 58s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e3b7cedf5b8c482c82999b3164ee98fc : SUCCESS in 55m 53s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f54bdcb471cb422299b39fc9be92abb9 : SUCCESS in 1h 02m 11s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/723b06c6cfe64b02aee2d55f4abca5fe : FAILURE in 51m 25s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/13bcd77716754da9a0ea998f96af8260 : SUCCESS in 51m 48s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b09df0ac637143f985e130e09b9086d8 : SUCCESS in 57m 58s\n- loci-manila https://zuul.opendev.org/t/openstack/build/43eb4899fe574a04bad71a46e099dec7 : SUCCESS in 1h 10m 03s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/aa9f14506a424e569f16db70ad8bd0e7 : FAILURE in 1h 08m 48s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fa4d47424add4953894ddd2ee48888e1 : SUCCESS in 1h 00m 01s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/625c4118ba6d423292f2f1cc81ec5c48 : SUCCESS in 58m 08s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c382c47b42b44af89b07a9db5cee7d82 : SUCCESS in 51m 04s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/da82a10e4bd645b884618e497e637cbf : TIMED_OUT in 41m 00s","accounts_in_message":[],"_revision_number":7},{"id":"353b8e401e9f70f0f38da370e2bbc33cd21e45b9","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 19:29:40.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"234fe57ff509d0d976ba070320b584e30db1e313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 20:52:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/626919af24134cfeaaa8846ec154021c : SUCCESS in 1h 03m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bf77f45da64343e1a806ef514aeb5ade : SUCCESS in 46m 23s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0531bcfa761a487486f8043595f2a60c : SUCCESS in 1h 19m 33s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/88b6825f058348f6ab41e4a182517553 : SUCCESS in 1h 19m 08s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/cd1d28ed083b422b843a47cf3b3bfda8 : SUCCESS in 1h 10m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/bc6df1be4bb2410b8e609d4c3da4b907 : SUCCESS in 1h 19m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7bde433f94954f5095142c49d229038d : SUCCESS in 45m 16s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cdbc6950bf46499c82b716d032071ab1 : SUCCESS in 1h 16m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/05a74c7d49f84b7d807536f26c20978a : SUCCESS in 1h 07m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/38199b0a4e17464a99429c1fd9b212f2 : SUCCESS in 56m 49s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a04bf15644ee4ee98e503cd3ef43268c : SUCCESS in 49m 06s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c0ce3ce0ba1c406c9de58df154c02ee3 : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":8},{"id":"fecff6441294ebaf15e283c41d509ee6312e2172","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 12:15:08.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"13165f8160afa488774972d3dcd08d8ee2e92733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 13:17:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/61481f6b2b9e4dd0ab288138c2737b5e : SUCCESS in 38m 12s\n- loci-glance https://zuul.opendev.org/t/openstack/build/44334b513b41462b97573778a1bd8f88 : TIMED_OUT in 1h 01m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/6c12dcbc0da84990a7e42376ae839f1b : SUCCESS in 45m 29s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f9020c0fe66d4160b4ff73808c56f7bb : SUCCESS in 45m 09s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a3daeaf15a1448b48d2149ae0fd1df5e : SUCCESS in 34m 01s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6312e53498b0419c96b33813fed85eaa : SUCCESS in 33m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/27c488bd85734ecea24a2490e94689ce : SUCCESS in 38m 29s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7ee3a439abf04698a1cd1abc0a96c711 : SUCCESS in 46m 54s\n- loci-nova https://zuul.opendev.org/t/openstack/build/93260c35145b4090a29c60268dd8c6b3 : SUCCESS in 43m 14s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/21e457a83fd94d3897d2b3d96ff26e05 : SUCCESS in 45m 02s\n- loci-placement https://zuul.opendev.org/t/openstack/build/308a8f814531430a839a9ecd097eb1a1 : SUCCESS in 55m 47s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/00486d15e3ba477d9a7f3661c51bf188 : SUCCESS in 30m 03s","accounts_in_message":[],"_revision_number":9},{"id":"7b403c0dabd96eefb4fdd22968b0d20b7108fa28","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 13:21:51.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"3446b2f3db28641d939fe7d2482dfbdf5b16e93b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:24:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d50d859793684ae4b4138f3ac620b872 : SUCCESS in 34m 52s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0983911118b24931a57c1c2e8a8e1261 : SUCCESS in 37m 42s\n- loci-heat https://zuul.opendev.org/t/openstack/build/65e68d272be04959a47474b4db00064b : SUCCESS in 35m 23s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4f2d10db300c419e9d6ba3c701c6e7b5 : SUCCESS in 37m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1e48660a82184d7bacb1367b696fd2bb : SUCCESS in 54m 00s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9f1505f3687945aeb0e3e17fddfdd920 : SUCCESS in 43m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/be804fadbc894a269b2cc82327cea660 : SUCCESS in 45m 26s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/309035071215409a8d3a9e71e844d208 : SUCCESS in 1h 00m 23s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ed906d06c8974ac8bfda84dc7e140ee7 : SUCCESS in 50m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/eca8bd13c48e4601ac43ecd1e7b7846a : SUCCESS in 57m 09s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ba62427391734ab8b974724b99686adb : SUCCESS in 44m 26s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/2e17e0816ded410ab9a8babbdb0d0e28 : SUCCESS in 38m 01s","accounts_in_message":[],"_revision_number":9},{"id":"5d773702f5362c67a3896f0e481f69796748c5fa","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 14:47:25.000000000","message":"Patch Set 9:\n\nWF please. Thank you!","accounts_in_message":[],"_revision_number":9},{"id":"97de2f314f0423750c92d26167356fb4f2ab669c","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2020-01-27 14:51:19.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"6d7d963802fd118a1c954bfe1223b83287ea6e4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:51:32.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5d81ba239449e908b73c14ead624e862146ee77c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 15:53:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"d7e3aa8165277141625f41841c4ba1efd28ee44b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 15:53:48.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ff1c30d1db8c4067acd0117859e14e99 : SUCCESS in 36m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8cce25ba85724c55b0f182cfb927cb32 : SUCCESS in 1h 00m 26s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b0b92a3ed77541139d78aa0bbdc6764e : SUCCESS in 44m 57s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/63681981c58f4c988a84c2c53bd459bb : SUCCESS in 42m 53s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/502beaf3270a4fb2b522fc3cbb5e4e44 : SUCCESS in 55m 59s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/947cc9b168444b28bdd2bd907ac81f80 : SUCCESS in 39m 40s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b6b10ad1a5604011a12603f37dd183e2 : SUCCESS in 51m 08s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2c3a4858190a4467b77fcac0ae3fdf4c : SUCCESS in 40m 14s\n- loci-nova https://zuul.opendev.org/t/openstack/build/58f89b325d2f474fb0bf9f44499fc112 : SUCCESS in 41m 12s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/beebbf1b1e1a4c98b73584853121a798 : SUCCESS in 45m 17s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7bbd90a4f5604237adb2bbf6effeb2ea : SUCCESS in 27m 53s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"8ed58b10f6bed12b2a74296e1571b457b577594a","revisions":{"da6e3ca04d6b8e8596427094db69b13c165114f7":{"kind":"REWORK","_number":1,"created":"2019-11-12 17:16:26.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/86/693886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/86/693886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/86/693886/1"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:15:48.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:16:16.000000000","tz":-360},"subject":"Allow comments in dep files","message":"Allow comments in dep files\n\nChange-Id: I2d4c65268cec70c1d03ed1556c5ab5dbab94c398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/da6e3ca04d6b8e8596427094db69b13c165114f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/da6e3ca04d6b8e8596427094db69b13c165114f7"}]},"branch":"refs/heads/master"},"76f542f796cbd5708583bd86278a45c1046ff65f":{"kind":"REWORK","_number":2,"created":"2019-11-12 17:28:04.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/86/693886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/86/693886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/86/693886/2"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:15:48.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:28:01.000000000","tz":0},"subject":"Allow comments in dep files","message":"Allow comments in dep files\n\nChange-Id: I2d4c65268cec70c1d03ed1556c5ab5dbab94c398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76f542f796cbd5708583bd86278a45c1046ff65f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76f542f796cbd5708583bd86278a45c1046ff65f"}]},"branch":"refs/heads/master"},"57d210e45aa5087fc0111b06b7d03baffb6341e8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-17 06:03:48.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/86/693886/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/86/693886/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/86/693886/3"}}},"commit":{"parents":[{"commit":"54efaeef9c3a2a515db4bade7ae9bd920f5bc6ee","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/54efaeef9c3a2a515db4bade7ae9bd920f5bc6ee"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:15:48.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 06:03:48.000000000","tz":0},"subject":"Allow comments in dep files","message":"Allow comments in dep files\n\nChange-Id: I2d4c65268cec70c1d03ed1556c5ab5dbab94c398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/57d210e45aa5087fc0111b06b7d03baffb6341e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/57d210e45aa5087fc0111b06b7d03baffb6341e8"}]},"branch":"refs/heads/master"},"2c3c3d4a98075e0fdea9f4700d0927ee1a968f45":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-01-18 11:43:41.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/86/693886/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/86/693886/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/86/693886/4"}}},"commit":{"parents":[{"commit":"21f7a0af263752977ecb577b7d52223dd637e3d4","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/21f7a0af263752977ecb577b7d52223dd637e3d4"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:15:48.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 11:43:41.000000000","tz":0},"subject":"Allow comments in dep files","message":"Allow comments in dep files\n\nChange-Id: I2d4c65268cec70c1d03ed1556c5ab5dbab94c398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2c3c3d4a98075e0fdea9f4700d0927ee1a968f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2c3c3d4a98075e0fdea9f4700d0927ee1a968f45"}]},"branch":"refs/heads/master"},"26f9eb530c92f99c4856cf6fc35328fc3d3f082d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-01-20 15:58:59.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/86/693886/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/86/693886/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/86/693886/5"}}},"commit":{"parents":[{"commit":"cafdc187afe29e6e245c56a83a9a21190bbaf8f3","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cafdc187afe29e6e245c56a83a9a21190bbaf8f3"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:15:48.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 15:58:59.000000000","tz":0},"subject":"Allow comments in dep files","message":"Allow comments in dep files\n\nChange-Id: I2d4c65268cec70c1d03ed1556c5ab5dbab94c398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/26f9eb530c92f99c4856cf6fc35328fc3d3f082d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/26f9eb530c92f99c4856cf6fc35328fc3d3f082d"}]},"branch":"refs/heads/master"},"89e078a6814c7a2882e950d42a2486d3601bbdb7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-01-20 21:14:48.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/86/693886/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/86/693886/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/86/693886/6"}}},"commit":{"parents":[{"commit":"bcd119e10d865252fd0de96e7c0b810c5f8e1d15","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bcd119e10d865252fd0de96e7c0b810c5f8e1d15"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:15:48.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 21:14:48.000000000","tz":0},"subject":"Allow comments in dep files","message":"Allow comments in dep files\n\nChange-Id: I2d4c65268cec70c1d03ed1556c5ab5dbab94c398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/89e078a6814c7a2882e950d42a2486d3601bbdb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/89e078a6814c7a2882e950d42a2486d3601bbdb7"}]},"branch":"refs/heads/master"},"6447387257934c3fb700d4b7e86fc12a23d94fd6":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-01-21 14:28:40.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/86/693886/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/86/693886/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/86/693886/7"}}},"commit":{"parents":[{"commit":"82fbed8c851a89952a8cd3b5293d73b755adc474","subject":"Enable Debian gate build and fix mirrors related issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/82fbed8c851a89952a8cd3b5293d73b755adc474"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:15:48.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 14:28:40.000000000","tz":0},"subject":"Allow comments in dep files","message":"Allow comments in dep files\n\nChange-Id: I2d4c65268cec70c1d03ed1556c5ab5dbab94c398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6447387257934c3fb700d4b7e86fc12a23d94fd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6447387257934c3fb700d4b7e86fc12a23d94fd6"}]},"branch":"refs/heads/master"},"5245cfc64fb347ac4b8f129c0ed3788069eefcb2":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-01-21 19:29:40.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/86/693886/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/86/693886/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/86/693886/8"}}},"commit":{"parents":[{"commit":"91f9d9bf7950e75354a138c082674099d4d9f9b5","subject":"Enable Debian gate build and fix mirrors related issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/91f9d9bf7950e75354a138c082674099d4d9f9b5"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:15:48.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 19:29:40.000000000","tz":0},"subject":"Allow comments in dep files","message":"Allow comments in dep files\n\nChange-Id: I2d4c65268cec70c1d03ed1556c5ab5dbab94c398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5245cfc64fb347ac4b8f129c0ed3788069eefcb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5245cfc64fb347ac4b8f129c0ed3788069eefcb2"}]},"branch":"refs/heads/master"},"8ed58b10f6bed12b2a74296e1571b457b577594a":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-01-27 12:15:08.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/86/693886/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/86/693886/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/86/693886/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/86/693886/9"}}},"commit":{"parents":[{"commit":"d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2019-11-12 17:15:48.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-27 12:15:08.000000000","tz":0},"subject":"Allow comments in dep files","message":"Allow comments in dep files\n\nChange-Id: I2d4c65268cec70c1d03ed1556c5ab5dbab94c398\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ed58b10f6bed12b2a74296e1571b457b577594a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8ed58b10f6bed12b2a74296e1571b457b577594a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
