)]}'
{"id":"openstack%2Floci~697424","triplet_id":"openstack%2Floci~master~Id8f2133429b08e941ead724915049073df96cff7","project":"openstack/loci","branch":"master","topic":"test-gate","hashtags":[],"change_id":"Id8f2133429b08e941ead724915049073df96cff7","subject":"DNM: testing zuul gate fixes","status":"ABANDONED","created":"2019-12-05 09:25:53.000000000","updated":"2019-12-13 08:30:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5459ffa0bf7fcdd082d976211faca93956423377","_number":697424,"virtual_id_number":697424,"owner":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-05 13:21:06.000000000","_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-05 10:03:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-05 10:03:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2710419215f2e88f73edd1680d45ccad8d83c4d","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2019-12-05 09:25:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc6e1ad529908747b5ab47b39f530d7f045fc425","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 10:03:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/de3ac5b57cd64567a455677b50feeb55 : FAILURE in 28m 10s\n- loci-glance https://zuul.opendev.org/t/openstack/build/639f387850ca4720bb3f170a3cdd148d : FAILURE in 30m 40s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3baec86abbd043aab38b241e39e9dd61 : FAILURE in 34m 09s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/7afd8edd34ee47ee8be98f839e4c9b57 : FAILURE in 30m 06s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/4b99600c2a524dcf863b8cf5e9136a0e : FAILURE in 28m 19s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9491ebc5a5eb435b93479c62805cf1c4 : FAILURE in 27m 45s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5cac0b033ee7404fa59f95380350740f : FAILURE in 27m 29s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/25f413b66a934c7e83e98c3aab60e5ed : FAILURE in 29m 59s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3ccd302d0d75437783ac8445e185bb78 : FAILURE in 27m 43s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/47c7bfb7c86645388c69eef5d0317bb1 : FAILURE in 29m 19s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8d3b4aa905044ead81e36e29d5b7155a : FAILURE in 34m 12s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/04b7900b82b741ec93ed6e430c80f1c9 : FAILURE in 29m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bb87867e3d18668bf962221ee0ef6bdde11156e0","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2019-12-05 13:21:06.000000000","message":"Patch Set 1:\n\nERROR: Package \u0027edgegrid-python\u0027 requires a different Python: 2.7.5 not in \u0027\u003e\u003d2.7.10\u0027","accounts_in_message":[],"_revision_number":1},{"id":"4d9e0ff152a2152e3a8382c5fd9d9f897f709a80","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2019-12-13 08:30:33.000000000","message":"Abandoned\n\nSeems to be working again","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82ad44699dcb28acd8296a07f2e2faf06cce19d5","revisions":{"82ad44699dcb28acd8296a07f2e2faf06cce19d5":{"kind":"REWORK","_number":1,"created":"2019-12-05 09:25:53.000000000","uploader":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"ref":"refs/changes/24/697424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/697424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/697424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/697424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/697424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/697424/1"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2019-12-05 09:27:02.000000000","tz":60},"committer":{"name":"Dantali0n","email":"info@dantalion.nl","date":"2019-12-05 09:27:02.000000000","tz":60},"subject":"DNM: testing zuul gate fixes","message":"DNM: testing zuul gate fixes\n\nChange-Id: Id8f2133429b08e941ead724915049073df96cff7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/82ad44699dcb28acd8296a07f2e2faf06cce19d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/82ad44699dcb28acd8296a07f2e2faf06cce19d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
