)]}'
{"id":"openstack%2Floci~703196","triplet_id":"openstack%2Floci~master~I6d5410da02ff9717418de55876162d34b6185e1d","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I6d5410da02ff9717418de55876162d34b6185e1d","subject":"Avoid storing cache in pip install for requirements image","status":"MERGED","created":"2020-01-17 21:20:47.000000000","updated":"2020-07-24 05:33:36.000000000","submitted":"2020-07-24 05:33:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703196-1595568816161-78954b85","meta_rev_id":"19ae6a887b19bc2cdbaf5df10e3ae72dd37accca","_number":703196,"virtual_id_number":703196,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 05:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"recommended":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":1,"date":"2020-07-02 18:16:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":1,"date":"2020-07-02 13:38:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2020-05-15 10:54:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-05 15:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2020-07-24 04:40:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2020-07-24 04:40:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 21:40:22.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-02-05 15:16:59.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-05-15 10:54:51.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-07-02 13:38:36.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"state":"REVIEWER"},{"updated":"2020-07-02 18:16:17.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2020-07-24 04:40:00.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-07-24 05:33:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e2a8ad741d5296acdaa2d406ac737d79ecd7a92","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 21:20:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3883a434f6fa3b9b154596ef0d1b06ca9aa165b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 21:59:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e7f8892dc2864922b81896fdfe6c018b : FAILURE in 31m 33s\n- loci-glance https://zuul.opendev.org/t/openstack/build/042612e642c24c0ab0f2ce00c0343543 : FAILURE in 27m 14s\n- loci-heat https://zuul.opendev.org/t/openstack/build/85e2d9cc0891472e89c308d657713428 : FAILURE in 25m 56s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f9426cbafdcb4918aa6ba203d959fe72 : FAILURE in 32m 08s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c71e98918c624910bafc7f9e5181addc : FAILURE in 26m 23s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/aecc40de333e4a0f83337085b064c20f : FAILURE in 28m 35s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5a2d2c5a9b1141719e728b323f4c8aad : FAILURE in 26m 26s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/92f98634264c493d8d112885df5718f6 : FAILURE in 27m 27s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e8a6a584ca22429284cd504b53393993 : FAILURE in 28m 34s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/95327e4c8e03496ea3225da6aeeb67cd : FAILURE in 27m 07s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ce2c88331b3e48209cfa16d98de55cfa : FAILURE in 27m 12s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/79c04ffe3898442e84a570aa529a0ec9 : FAILURE in 26m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ee5936e9ffd76748c5641b1308b803eb73dfbfdf","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 14:54:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2fb3eab45e9f2c9f5470ba5bd9bf8e7285913023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 15:44:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8ef208e279c14e4ab1407efb465a0545 : SUCCESS in 38m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bc287e9c487f4c649591287b82a60c6d : SUCCESS in 43m 57s\n- loci-heat https://zuul.opendev.org/t/openstack/build/eb8184e2220d488cb6b19534f8c04b99 : SUCCESS in 40m 23s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5db1ac0b361045768257d12fc83a1a88 : SUCCESS in 39m 10s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f185f05a4b4c45598b0c7d25c8bf25f2 : SUCCESS in 42m 02s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/db454f940c1d45bebc318d92a7d4537e : SUCCESS in 37m 27s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a52f603624f447709934010be17c1273 : SUCCESS in 36m 38s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/1a1f2e439b6247a3a4411c9b6859ecc2 : SUCCESS in 45m 20s\n- loci-nova https://zuul.opendev.org/t/openstack/build/20275b454c784daf96db53a0127fab58 : SUCCESS in 40m 53s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b2317764492940dcb35102f1d6201564 : SUCCESS in 43m 26s\n- loci-placement https://zuul.opendev.org/t/openstack/build/64f88de9c93d45a99defc1d039f7246d : FAILURE in 36m 39s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ab03f26ed1924498854eefe57b0ff27d : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":2},{"id":"63f3481da2d88a928500231e6a8a27644ac7a65a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 15:59:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b83955eb31caeb9bc073ecd8fdaaaa91aacb4915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 16:52:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/dc87e78eb5d34af7819643d57bf13f2d : FAILURE in 39m 16s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ea9d7847b71f4ec6b449888b2a872127 : FAILURE in 29m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/756a3e1385f94e8eb0e5d1e19f39a997 : FAILURE in 22m 52s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e4b292b09976488eaf7ad8dd3d18171b : FAILURE in 36m 10s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/81b0809620cf45448521208de3e54481 : FAILURE in 44m 26s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4c813cf1a8854d94b5b74fd56f1f6ee8 : FAILURE in 44m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/90d9521cf17d41c5b39d7ffd3f2b9d66 : FAILURE in 44m 39s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7ca2274cff834f3bb0b1b3a589d368e3 : FAILURE in 40m 13s\n- loci-nova https://zuul.opendev.org/t/openstack/build/98f9a24077284d9fb78bce3b3264ed9e : FAILURE in 43m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e3f457577122437c827d3c851dab0f42 : FAILURE in 48m 59s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f8bece1b0d104e6092999f68526548f2 : FAILURE in 32m 15s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6bb8ee9892904e3fbe73e8e0781f57c4 : FAILURE in 31m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2f3ad9e2ebbc742a505fa001d96cae0e726e658f","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 19:46:09.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ddfb4e44059ff75b70187b655034a70e15c68f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 20:48:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8a8c14b914fe4cd1b1fd0d163f47201e : FAILURE in 32m 08s\n- loci-glance https://zuul.opendev.org/t/openstack/build/501cc7491f1f49bca182bb6095dfa2c2 : SUCCESS in 49m 14s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d84fe2920ea34ee6af34ee31293da8f7 : SUCCESS in 44m 14s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/295f8725a400462c8cb1f513fb93c270 : SUCCESS in 58m 06s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e98db839886941ff8f69050551c04dc0 : SUCCESS in 48m 43s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9c7e8a84c8634142a60d6f4caecead35 : SUCCESS in 39m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/04439d9509a04ae681fb5115c53ec0a1 : SUCCESS in 46m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c941cdd4c6f54418be79c6f1e9842839 : SUCCESS in 43m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/551a59ddae8d4d4ea269c9b910c08ba1 : SUCCESS in 44m 42s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/afa194aa4b114ba18aca6a2719801581 : SUCCESS in 42m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/abefb22af6da474884a565bbc60b0fb4 : SUCCESS in 33m 52s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0a6389743d0949a0af97584ca334b7f2 : TIMED_OUT in 41m 19s","accounts_in_message":[],"_revision_number":4},{"id":"c5b3c5246f08ef17b70cb4d3e3dc897eae61ffd3","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 20:48:43.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5c9919fc267a8586075e46560b6930c58654a51a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 21:52:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/23dc750de0054b2f82ff3d266765f5bd : SUCCESS in 45m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/569dc05fc7324153a7b94a484bd9eda6 : SUCCESS in 52m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/646587ff55b142e3b32ecd7731ffd285 : SUCCESS in 48m 12s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2ff356ec8538443db7bb38d1ce2753cd : SUCCESS in 56m 36s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/7094fb823388495d91489e5b3e20be28 : SUCCESS in 45m 54s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/abb2bdc5504b4569ae27a2cdc09fd653 : SUCCESS in 45m 51s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b8a5b23cf96f4a479bc84c6a5439e41e : SUCCESS in 53m 19s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f7020a56e7924ed8bb5bc4d55b83af84 : SUCCESS in 48m 00s\n- loci-nova https://zuul.opendev.org/t/openstack/build/cabe9416b627403daceab77cf6cd46c7 : SUCCESS in 58m 37s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/bad96e2ebae14b7a8f6a265e7009ca4a : SUCCESS in 48m 37s\n- loci-placement https://zuul.opendev.org/t/openstack/build/31a494605e1d49449398396d96196985 : SUCCESS in 41m 45s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ca902f10f06d430c9c3a522745c0aa47 : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":4},{"id":"6bb76862fc1e2e229e006080adf558df07b1d438","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 18:48:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"321d02a6dd24fd06e25dd5a8bf87a8bb9b95e5da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 19:46:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ebebc7caedc047cc81e880700e8400ba : SUCCESS in 41m 11s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cd77b407b2074f41a722d891873fc92d : SUCCESS in 46m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/58e1e0866c094c81b31defa006d6a5cc : SUCCESS in 41m 12s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c50919038bf54f809766992ab215b4b0 : SUCCESS in 41m 21s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/88ef30af57484995adb5494f4491a692 : SUCCESS in 39m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f5acc910c98342e9bfe36bf98cb0a974 : SUCCESS in 44m 46s\n- loci-manila https://zuul.opendev.org/t/openstack/build/1dc0e7000ec0409680ca5c64417dd247 : SUCCESS in 43m 10s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/785332e08c57448894b8a24559ef23ba : SUCCESS in 50m 47s\n- loci-nova https://zuul.opendev.org/t/openstack/build/348fc0a35bd84466ba4f82f4bfc8d89d : SUCCESS in 45m 21s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f2e7a5c96a3446719131168a6b48ed4f : SUCCESS in 37m 00s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4143b375e38b46da8c212160f52e16f8 : SUCCESS in 31m 46s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1e998d3fcf024380aec797ee8e8fb93d : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":5},{"id":"e24aa81a7f1e4c9d920317f1efa37f3998dfd541","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-05 15:16:59.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c69360ae39092d9fb172209d0997da6f0864358b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-15 10:54:51.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"6c33a02b822ecc262aea48a2f10429cf97f63f1c","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2020-07-02 13:38:36.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d34e6728329909d65e411873ef32085f9a32b582","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-07-02 18:16:17.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"24bd3e18a9a81dbf9de14194b20f092a7ba63366","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-07-24 04:40:00.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ad8a58824e5df92894a29dd39ab674fcab087124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 04:40:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"78f9e9f211de5702e0a869b26ea94bf70e54e94d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 05:33:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/81c87d8f2cce40b4b075d0f112cd8761 : SUCCESS in 43m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/78ce97a4f2de46778971ba7ba7e8a196 : SUCCESS in 38m 28s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c9a3c5453365469084fcfe7611711907 : SUCCESS in 40m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/50a709457de843c195f890960d087f9b : SUCCESS in 41m 58s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0bd9114ab2184c4cab2f0edf5fe306a6 : SUCCESS in 40m 26s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1bfc34f214b1442abc3f01b4eb181255 : SUCCESS in 39m 37s\n- loci-manila https://zuul.opendev.org/t/openstack/build/dcba7bc4a4204e0ba120d16d9242006e : SUCCESS in 40m 44s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5d4a077a7673482db564d5375d4b955e : SUCCESS in 42m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/42203e37e5e54c01b7dce12dfbb994f3 : SUCCESS in 44m 48s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7d99bb917722402fb40c54e84eea0d55 : SUCCESS in 41m 07s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e0fdd199c55146b0aa0047353d398add : SUCCESS in 30m 35s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/ff1d89856a1244ddb7bbd83bba750df4 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":5},{"id":"19ae6a887b19bc2cdbaf5df10e3ae72dd37accca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 05:33:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a7b46b1aa59ed65c421ce8d4872d9cc8906c13ec","revisions":{"f8062f45597a52b639849a128e6773ff6712bb61":{"kind":"REWORK","_number":1,"created":"2020-01-17 21:20:47.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/96/703196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/703196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/703196/1"}}},"commit":{"parents":[{"commit":"c94e375f8da31363c5c8155027e2bfeb4dc7accb","subject":"Fix broken pydep for older releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94e375f8da31363c5c8155027e2bfeb4dc7accb"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 21:14:15.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 21:14:15.000000000","tz":-360},"subject":"Avoid storing cache in pip install for requirements image","message":"Avoid storing cache in pip install for requirements image\n\nIf additional pip packages are provided for requirements image, they\nwill be installed into system before building wheels and, in addition,\nwheel stored to local pip cache. Later, during wheels creation, this\ncached wheel will be used and then deleted as system will think that\npackage is available in wheel form.\n\nThis only matters when package is not available in wheel form and needs\nto be a part of requirements image.\n\nChange-Id: I6d5410da02ff9717418de55876162d34b6185e1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}]},"branch":"refs/heads/master"},"dfcf5ced1933758603641480bc4075759ba99692":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-18 14:54:24.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/96/703196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/703196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/703196/2"}}},"commit":{"parents":[{"commit":"4a948abf4457e102293f64a56472165de6f7689b","subject":"Install libpython[23].x after python installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4a948abf4457e102293f64a56472165de6f7689b"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 21:14:15.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 14:54:24.000000000","tz":0},"subject":"Avoid storing cache in pip install for requirements image","message":"Avoid storing cache in pip install for requirements image\n\nIf additional pip packages are provided for requirements image, they\nwill be installed into system before building wheels and, in addition,\nwheel stored to local pip cache. Later, during wheels creation, this\ncached wheel will be used and then deleted as system will think that\npackage is available in wheel form.\n\nThis only matters when package is not available in wheel form and needs\nto be a part of requirements image.\n\nChange-Id: I6d5410da02ff9717418de55876162d34b6185e1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfcf5ced1933758603641480bc4075759ba99692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfcf5ced1933758603641480bc4075759ba99692"}]},"branch":"refs/heads/master"},"cfe5579a294ab572eb9e2ed0edf80295524e3b89":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-20 15:59:12.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/96/703196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/703196/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/703196/3"}}},"commit":{"parents":[{"commit":"3987c421a2914cbc7e6dc20af57ce2679ad10026","subject":"Install libpython[23].x after python installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3987c421a2914cbc7e6dc20af57ce2679ad10026"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 21:14:15.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 15:59:12.000000000","tz":0},"subject":"Avoid storing cache in pip install for requirements image","message":"Avoid storing cache in pip install for requirements image\n\nIf additional pip packages are provided for requirements image, they\nwill be installed into system before building wheels and, in addition,\nwheel stored to local pip cache. Later, during wheels creation, this\ncached wheel will be used and then deleted as system will think that\npackage is available in wheel form.\n\nThis only matters when package is not available in wheel form and needs\nto be a part of requirements image.\n\nChange-Id: I6d5410da02ff9717418de55876162d34b6185e1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cfe5579a294ab572eb9e2ed0edf80295524e3b89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cfe5579a294ab572eb9e2ed0edf80295524e3b89"}]},"branch":"refs/heads/master"},"9eacb98344735d84e0f72e0a7cfa55e997fc8206":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-01-21 19:46:09.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/96/703196/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/703196/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/703196/4"}}},"commit":{"parents":[{"commit":"d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 21:14:15.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 19:46:09.000000000","tz":0},"subject":"Avoid storing cache in pip install for requirements image","message":"Avoid storing cache in pip install for requirements image\n\nIf additional pip packages are provided for requirements image, they\nwill be installed into system before building wheels and, in addition,\nwheel stored to local pip cache. Later, during wheels creation, this\ncached wheel will be used and then deleted as system will think that\npackage is available in wheel form.\n\nThis only matters when package is not available in wheel form and needs\nto be a part of requirements image.\n\nChange-Id: I6d5410da02ff9717418de55876162d34b6185e1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9eacb98344735d84e0f72e0a7cfa55e997fc8206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9eacb98344735d84e0f72e0a7cfa55e997fc8206"}]},"branch":"refs/heads/master"},"a7b46b1aa59ed65c421ce8d4872d9cc8906c13ec":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-01-27 18:48:42.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/96/703196/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/96/703196/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/96/703196/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/96/703196/5"}}},"commit":{"parents":[{"commit":"497691cd2c1ceb14ad261353772966c55d6ef10b","subject":"Install libpython[23].x after python installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/497691cd2c1ceb14ad261353772966c55d6ef10b"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 21:14:15.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-27 18:48:42.000000000","tz":0},"subject":"Avoid storing cache in pip install for requirements image","message":"Avoid storing cache in pip install for requirements image\n\nIf additional pip packages are provided for requirements image, they\nwill be installed into system before building wheels and, in addition,\nwheel stored to local pip cache. Later, during wheels creation, this\ncached wheel will be used and then deleted as system will think that\npackage is available in wheel form.\n\nThis only matters when package is not available in wheel form and needs\nto be a part of requirements image.\n\nChange-Id: I6d5410da02ff9717418de55876162d34b6185e1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a7b46b1aa59ed65c421ce8d4872d9cc8906c13ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a7b46b1aa59ed65c421ce8d4872d9cc8906c13ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
