)]}'
{"id":"openstack%2Floci~703205","triplet_id":"openstack%2Floci~master~Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075","subject":"Move hardcoded packages for requirements to config","status":"ABANDONED","created":"2020-01-17 22:50:05.000000000","updated":"2020-07-24 04:44:18.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b1fb974df61053295d39a178b530748f0e9a6a63","_number":703205,"virtual_id_number":703205,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-05 19:56:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"date":"2020-02-05 18:56:16.000000000","_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"date":"2020-02-12 20:58:18.000000000","_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-18 18:10:28.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-01-22 07:41:18.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"state":"REVIEWER"},{"updated":"2020-02-05 15:16:12.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-02-05 19:56:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-12 20:58:18.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"}],"messages":[{"id":"831bdf47c30cbe619817f1a4d211cd5bad7f7ce5","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 22:50:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3c00358eef784d6d9a199a16c76d1efd9587a57","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 23:28:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"69693b64b33cc49d2ee113385faa9d2402689d9b","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 23:33:29.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"ffe4db44b4677212c44c537815364b70e25574f0","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 23:33:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b952e98f7ea790a1e8960bc69a494e6e71f2a96c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 23:46:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8131534f9b3c2fce0f1fb15caab82ea53a80d00a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 00:05:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/160560f0d40241ddad2dfb80d5a43939 : FAILURE in 3m 32s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e67ab321f619499882de826274a38e36 : FAILURE in 5m 49s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2992d1cf88d04544acb9b149a4bce707 : FAILURE in 3m 21s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f2ea207d243745dcaa682b66709695f0 : FAILURE in 3m 34s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9ec3e6e5a6354c2aacef5cc4f5ba3f27 : FAILURE in 4m 12s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8c516421ff8e471caa7aab5f16dc4a08 : FAILURE in 4m 33s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c0d48af675dc409eabc89d807470ea06 : FAILURE in 5m 25s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6c7fee70075f46fe930134da9e76cdea : FAILURE in 6m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8c907eac1d77477595dc5b30e49e4939 : FAILURE in 5m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/eea1688b1c694c5980750c0e112b7d0c : FAILURE in 3m 23s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cab95abc26dc4958b02e22eae2e2f961 : FAILURE in 3m 25s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/30543c1ea10b48f58f34764d0eb2b19a : FAILURE in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"40a2024f56494026235dbb94e2c0aac3ad9fec1a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 00:18:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9122e416c1853e0974b4b373d711ffeecaefab6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 00:25:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/923e7d3e5c4144ad9b0fd043845eea4d : FAILURE in 4m 11s\n- loci-glance https://zuul.opendev.org/t/openstack/build/90c508702ce241c2a16307a6570be956 : FAILURE in 5m 59s\n- loci-heat https://zuul.opendev.org/t/openstack/build/30e3f33aec584f40b60a1b75e33f64f3 : FAILURE in 5m 25s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a1a587d41dce43d5a6b27134c74c3355 : FAILURE in 4m 56s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b1845f1eddb44b19989246fe9a240b3b : FAILURE in 5m 17s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d7d70d42922f49d5831a9b2e34fe72d8 : FAILURE in 3m 57s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e2be1b42208143f3837eed8704d6a1bc : FAILURE in 3m 52s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/80c7a5a3f20041e3920e60fde0afbe04 : FAILURE in 4m 05s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7d31911882bf41708b8abd4e0a6ad77d : FAILURE in 4m 19s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/475431bb101a41c9a82eec8c6c236ab1 : FAILURE in 4m 24s\n- loci-placement https://zuul.opendev.org/t/openstack/build/09e07fd960cb4283b9e2bbc65c5cc53d : FAILURE in 5m 45s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/759e0a7906ba4ab9ad2f4e64d5b5d190 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":4},{"id":"33cede0faf5941d98265435938097723681602cc","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 00:30:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fb6ff0faf961cbdb0d481066cff3aac1741903b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 00:37:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4d10f32e9b7a4c6fae876413e29857d4 : FAILURE in 3m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1facc84a88804ef78e5b36da7228ad0a : FAILURE in 3m 03s\n- loci-heat https://zuul.opendev.org/t/openstack/build/53b392b3bd194c42a69c7c5f567fb207 : FAILURE in 3m 41s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/16357bd556604dbca5ebca96e0812b43 : FAILURE in 2m 38s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b4a07f74fcec4be380f28b055b51d6e2 : FAILURE in 4m 23s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/0447c64fa3cf4fb2b9c13a0fc666247b : FAILURE in 4m 19s\n- loci-manila https://zuul.opendev.org/t/openstack/build/28c51ad596584b13975aa5cb3c6d0cad : FAILURE in 4m 43s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4fb188a81c8740baaaa20ed8a83dda59 : FAILURE in 4m 18s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c56cc3dbb09842bf80093da8d4dc380f : FAILURE in 4m 34s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/52916a0ca2b64ac6b357d1be65266368 : FAILURE in 4m 41s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3db42406479c4c3a92d4a3f6e79e41ac : FAILURE in 4m 28s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e4c21af589044b2b88d3d915b34e6dee : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":5},{"id":"088e70876caae0625a1fa891a2861a7ccb41017c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 00:47:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7f6a2a89c3b24fba02fbca0b7a928c7d2e65ef22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 00:55:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4cded48e2a854c17940b9401a3f97abe : FAILURE in 3m 25s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8647924dc5974587bef5ce57ea768470 : FAILURE in 3m 00s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f6a622259a4b4ee58ee7780bf766c46d : FAILURE in 2m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/bda09e927349419090415b77f5c8e921 : FAILURE in 2m 26s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a11931b2b6534425ba9db40d6d192b20 : FAILURE in 2m 46s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a13f6f8d6ed74c0fb864054233034afe : FAILURE in 2m 40s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0cf3aced3c2940b68a2ae605c81c0f92 : FAILURE in 3m 01s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/201363000d9649c6bab04ceaf47842fc : FAILURE in 2m 40s\n- loci-nova https://zuul.opendev.org/t/openstack/build/dc28932927074546a8c271e487d63f7d : FAILURE in 3m 20s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/53f400a7ff8348a9ae11026e04fe9aba : FAILURE in 2m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/6e9602c9df5f4b22b20089aecb133fb1 : FAILURE in 2m 35s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7db799aa92e542eeaedc155594add6ee : FAILURE in 3m 08s","accounts_in_message":[],"_revision_number":6},{"id":"c9284db683aa800cdb3212fd1ebcbbfc70344940","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 01:02:30.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"672c18d69ef090416e5cb9618e74efe8bbe48ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 01:11:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ffbcea15f5ad40258356d190c3e2e924 : FAILURE in 3m 36s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2c1f27e46ded4b5ebb5fe8b9664f8426 : FAILURE in 2m 51s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e61a536aebe948538d0eeb0ffdd80ee4 : FAILURE in 2m 58s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ddbf8fff97174bc39a944f970c620bbc : FAILURE in 3m 22s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e6ac0af7a51b4ca6abdb034593537ac0 : FAILURE in 5m 15s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f1c647b1c63f4f6589ea7b85d398712d : FAILURE in 2m 45s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6884ee5f7c1346c1894cc5d6936dcb5f : FAILURE in 2m 53s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/1d5f3dfbc4d9469bba72dd1739cf4b28 : FAILURE in 2m 36s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c479f3cd623d46d68624aff695f44044 : FAILURE in 2m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/dfc8798d39c9491592751b516ae9dc5c : FAILURE in 4m 57s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e6729f2086474c66b1fdbe03c2cd956c : FAILURE in 4m 03s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/dc9b731ce2544450b92640ba77849178 : FAILURE in 3m 55s","accounts_in_message":[],"_revision_number":7},{"id":"93d3df33ac923e438582f2940df5760d519566ad","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 14:54:32.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"b14f490148b19c1baba695191dea95e0cf068c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 15:02:40.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/58e1b909553f4dd6a4ea9cc24b8b5866 : FAILURE in 3m 03s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9ceda54280234a44b8668128e12ad549 : FAILURE in 2m 52s\n- loci-heat https://zuul.opendev.org/t/openstack/build/daa29c9643384e2da0ec272ff29b53c2 : FAILURE in 3m 10s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/74df4136fcac48bea93cb8929d0b1cde : FAILURE in 2m 59s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a3e283cd9bd349aa9273b6b885f4b799 : FAILURE in 2m 50s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/0657e67cdafe43c7894d63a10ebe4ac2 : FAILURE in 2m 55s\n- loci-manila https://zuul.opendev.org/t/openstack/build/ae5e30fdd01145c298cfa4353fc4fabd : FAILURE in 2m 39s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7fac18e620454a3d9edf04940b64867d : FAILURE in 3m 00s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0a3f24ca88bb426cae670e677e1ef791 : FAILURE in 3m 05s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/6ed245919114467eac506b974ac7ed18 : FAILURE in 3m 00s\n- loci-placement https://zuul.opendev.org/t/openstack/build/fb58cf7b9f7347ed93a0182a20b5c5ba : FAILURE in 3m 10s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4146aad0eb624237b3ecfe80ea6a3e3d : FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":8},{"id":"561676c6ef32a62dc9088d048377911fce1c8ed0","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 15:23:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4b05eaf6637b57214092fc44e0ad3856127c6361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 15:39:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/8a9c5db3a6894c21b7889d64709aa060 : FAILURE in 8m 31s\n- loci-glance https://zuul.opendev.org/t/openstack/build/66fdbda0c2524969a901af700983a89e : FAILURE in 9m 24s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0262faaa61154770b0b5ad58adcea673 : FAILURE in 10m 23s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6c73f1e5d02140658572a7c14f482941 : FAILURE in 9m 15s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0089cd74c7b34836826f3f2dcf5fde09 : FAILURE in 9m 24s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c2d36b1d20a940a68142eda411838d44 : FAILURE in 9m 40s\n- loci-manila https://zuul.opendev.org/t/openstack/build/dfb2a202049746959c9f569f2ec0f41d : FAILURE in 9m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6b9b5181ee9b444286f5ba01d7125ad4 : FAILURE in 9m 49s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bedfa137b44c472e98a9b240e6a4fbf9 : FAILURE in 10m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/df01cb646d3045d08d5beb0b3d16b056 : FAILURE in 9m 58s\n- loci-placement https://zuul.opendev.org/t/openstack/build/bb525106c753409399a97cc6b7d9b08f : FAILURE in 11m 09s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0938fd45fca94a078dcd7cbd06f34080 : FAILURE in 9m 11s","accounts_in_message":[],"_revision_number":9},{"id":"a43966142b0794cc05916e04195077363ade2fe7","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 15:55:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cea45e63b910f4def4817ad4a2ab2faf8c083c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 16:58:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/70f92f5a084848bea3be662ebdd548d5 : TIMED_OUT in 1h 00m 47s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8b60537404a74fd5af8c769969168bee : FAILURE in 44m 42s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c986059915fa43f5add94bb04794aaac : FAILURE in 46m 01s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/89ba71d56d0047908a98ffdd06360b23 : TIMED_OUT in 1h 00m 57s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/68a3529abaf74e5bb37d0ec0aa8fdd9d : TIMED_OUT in 1h 00m 53s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/346273e720b04f55bef84eaa014f2244 : TIMED_OUT in 1h 00m 55s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0b7e46f15d104bdf9d1b2cb8ae96c632 : TIMED_OUT in 1h 00m 49s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/02a91719a76747f9ba1e9821de82ba67 : TIMED_OUT in 1h 00m 51s\n- loci-nova https://zuul.opendev.org/t/openstack/build/eb419325cb7344d1a83c96c3d20d72f1 : TIMED_OUT in 1h 00m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3eaea8ccfe9446f7ba15121987a3e109 : TIMED_OUT in 1h 01m 14s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ea85e9b9f66a45ea9ddb5f4bfc9c1e9e : TIMED_OUT in 1h 00m 50s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/be716d3ac6f240cd9156af90e299f28e : TIMED_OUT in 40m 53s","accounts_in_message":[],"_revision_number":10},{"id":"ffd01e341ab896a6c00e1e2d826f27bbd0c9146b","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 17:12:47.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"982654e4e4dd3dda312688c73556d75d4db0cdf9","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 17:16:31.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ec80ae6c4dd82879573495bb716083b1158c920c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 18:11:44.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d15017e95af940be8b6046300d371f7b : SUCCESS in 52m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1ef749fb016c45c793eafae30663cd1a : SUCCESS in 43m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3095394308294b3fb8d63d6fdf02c979 : SUCCESS in 51m 40s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c6a1ec1baecb4cf386348d7613bfd788 : SUCCESS in 50m 09s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ebf53244e9d047ba88d1a1f9e6c6ef42 : SUCCESS in 37m 03s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/724813a87b8e4f32bbc68abb38e3bbdf : SUCCESS in 44m 44s\n- loci-manila https://zuul.opendev.org/t/openstack/build/bb08b3864dcb436e983d4847ae8d44ef : SUCCESS in 37m 00s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cc216d08450d4932b6efdcaeae22f132 : SUCCESS in 48m 49s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3cd2ecdc3a7b447f80739e083bef5e23 : SUCCESS in 40m 53s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/116de95e2fa6459483156544956dde17 : SUCCESS in 44m 20s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5b7f04cc51f64a5dba2cf58b08014f20 : FAILURE in 36m 33s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/3c3a9139fb754652bb69930413a77e7e : SUCCESS in 34m 43s","accounts_in_message":[],"_revision_number":12},{"id":"99fa6fa2d8f4887707cd56012b14308fbe917118","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 15:59:18.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"b5802f228f47b4a7f85efaa0de46c3d69ff8a973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 16:59:12.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/93eef3f797924c4198b1b6cd6098c72f : FAILURE in 42m 51s\n- loci-glance https://zuul.opendev.org/t/openstack/build/60ef7b62607b40b99d937eed7461b00f : FAILURE in 26m 09s\n- loci-heat https://zuul.opendev.org/t/openstack/build/16499c04b7a541f1959763159cbb5205 : FAILURE in 21m 39s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/bd6a59e44f0940a4a53b401d0ff0a71c : FAILURE in 42m 56s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1e6f9b7f638d499eb8244e4784b89af5 : FAILURE in 40m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a2e259fec5214b67a9955c12d88f43d6 : FAILURE in 37m 15s\n- loci-manila https://zuul.opendev.org/t/openstack/build/d20d3f80159a4f388ab21f456818630c : FAILURE in 43m 59s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/23979f003e1e43eb9c902785f40270af : FAILURE in 30m 19s\n- loci-nova https://zuul.opendev.org/t/openstack/build/14d2a5c26e6f45ba80c30811112b8ccd : FAILURE in 37m 16s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/74b4515b1d2645e4b7137f1b7f476ff4 : FAILURE in 24m 47s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b5851e1dbdc247be816caa0e8389fd2a : FAILURE in 38m 21s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/95f0b297dc144215b0b7bbd771e102ac : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":13},{"id":"547aa77e799ce489d9210e7eab44735837199642","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 18:49:19.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"e18b25c5e042db9a955629d2662c46b88d53967d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 19:48:19.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9fc3f7cc75eb48349f3fbb2721df19da : SUCCESS in 37m 20s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0a8480bac5344b98b15e24ebb1267dcc : SUCCESS in 35m 11s\n- loci-heat https://zuul.opendev.org/t/openstack/build/b366bbcba774483bae71b7e8d460a60b : SUCCESS in 37m 51s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/11a7ffa677f045b9a93e03af21e49d2b : SUCCESS in 39m 14s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f8f11fd4923540bdbe8d43025dda69d9 : SUCCESS in 36m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/799eecfdbbcf4b24b160a25fadaca8bc : SUCCESS in 36m 59s\n- loci-manila https://zuul.opendev.org/t/openstack/build/58fbe33f2cdd4e2bab6d5614259a6680 : SUCCESS in 42m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4b9ea1875e204a60bf90e58d042f1965 : SUCCESS in 45m 00s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ca9ad52c13864990ba97dc4a5b02b3be : SUCCESS in 44m 48s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5150ed6035b74277880bf4b77b66d270 : SUCCESS in 50m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5d513707683a45e0ac635c888f6acdb5 : SUCCESS in 38m 08s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d1417747daaf4a81a4336add8c414be4 : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":14},{"id":"f0ac17866694e0b3a4ab6acaaa8dec2058face0e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-05 15:16:12.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"98e2e2d0bec9f2f091c6d4e581e5c028870c3f82","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-05 17:30:04.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"cf7be358413dc38d690c68ad81d9758d8bbced11","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-05 17:32:05.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"bd02abf70c9aca4e794812878cc40caf608587a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 18:42:56.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4c9be27c5ebb4774b748541e046cfb50 : FAILURE in 35m 59s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0530b77ec31349c3a40f3f75932ab13f : SUCCESS in 41m 13s\n- loci-heat https://zuul.opendev.org/t/openstack/build/4de6bd8afb8c43eeaa5ec975d1c9b71e : SUCCESS in 45m 38s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/fd251e7d132d4e3e948b9dff4872fe0c : SUCCESS in 56m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b57cc17a0a08475ba9943b035be3ee76 : SUCCESS in 44m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/bba69401947b456e8c3c73b129660bbc : SUCCESS in 41m 38s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6e40c686c30d4dcaa22d108ac2a189ee : TIMED_OUT in 1h 01m 13s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/519b5d3cff074affbd003d22b956258f : SUCCESS in 45m 18s\n- loci-nova https://zuul.opendev.org/t/openstack/build/361d494a5d42481694cafb0aafc44030 : SUCCESS in 52m 31s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/1f9ce0d827ce47cf855a9a5dd156b0ef : SUCCESS in 40m 00s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e002cc6690b046c1bd37f281cfc52bcf : SUCCESS in 32m 53s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4108ebdbc0c14ecc992b77927381d19b : SUCCESS in 28m 01s","accounts_in_message":[],"_revision_number":15},{"id":"8701fc4d0ae312622dea76e1ba360cf673639230","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-05 18:56:16.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"e01ed79faa44224ca361322b9c2020e221ca80e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 19:56:54.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/119ae999d1bd41898450ca6fa3c4f2c3 : SUCCESS in 52m 01s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bc17ee8a68cc4b5381ff52642e996e80 : SUCCESS in 37m 47s\n- loci-heat https://zuul.opendev.org/t/openstack/build/78705f42450d4515b5cc4ae8ca4b940d : SUCCESS in 54m 35s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9bc8661cd9024fc791e35f23f308efae : SUCCESS in 52m 42s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c8f87fd950fc4afb86f06590f7d6c687 : SUCCESS in 38m 04s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b6bfd354895a4e4192b95f30d50cdeda : SUCCESS in 45m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/d89cccbfb6aa476699586414d169af3e : SUCCESS in 39m 30s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/80ae584519a7418da82632d613608b81 : SUCCESS in 40m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/74a242ce16324d279e965902d3c37452 : SUCCESS in 40m 33s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3913ca6f0d3e448ca0bdef54e0a98d8d : SUCCESS in 38m 22s\n- loci-placement https://zuul.opendev.org/t/openstack/build/72a88a62d90f411b8e11b6fa1f61ab70 : SUCCESS in 37m 15s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/21fdeea019834b0ca454e8fc3602c2c7 : SUCCESS in 32m 03s","accounts_in_message":[],"_revision_number":15},{"id":"b1fb974df61053295d39a178b530748f0e9a6a63","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-07-24 04:44:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"449d6ecd5f91405d506de2669f7a5a6969754b86","revisions":{"c23aba77fd0d9e8856b71f8c4ee6fb2b2c944c1f":{"kind":"REWORK","_number":1,"created":"2020-01-17 22:50:05.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/1"}}},"commit":{"parents":[{"commit":"f8062f45597a52b639849a128e6773ff6712bb61","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:49:41.000000000","tz":-360},"subject":"Add purestorage installation for requirements","message":"Add purestorage installation for requirements\n\nPurestorage is not available in wheel form from PyPi.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c23aba77fd0d9e8856b71f8c4ee6fb2b2c944c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c23aba77fd0d9e8856b71f8c4ee6fb2b2c944c1f"}]},"branch":"refs/heads/master"},"b60a3f9d949bb2a6cc241c188e30245c35470742":{"kind":"REWORK","_number":2,"created":"2020-01-17 23:33:30.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/2"}}},"commit":{"parents":[{"commit":"f8062f45597a52b639849a128e6773ff6712bb61","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 23:32:50.000000000","tz":-360},"subject":"Add purestorage installation for requirements","message":"Add purestorage installation for requirements\n\nPurestorage is not available in wheel form from PyPi.\nAlso adds required code to process pydep for requirements image build.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b60a3f9d949bb2a6cc241c188e30245c35470742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b60a3f9d949bb2a6cc241c188e30245c35470742"}]},"branch":"refs/heads/master"},"fef73d79c632c0b0ccd9bec0b60c6867decdb7be":{"kind":"REWORK","_number":3,"created":"2020-01-17 23:46:31.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/3"}}},"commit":{"parents":[{"commit":"f8062f45597a52b639849a128e6773ff6712bb61","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 23:39:32.000000000","tz":-360},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\nAlso adds purestorage to pydep for requirement image as it is not\navailable in wheel form from PyPi.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fef73d79c632c0b0ccd9bec0b60c6867decdb7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/fef73d79c632c0b0ccd9bec0b60c6867decdb7be"}]},"branch":"refs/heads/master"},"9d0e03b3c72997d827437808acebd3b9578a344c":{"kind":"REWORK","_number":4,"created":"2020-01-18 00:18:09.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/4"}}},"commit":{"parents":[{"commit":"f8062f45597a52b639849a128e6773ff6712bb61","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 00:18:01.000000000","tz":-360},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\nAlso adds purestorage to pydep for requirement image as it is not\navailable in wheel form from PyPi.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9d0e03b3c72997d827437808acebd3b9578a344c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9d0e03b3c72997d827437808acebd3b9578a344c"}]},"branch":"refs/heads/master"},"59362c24b00912d6f1dc9962f99366bd28ff4517":{"kind":"REWORK","_number":5,"created":"2020-01-18 00:30:42.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/5"}}},"commit":{"parents":[{"commit":"f8062f45597a52b639849a128e6773ff6712bb61","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 00:30:39.000000000","tz":-360},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\nAlso adds purestorage to pydep for requirement image as it is not\navailable in wheel form from PyPi.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/59362c24b00912d6f1dc9962f99366bd28ff4517"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/59362c24b00912d6f1dc9962f99366bd28ff4517"}]},"branch":"refs/heads/master"},"f1d6f1b94c0e8e5fc4cd6f33f8dfb2b7cb973cec":{"kind":"REWORK","_number":6,"created":"2020-01-18 00:47:22.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/6"}}},"commit":{"parents":[{"commit":"f8062f45597a52b639849a128e6773ff6712bb61","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 00:47:18.000000000","tz":-360},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\nAlso adds purestorage to pydep for requirement image as it is not\navailable in wheel form from PyPi.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f1d6f1b94c0e8e5fc4cd6f33f8dfb2b7cb973cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f1d6f1b94c0e8e5fc4cd6f33f8dfb2b7cb973cec"}]},"branch":"refs/heads/master"},"6c58b06e3d4cf6db17e4b343c6d884338cbe10c8":{"kind":"REWORK","_number":7,"created":"2020-01-18 01:02:30.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/7"}}},"commit":{"parents":[{"commit":"f8062f45597a52b639849a128e6773ff6712bb61","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 01:00:30.000000000","tz":-360},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6c58b06e3d4cf6db17e4b343c6d884338cbe10c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6c58b06e3d4cf6db17e4b343c6d884338cbe10c8"}]},"branch":"refs/heads/master"},"2d86f67859ef1ad3ebe3097d48da942b4f8db222":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-01-18 14:54:32.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/8"}}},"commit":{"parents":[{"commit":"dfcf5ced1933758603641480bc4075759ba99692","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfcf5ced1933758603641480bc4075759ba99692"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 14:54:32.000000000","tz":0},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2d86f67859ef1ad3ebe3097d48da942b4f8db222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2d86f67859ef1ad3ebe3097d48da942b4f8db222"}]},"branch":"refs/heads/master"},"84b28d00f7b74308f071392815574dbfffade681":{"kind":"REWORK","_number":9,"created":"2020-01-18 15:23:35.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/9"}}},"commit":{"parents":[{"commit":"dfcf5ced1933758603641480bc4075759ba99692","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfcf5ced1933758603641480bc4075759ba99692"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 15:23:17.000000000","tz":-360},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/84b28d00f7b74308f071392815574dbfffade681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/84b28d00f7b74308f071392815574dbfffade681"}]},"branch":"refs/heads/master"},"214da4b2f9d1c7983020ccb0c8d57cb02c072e67":{"kind":"REWORK","_number":10,"created":"2020-01-18 15:55:45.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/10"}}},"commit":{"parents":[{"commit":"dfcf5ced1933758603641480bc4075759ba99692","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfcf5ced1933758603641480bc4075759ba99692"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 15:55:40.000000000","tz":-360},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/214da4b2f9d1c7983020ccb0c8d57cb02c072e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/214da4b2f9d1c7983020ccb0c8d57cb02c072e67"}]},"branch":"refs/heads/master"},"85aa7bbfc10899184d2e4566eefca9d85e589863":{"kind":"REWORK","_number":11,"created":"2020-01-18 17:12:47.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/11"}}},"commit":{"parents":[{"commit":"dfcf5ced1933758603641480bc4075759ba99692","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfcf5ced1933758603641480bc4075759ba99692"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 17:12:43.000000000","tz":-360},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/85aa7bbfc10899184d2e4566eefca9d85e589863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/85aa7bbfc10899184d2e4566eefca9d85e589863"}]},"branch":"refs/heads/master"},"5c9c35ff8d1e40dab4a5f0f9b19afa32f64ca361":{"kind":"REWORK","_number":12,"created":"2020-01-18 17:16:31.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/12"}}},"commit":{"parents":[{"commit":"dfcf5ced1933758603641480bc4075759ba99692","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfcf5ced1933758603641480bc4075759ba99692"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-18 17:15:07.000000000","tz":-360},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\n\nDefault PROFILES in Dockerfile is defined to deal with set -u in\nrequirements.sh.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c9c35ff8d1e40dab4a5f0f9b19afa32f64ca361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c9c35ff8d1e40dab4a5f0f9b19afa32f64ca361"}]},"branch":"refs/heads/master"},"00be04d8a8a7e1756d81faccefe42e0921b9a045":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2020-01-20 15:59:18.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/13"}}},"commit":{"parents":[{"commit":"cfe5579a294ab572eb9e2ed0edf80295524e3b89","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cfe5579a294ab572eb9e2ed0edf80295524e3b89"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 15:59:18.000000000","tz":0},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\n\nDefault PROFILES in Dockerfile is defined to deal with set -u in\nrequirements.sh.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/00be04d8a8a7e1756d81faccefe42e0921b9a045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/00be04d8a8a7e1756d81faccefe42e0921b9a045"}]},"branch":"refs/heads/master"},"d4d92cfe6189991871d3dc0c67da71f53f79c561":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2020-01-27 18:49:19.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/14"}}},"commit":{"parents":[{"commit":"a7b46b1aa59ed65c421ce8d4872d9cc8906c13ec","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a7b46b1aa59ed65c421ce8d4872d9cc8906c13ec"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-27 18:49:19.000000000","tz":0},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\n\nDefault PROFILES in Dockerfile is defined to deal with set -u in\nrequirements.sh.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d4d92cfe6189991871d3dc0c67da71f53f79c561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d4d92cfe6189991871d3dc0c67da71f53f79c561"}]},"branch":"refs/heads/master"},"449d6ecd5f91405d506de2669f7a5a6969754b86":{"kind":"REWORK","_number":15,"created":"2020-02-05 17:32:05.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/05/703205/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/05/703205/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/05/703205/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/05/703205/15"}}},"commit":{"parents":[{"commit":"a7b46b1aa59ed65c421ce8d4872d9cc8906c13ec","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a7b46b1aa59ed65c421ce8d4872d9cc8906c13ec"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 22:47:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-05 17:30:35.000000000","tz":0},"subject":"Move hardcoded packages for requirements to config","message":"Move hardcoded packages for requirements to config\n\nMoves packages to pydep and adds related code to process pydep\nconfiguration for requirements image build.\n\nDefault PROFILES in Dockerfile is defined to deal with set -u in\nrequirements.sh.\n\nChange-Id: Ia2ff838b523fdfd8eb2e2d3ae51780d564d9a075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/449d6ecd5f91405d506de2669f7a5a6969754b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/449d6ecd5f91405d506de2669f7a5a6969754b86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
