)]}'
{"id":"openstack%2Floci~703209","triplet_id":"openstack%2Floci~master~I6746236f5ad743f62b22910686831409d7b724a1","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I6746236f5ad743f62b22910686831409d7b724a1","subject":"Add purestorage to requirements image","status":"ABANDONED","created":"2020-01-17 23:48:21.000000000","updated":"2020-01-18 00:33:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7fbafbc9eb0b1b826701d5994cb3093fa8479a15","_number":703209,"virtual_id_number":703209,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-18 00:32:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-18 00:32:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a404d35ac1e3c087e35b2c2c4c7665c859dc9e46","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-17 23:48:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f99529e7e77acf753a2473277ae6df43c068e365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 00:32:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/dfe57d49998e49e89f9e63a79feedcf7 : FAILURE in 26m 58s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b763b4ef658d4ec8a2c2907e751b1900 : FAILURE in 26m 19s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9892cd395bce4a5783d9560df6336846 : FAILURE in 31m 07s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/22305262a0dc4900b65128def02604d6 : FAILURE in 30m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3c430b03d8f640b082c74c92f2995dd0 : FAILURE in 30m 33s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ceeb5066eb954d5fa7067a7bb0c0980f : FAILURE in 37m 44s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9b42907ccb6f4e799fceda41889864c7 : FAILURE in 37m 17s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7227bc06cddd4f4398ed81f188db9aeb : FAILURE in 30m 51s\n- loci-nova https://zuul.opendev.org/t/openstack/build/945fc818615b4cc3bb95614952680756 : FAILURE in 30m 02s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c07394c5aaba4ac193afd79a99cd1302 : FAILURE in 31m 57s\n- loci-placement https://zuul.opendev.org/t/openstack/build/820b182788f547c39e53528e662c0ed2 : FAILURE in 30m 36s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/af151da9e0fb4daf9783b6fddc17bfce : FAILURE in 31m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7fbafbc9eb0b1b826701d5994cb3093fa8479a15","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-18 00:33:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fdef1df1057c7fdb7ed654b11496bd87c31698a","revisions":{"4fdef1df1057c7fdb7ed654b11496bd87c31698a":{"kind":"REWORK","_number":1,"created":"2020-01-17 23:48:21.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/09/703209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/09/703209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/703209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/703209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/703209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/09/703209/1"}}},"commit":{"parents":[{"commit":"f8062f45597a52b639849a128e6773ff6712bb61","subject":"Avoid storing cache in pip install for requirements image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f8062f45597a52b639849a128e6773ff6712bb61"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 23:47:28.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-17 23:47:28.000000000","tz":-360},"subject":"Add purestorage to requirements image","message":"Add purestorage to requirements image\n\nChange-Id: I6746236f5ad743f62b22910686831409d7b724a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4fdef1df1057c7fdb7ed654b11496bd87c31698a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4fdef1df1057c7fdb7ed654b11496bd87c31698a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
