)]}'
{"id":"openstack%2Floci~703498","triplet_id":"openstack%2Floci~master~Ib40190427c69dad7c62a402d4c663eb464d3c4f5","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"Ib40190427c69dad7c62a402d4c663eb464d3c4f5","subject":"Enable leap15 build disabled due to repo issues","status":"ABANDONED","created":"2020-01-20 20:27:30.000000000","updated":"2020-01-21 01:38:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7d44259f18fd11a0a88be69d3de6d519df689668","_number":703498,"virtual_id_number":703498,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-20 22:18:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-01-21 00:58:43.000000000","_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-20 22:00:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a0a1cf389dd4c4cb696b66faa7de92eccbdfe99","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 20:27:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6fba5060ad0ebcd103f252ec3ad598f6a602dc2","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 20:35:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"496f3e4aee04ab8bcd7667b4e9c34396804bba10","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 20:40:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5503f2baf78bf336355626742e30c8524e214bf0","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 20:46:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6948a24a818b4b393b4d5ca11476d2780fd4b023","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 20:55:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c4f51f82856ae3b15a578d7706f4201d394ce1d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 20:55:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5bf0501fe55e48329e5ddd3198ae5c33 : SUCCESS in 5m 48s\n- loci-glance https://zuul.opendev.org/t/openstack/build/15b1cd10ae244ff49b159226077212aa : SUCCESS in 4m 58s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0000d848df8740caabee3eb3510e829d : SUCCESS in 2m 59s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/5b3fecae18ec4921a737485ecb53b955 : SUCCESS in 4m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/81d8d9bf5e41479ba57d44dd0c71eefa : SUCCESS in 4m 34s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/fa544bd4f385479fb08ee198abb996b2 : SUCCESS in 3m 31s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a872b63b671d4eaea4848ff94fb09981 : SUCCESS in 3m 03s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/df10308e62f740c79d04a793a0c6f1d2 : SUCCESS in 5m 31s\n- loci-nova https://zuul.opendev.org/t/openstack/build/cba46dc67db942f3b6c56cde0940f367 : SUCCESS in 4m 23s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/fb5d0156b9fd431dbf2540529ef24cb7 : SUCCESS in 5m 12s\n- loci-placement https://zuul.opendev.org/t/openstack/build/62c90f198dd6457fb802a8ad512b978a : SUCCESS in 4m 10s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1df310776ef94b93aca2e32917310966 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":4},{"id":"c6278aed7f913c0839c2de88ed769b800222f7f7","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 21:14:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5},{"id":"dc90d3da2744a32cd42e48a46b4886a1e1ef6192","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 21:43:54.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"d3753d7a8ac4d08a5951e0c724e967ff198498cf","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 21:47:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6a0a40c7062a6ece3b7d7f3428fb9bd314054f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 22:00:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7047013126d4433aa250b29ae8e70c8f : FAILURE in 5m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/43eb435ae21c4817b3f7c531112db4bc : FAILURE in 5m 45s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5ed59b3363894df286629b10f0a852d7 : FAILURE in 4m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/acea7c0192ff4fbd963d00c052cba361 : FAILURE in 10m 07s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/cf22396f74c94192b0f1f462f88ca951 : FAILURE in 7m 41s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/eea465a602254051957987e126d6baf0 : FAILURE in 6m 33s\n- loci-manila https://zuul.opendev.org/t/openstack/build/98d67275b9a84c98aa6edc4f1af9fc2b : FAILURE in 9m 24s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/98f454a926904223b009ecb770c19e18 : FAILURE in 8m 21s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b133d43fb9164179a4582c1f1e0255af : FAILURE in 6m 14s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a977368c59804d60a4aacda804dc68e6 : FAILURE in 9m 36s\n- loci-placement https://zuul.opendev.org/t/openstack/build/32041b5afdf8460e8fda78a020ce92ef : FAILURE in 5m 50s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7cd028c7b1714eecbb0b906653baaf55 : FAILURE in 8m 03s","accounts_in_message":[],"_revision_number":6},{"id":"aa72c67534951d709848410ec6bd1553e6f76e04","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-20 22:18:02.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"7e5e36b6ad6550544717e3e4783e735ba644ec89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 22:26:26.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/605377f1761d46cfa4f309b90c67c755 : FAILURE in 6m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8032243279f94fdeaa8c64be7f76331e : FAILURE in 5m 24s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8f7e3b9a735c4d0e8812a13a6b69b766 : FAILURE in 5m 23s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/71501fa64144462192922dc6681f9d3c : FAILURE in 5m 17s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/51a84049b08c4406879cd1cf6412afb8 : FAILURE in 5m 04s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1071b00d71af45efa48c89035e41c234 : FAILURE in 4m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a3e0f74907a34b41a5159ce55c26e40c : FAILURE in 5m 00s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6620e98c22ff49f8a9cc5bc723575846 : FAILURE in 5m 57s\n- loci-nova https://zuul.opendev.org/t/openstack/build/9d583911ace2413f95fce2be1d76373b : FAILURE in 6m 22s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2e2409267f07464fb96cc96e14b04018 : FAILURE in 5m 18s\n- loci-placement https://zuul.opendev.org/t/openstack/build/28356e915abd4a6091707ec2d2f05ea4 : FAILURE in 5m 54s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/94b68a4a1eaa4b1abb2a8b4f7694b9b6 : FAILURE in 5m 45s","accounts_in_message":[],"_revision_number":7},{"id":"dcca0c816ed462efbcbf177835b12281eddceb6e","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 00:58:43.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7d44259f18fd11a0a88be69d3de6d519df689668","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 01:38:29.000000000","message":"Abandoned\n\nno longer needed. mirrors are ok now","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0c11c1819f2ba405e1c58df6ab8da30bc79dcbb1","revisions":{"65ef67e929c42feeb99f61525b804ca79b4aea1f":{"kind":"REWORK","_number":1,"created":"2020-01-20 20:27:30.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/98/703498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/98/703498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/98/703498/1"}}},"commit":{"parents":[{"commit":"bcd119e10d865252fd0de96e7c0b810c5f8e1d15","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bcd119e10d865252fd0de96e7c0b810c5f8e1d15"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 19:38:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 20:27:18.000000000","tz":-360},"subject":"test","message":"test\n\nChange-Id: Ib40190427c69dad7c62a402d4c663eb464d3c4f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/65ef67e929c42feeb99f61525b804ca79b4aea1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/65ef67e929c42feeb99f61525b804ca79b4aea1f"}]},"branch":"refs/heads/master"},"ccc4696554c6b83a60ee1a6b0256497ddf8a2016":{"kind":"REWORK","_number":2,"created":"2020-01-20 20:35:54.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/98/703498/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/98/703498/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/98/703498/2"}}},"commit":{"parents":[{"commit":"bcd119e10d865252fd0de96e7c0b810c5f8e1d15","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bcd119e10d865252fd0de96e7c0b810c5f8e1d15"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 19:38:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 20:35:51.000000000","tz":-360},"subject":"test","message":"test\n\nChange-Id: Ib40190427c69dad7c62a402d4c663eb464d3c4f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ccc4696554c6b83a60ee1a6b0256497ddf8a2016"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ccc4696554c6b83a60ee1a6b0256497ddf8a2016"}]},"branch":"refs/heads/master"},"6f5bec1d61ee5c8fbb756d0166c1d0e12ef219d9":{"kind":"REWORK","_number":3,"created":"2020-01-20 20:40:40.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/98/703498/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/98/703498/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/98/703498/3"}}},"commit":{"parents":[{"commit":"bcd119e10d865252fd0de96e7c0b810c5f8e1d15","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bcd119e10d865252fd0de96e7c0b810c5f8e1d15"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 19:38:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 20:40:37.000000000","tz":-360},"subject":"test","message":"test\n\nChange-Id: Ib40190427c69dad7c62a402d4c663eb464d3c4f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6f5bec1d61ee5c8fbb756d0166c1d0e12ef219d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6f5bec1d61ee5c8fbb756d0166c1d0e12ef219d9"}]},"branch":"refs/heads/master"},"1e26dc4f4707eaf19637fa9b856b62de353a3591":{"kind":"REWORK","_number":4,"created":"2020-01-20 20:46:56.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/98/703498/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/98/703498/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/98/703498/4"}}},"commit":{"parents":[{"commit":"bcd119e10d865252fd0de96e7c0b810c5f8e1d15","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bcd119e10d865252fd0de96e7c0b810c5f8e1d15"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 19:38:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 20:46:53.000000000","tz":-360},"subject":"test","message":"test\n\nChange-Id: Ib40190427c69dad7c62a402d4c663eb464d3c4f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1e26dc4f4707eaf19637fa9b856b62de353a3591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1e26dc4f4707eaf19637fa9b856b62de353a3591"}]},"branch":"refs/heads/master"},"1ede319599129cbbc753f2dda44c52a6dfe9b44e":{"kind":"REWORK","_number":5,"created":"2020-01-20 20:55:44.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/98/703498/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/98/703498/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/98/703498/5"}}},"commit":{"parents":[{"commit":"bcd119e10d865252fd0de96e7c0b810c5f8e1d15","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bcd119e10d865252fd0de96e7c0b810c5f8e1d15"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 19:38:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 20:55:41.000000000","tz":-360},"subject":"test","message":"test\n\nChange-Id: Ib40190427c69dad7c62a402d4c663eb464d3c4f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1ede319599129cbbc753f2dda44c52a6dfe9b44e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1ede319599129cbbc753f2dda44c52a6dfe9b44e"}]},"branch":"refs/heads/master"},"06c60f692b93ed1331290e05e563aeffe31f9168":{"kind":"REWORK","_number":6,"created":"2020-01-20 21:47:23.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/98/703498/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/98/703498/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/98/703498/6"}}},"commit":{"parents":[{"commit":"9b19db4394be22494341b427c9e20ea08dd8efd8","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9b19db4394be22494341b427c9e20ea08dd8efd8"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 19:38:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 21:46:31.000000000","tz":-360},"subject":"Enable leap15 build disabled due to repo issues","message":"Enable leap15 build disabled due to repo issues\n\nChange-Id: Ib40190427c69dad7c62a402d4c663eb464d3c4f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/06c60f692b93ed1331290e05e563aeffe31f9168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/06c60f692b93ed1331290e05e563aeffe31f9168"}]},"branch":"refs/heads/master"},"0c11c1819f2ba405e1c58df6ab8da30bc79dcbb1":{"kind":"NO_CHANGE","_number":7,"created":"2020-01-20 22:18:02.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/98/703498/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/98/703498/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/98/703498/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/98/703498/7"}}},"commit":{"parents":[{"commit":"ba116c2120c496ef3333a6ad5b741566cdec6979","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ba116c2120c496ef3333a6ad5b741566cdec6979"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 19:38:59.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-20 22:18:02.000000000","tz":0},"subject":"Enable leap15 build disabled due to repo issues","message":"Enable leap15 build disabled due to repo issues\n\nChange-Id: Ib40190427c69dad7c62a402d4c663eb464d3c4f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0c11c1819f2ba405e1c58df6ab8da30bc79dcbb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0c11c1819f2ba405e1c58df6ab8da30bc79dcbb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
