)]}'
{"id":"openstack%2Floci~703647","triplet_id":"openstack%2Floci~master~If602d4ccdaa66eed98dc8e9c600137a2f6fbb828","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"If602d4ccdaa66eed98dc8e9c600137a2f6fbb828","subject":"Remove an entry for building project from uc","status":"ABANDONED","created":"2020-01-21 15:50:17.000000000","updated":"2020-01-30 14:35:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"06453c6c1a1365945448579bc6518159dc40bd1a","_number":703647,"virtual_id_number":703647,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-21 19:38:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"disliked":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"all":[{"value":1,"date":"2020-01-27 19:35:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":-1,"date":"2020-01-28 16:18:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"value":1,"date":"2020-01-21 19:55:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"date":"2020-01-22 07:40:36.000000000","_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 16:55:29.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-01-21 16:56:03.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-01-21 19:38:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-21 19:55:28.000000000","updated_by":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"reviewer":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"state":"REVIEWER"},{"updated":"2020-01-22 07:40:36.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"state":"REVIEWER"},{"updated":"2020-01-27 19:35:48.000000000","updated_by":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"reviewer":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"state":"REVIEWER"}],"messages":[{"id":"c2a6f5a3856c44f267da4eaabeaac4a8581ffa7a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 15:50:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"124df1efe701f6c147b97f09cde2b4c35cfa6586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 16:55:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/fde35c91dd4e404dac92ce436d622908 : SUCCESS in 47m 03s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a2d8603f072844c18e1dd18f817c0c52 : SUCCESS in 44m 57s\n- loci-heat https://zuul.opendev.org/t/openstack/build/553f16c3ab7a412db107774a0abc3da5 : SUCCESS in 56m 27s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/7c31570122594e89827e1025e2eed6e7 : SUCCESS in 50m 20s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/af41e5c6de9f4a51b44738fd42449b51 : SUCCESS in 41m 46s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/035992fc6e6547ad8c610d848481bbe4 : SUCCESS in 45m 01s\n- loci-manila https://zuul.opendev.org/t/openstack/build/fd8d20de792e4cb5a0076325f5bdd497 : SUCCESS in 48m 41s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7dbf3174ae17444f9f27ccefb0351ea5 : SUCCESS in 51m 11s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3a21bda81acd4f3db75a9cec50ff92bd : SUCCESS in 53m 37s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2b5f77ec41d84f81b97858e6d695b451 : SUCCESS in 46m 59s\n- loci-placement https://zuul.opendev.org/t/openstack/build/22c0b6ecf95341bc9841880e95ebd58b : SUCCESS in 34m 52s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e559a75a03b54a7c90be8132c2742a21 : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":1},{"id":"57e431760c2f0c8fadd0b27f41663d5bc2ea10a7","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2020-01-21 17:06:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ce5305bfdc31d19e8eebd7744ebd6481aad181c9","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 17:08:10.000000000","message":"Patch Set 1:\n\ni\u0027m sorry. that\u0027s not right","accounts_in_message":[],"_revision_number":1},{"id":"77217ef1f0a78ef65260ce8dabae660015db3db1","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 17:08:19.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f086f3196186a918326195aa26078e0d7b2475cc","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 17:17:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6f2c245f162cadbaa003e949399cd0b3a523f93","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 17:19:31.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2d142dcdda1c5d8af048283b6f748d9e5c8aaf83","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 17:22:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"58da8caf2f54e59c0cc7b1b3977fb62f5c1ba3e0","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 17:23:41.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e4e9ad51b1928516ff534ef616898afc3a2fe514","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 17:31:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f1f0f0818a64138e028c249d6c75700eaa4c2515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 18:34:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/97ae308ff8a54368a56e62e1b5efb9e1 : SUCCESS in 44m 47s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e3ae02d767764263ac0e42a4031efacc : SUCCESS in 43m 47s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5e49dbd88219458e939b1d373a245e2d : FAILURE in 46m 01s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/08515f052b174f71b6d7837621330b7d : SUCCESS in 51m 44s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/983bdcc9e11f486cb20042bc22babd89 : SUCCESS in 50m 34s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/cd274c360266498bb90d7fbd5c9e96c4 : SUCCESS in 46m 45s\n- loci-manila https://zuul.opendev.org/t/openstack/build/203bfbde88a14b3080b7824a597d3d8f : SUCCESS in 46m 03s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f1bf19c49b474404bb0b5a17361c12a6 : FAILURE in 36m 21s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e3aa680e0c6a4c0e950c18eb66706cf0 : TIMED_OUT in 1h 01m 01s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/d0fbb82ba3e347c284e16ab5d76e2935 : SUCCESS in 53m 17s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c1ea93950f8a4da7a3a655b0b7a2ba99 : SUCCESS in 40m 20s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/b87a66a147bd49599ca0bd664eaf71d6 : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":6},{"id":"201f2d532ec04bba5c7b72d21afe0a9ef9553cef","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-21 18:36:54.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"7ed5b7b00e3260bfeb23287387a0d2ef28b8b5b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 19:38:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/df2243ace3cc4d7d8bb123dc734a2cd7 : SUCCESS in 48m 14s\n- loci-glance https://zuul.opendev.org/t/openstack/build/256639d8c312435699aabad13629422e : SUCCESS in 39m 25s\n- loci-heat https://zuul.opendev.org/t/openstack/build/240ab3445e164e788c96216977ca9f1d : SUCCESS in 40m 56s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/32dd1f712c4840a7a6678286bb518bfb : SUCCESS in 41m 10s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d098f317cd844186828105112b4d8847 : SUCCESS in 53m 07s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8db2d455a8ee408585de90e813cc5ed0 : SUCCESS in 53m 05s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7138d5239d0a4e06b6f4231fd9242222 : SUCCESS in 38m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cbc8c89bd629413b8fd5334fc4daeacf : SUCCESS in 52m 10s\n- loci-nova https://zuul.opendev.org/t/openstack/build/aadf065e5509441a97ff3e1ba2ff5a6f : SUCCESS in 53m 02s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/14dff693fcd24f419b70f75cd1dec553 : SUCCESS in 48m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c05f12c18847482db12088978e1794a4 : SUCCESS in 42m 40s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f1f9d659f7c84deeb48bb8605141eb3a : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":6},{"id":"8bb506cbfcd9a249fe7e6c616594fb8901460135","author":{"_account_id":28719,"name":"Phil Sphicas","email":"phil.sphicas@att.com","username":"ps3910"},"date":"2020-01-21 19:55:28.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"bfc4ecbb0b1ddac53870c566621d50df5771f6d6","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-23 16:35:12.000000000","message":"Patch Set 6:\n\nPlease review. Thank you!","accounts_in_message":[],"_revision_number":6},{"id":"eaf95550801e0718d01ece62aba4edbe456cade7","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-27 19:32:56.000000000","message":"Patch Set 6:\n\nPlease review. Stein, Train and Rocky are broken w/o it. Thank you!","accounts_in_message":[],"_revision_number":6},{"id":"a07d39fb930845f240c651155bc355b2523f0d99","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-01-27 19:35:48.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"8b8d21eec144f53585cbd04d467c982dc176c892","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-28 00:52:48.000000000","message":"Patch Set 6:\n\ntest run https://review.opendev.org/#/c/704444/","accounts_in_message":[],"_revision_number":6},{"id":"194d22d51f7542ae9c3254325c7888d2bb14abca","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-28 16:11:04.000000000","message":"Patch Set 6:\n\nplease consider https://review.opendev.org/#/c/704433/ instead","accounts_in_message":[],"_revision_number":6},{"id":"6a64f04c7f887ff5ef44597147d5f7c1e578c79a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-28 16:18:45.000000000","message":"Patch Set 6: Code-Review-1","accounts_in_message":[],"_revision_number":6},{"id":"06453c6c1a1365945448579bc6518159dc40bd1a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-01-30 14:35:20.000000000","message":"Abandoned\n\nalternative commit is merged https://review.opendev.org/#/c/704433/","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bf478313c0963bf8d82a091874a1a756258d6626","revisions":{"33f79d966122c07cfbe40cc24c701e54417a7c75":{"kind":"REWORK","_number":1,"created":"2020-01-21 15:50:17.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/47/703647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/47/703647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/47/703647/1"}}},"commit":{"parents":[{"commit":"d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 15:47:32.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 15:47:32.000000000","tz":-360},"subject":"Remove an entry for building project from uc","message":"Remove an entry for building project from uc\n\nIn some releases (Stein, Train) uc contains at least an entry for\nhorizon, that breaks build for this component with unsatisfied\nconstraints error. Obviously we don\u0027t need this constraint for the\ncomponent we build.\n\nChange-Id: If602d4ccdaa66eed98dc8e9c600137a2f6fbb828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/33f79d966122c07cfbe40cc24c701e54417a7c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/33f79d966122c07cfbe40cc24c701e54417a7c75"}]},"branch":"refs/heads/master"},"f7641c94ed76c37ba3c91e18802f6a7449a40cd9":{"kind":"REWORK","_number":2,"created":"2020-01-21 17:17:28.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/47/703647/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/47/703647/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/47/703647/2"}}},"commit":{"parents":[{"commit":"d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 15:47:32.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 17:17:23.000000000","tz":-360},"subject":"Remove an entry for building project from uc","message":"Remove an entry for building project from uc\n\nIn some releases (Stein, Train) uc contains at least an entry for\nhorizon, that breaks build for this component with unsatisfied\nconstraints error. Obviously we don\u0027t need this constraint for the\ncomponent we build.\n\nChange-Id: If602d4ccdaa66eed98dc8e9c600137a2f6fbb828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f7641c94ed76c37ba3c91e18802f6a7449a40cd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f7641c94ed76c37ba3c91e18802f6a7449a40cd9"}]},"branch":"refs/heads/master"},"73776edcf85e6cec445bd8c03d3fd5e437ec1545":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-21 17:19:31.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/47/703647/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/47/703647/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/47/703647/3"}}},"commit":{"parents":[{"commit":"d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 15:47:32.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 17:18:57.000000000","tz":-360},"subject":"Remove an entry for building project from uc","message":"Remove an entry for building project from uc\n\nIn some releases (Stein, Train) uc contains at entries for neutron and\nhorizon, that breaks build for these components with unsatisfied\nconstraints error. Obviously we don\u0027t need this constraint for the\ncomponent we build.\n\nChange-Id: If602d4ccdaa66eed98dc8e9c600137a2f6fbb828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/73776edcf85e6cec445bd8c03d3fd5e437ec1545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/73776edcf85e6cec445bd8c03d3fd5e437ec1545"}]},"branch":"refs/heads/master"},"76f4ee52118d01ca413079576c2303dbfb64158f":{"kind":"REWORK","_number":4,"created":"2020-01-21 17:22:41.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/47/703647/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/47/703647/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/47/703647/4"}}},"commit":{"parents":[{"commit":"d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 15:47:32.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 17:22:39.000000000","tz":-360},"subject":"Remove an entry for building project from uc","message":"Remove an entry for building project from uc\n\nIn some releases (Stein, Train) uc contains at entries for neutron and\nhorizon, that breaks build for these components with unsatisfied\nconstraints error. Obviously we don\u0027t need this constraint for the\ncomponent we build.\n\nChange-Id: If602d4ccdaa66eed98dc8e9c600137a2f6fbb828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76f4ee52118d01ca413079576c2303dbfb64158f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76f4ee52118d01ca413079576c2303dbfb64158f"}]},"branch":"refs/heads/master"},"d962708a1928da536041a3b9e197a70ba880c350":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-01-21 17:23:41.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/47/703647/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/47/703647/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/47/703647/5"}}},"commit":{"parents":[{"commit":"d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 15:47:32.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 17:23:34.000000000","tz":-360},"subject":"Remove an entry for building project from uc","message":"Remove an entry for building project from uc\n\nIn some releases (Stein, Train) uc contains entries for neutron and\nhorizon, that breaks build for these components with unsatisfied\nconstraints error. Obviously we don\u0027t need this constraint for the\ncomponent we build.\n\nChange-Id: If602d4ccdaa66eed98dc8e9c600137a2f6fbb828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d962708a1928da536041a3b9e197a70ba880c350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d962708a1928da536041a3b9e197a70ba880c350"}]},"branch":"refs/heads/master"},"bf478313c0963bf8d82a091874a1a756258d6626":{"kind":"REWORK","_number":6,"created":"2020-01-21 17:31:33.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/47/703647/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/47/703647/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/47/703647/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/47/703647/6"}}},"commit":{"parents":[{"commit":"d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe","subject":"Update dockerfiles and gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5ff58a973f71f8eacb33ed5f6e20aaf6f8f7abe"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 15:47:32.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-01-21 17:31:19.000000000","tz":-360},"subject":"Remove an entry for building project from uc","message":"Remove an entry for building project from uc\n\nIn some releases (Stein, Train) uc contains entries for neutron and\nhorizon, that breaks build for these components with unsatisfied\nconstraints error. Obviously we don\u0027t need this constraint for the\ncomponent we build.\n\nChange-Id: If602d4ccdaa66eed98dc8e9c600137a2f6fbb828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bf478313c0963bf8d82a091874a1a756258d6626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/bf478313c0963bf8d82a091874a1a756258d6626"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
