)]}'
{"id":"openstack%2Floci~707032","triplet_id":"openstack%2Floci~master~I1023d6b2af4421ed014ba5b7c98e57294c5d46d9","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"I1023d6b2af4421ed014ba5b7c98e57294c5d46d9","subject":"Add support for virtualenv \u003e 20","status":"MERGED","created":"2020-02-10 23:58:34.000000000","updated":"2020-03-06 18:02:56.000000000","submitted":"2020-03-06 18:02:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707032-1583517776106-cb471cbf","meta_rev_id":"ed6be218cef46b357478252577a9273947001a11","_number":707032,"virtual_id_number":707032,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 18:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"date":"2020-03-06 15:53:10.000000000","_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"recommended":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"all":[{"value":1,"date":"2020-03-04 04:35:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":1,"date":"2020-03-03 18:56:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":1,"date":"2020-03-01 09:44:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":2,"date":"2020-03-06 03:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2020-03-03 19:43:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2020-03-03 18:31:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":1,"date":"2020-03-06 03:43:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 00:41:00.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-03-01 09:44:58.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"state":"REVIEWER"},{"updated":"2020-03-03 18:31:02.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2020-03-03 18:56:13.000000000","updated_by":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"reviewer":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"state":"REVIEWER"},{"updated":"2020-03-03 19:43:47.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-03-04 04:35:37.000000000","updated_by":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"reviewer":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"state":"REVIEWER"},{"updated":"2020-03-06 03:43:54.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-03-06 18:02:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"562bf4190598b278d0c98a5550bc1ce4ae118011","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-10 23:58:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10eec4221741d6574afec017443c10c9175f34c0","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-11 00:19:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"21541a2ee4ba9c827b97add102eca1c3d0f38822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 01:17:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/dc011e9bd4744736b692f93acf862234 : FAILURE in 29m 10s\n- loci-glance https://zuul.opendev.org/t/openstack/build/aeb485190c524923896f2b64343caec8 : FAILURE in 31m 58s\n- loci-heat https://zuul.opendev.org/t/openstack/build/64c6873297294cf7a6084c0b62dc7a50 : FAILURE in 30m 04s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ed522cf2a0ea489f8bc18d36600e690c : FAILURE in 31m 28s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b534a97117cb41aa908bbab1fc0cd51d : FAILURE in 32m 23s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/33334ece25554e3eac1b01a1f2bcadae : FAILURE in 26m 55s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f9bd6640690846189b25af6718d4d30c : FAILURE in 40m 24s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a96e3c086e094bb2ab5c98a1463d98f1 : FAILURE in 34m 09s\n- loci-nova https://zuul.opendev.org/t/openstack/build/75df49e288b745dd946cbf2088197c28 : FAILURE in 28m 09s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/68ff2a4afae64160921f27729b6cc556 : FAILURE in 26m 34s\n- loci-placement https://zuul.opendev.org/t/openstack/build/c57804bb4a3f4efdae829891765356e5 : SUCCESS in 34m 26s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/13d1c6ae0f7c489d9312fbbb5b3821a5 : FAILURE in 33m 33s","accounts_in_message":[],"_revision_number":2},{"id":"d1aafe6259c88b6ffe6bab7c9ea41a5ad521cc05","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-11 01:37:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b31d9ca95fca27067b837f75fdcf2608bf4f0d1e","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-11 01:51:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d58d1c26e5d1e9ed224a415927a5cd7436e9f215","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-11 02:10:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e5751ba2d0fb2041256e36619d6dbb80132e715e","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-11 02:32:58.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"dca6a2deddd838e7918da5d2bb087171a3b6ff7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 03:26:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d3c29c2707a544b38461fa07bd221759 : SUCCESS in 47m 47s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c1c4087a2f844b9b80fd2c6c88228866 : SUCCESS in 48m 26s\n- loci-heat https://zuul.opendev.org/t/openstack/build/1a044f599042469cbfb9ada00342cbeb : SUCCESS in 36m 49s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2ba6f479f8a0428eade17a60946ea171 : SUCCESS in 47m 00s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/469e0492b9a441c6a315f6736dd526ae : SUCCESS in 35m 19s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/664fe14fa07644588360c16191bec7ff : SUCCESS in 43m 40s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8873f85dc67a443abce49aaded7d06a0 : SUCCESS in 47m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ab02c796365e46868f9c499cefb60887 : SUCCESS in 46m 24s\n- loci-nova https://zuul.opendev.org/t/openstack/build/31bae86109424b089b4c3e2fb3d0cbbb : SUCCESS in 50m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c05c625bb1924cf69989cbed8842217a : SUCCESS in 46m 17s\n- loci-placement https://zuul.opendev.org/t/openstack/build/fd9f7104a97740b8be388c2960d9f9c4 : SUCCESS in 35m 07s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8602fc0c4b4f4daf8f9ad1c46bbddb93 : SUCCESS in 31m 25s","accounts_in_message":[],"_revision_number":6},{"id":"0948797afa85fbe20c7a070ab088eb23a300d39c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-11 04:51:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6},{"id":"f955cc9e3f109538614ecb23333277ab26470e78","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-24 23:59:09.000000000","message":"Restored","accounts_in_message":[],"_revision_number":6},{"id":"48d0040ca3ccfe563edb01fdf185f7a691b3398a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-24 23:59:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8f7a65e191d72a0af4cf7cae7168ada2cae6ba30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 00:25:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/384d2491087d4edca717f1e2d8b19b4f : FAILURE in 19m 18s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ee370af023f1419cad8208cbf6b293ea : FAILURE in 18m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/46806a0690324c27a70063ac18138445 : FAILURE in 15m 13s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/24b94c8dfaab4676ab14888c806f7145 : FAILURE in 16m 24s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/77c06e3c46184e56a8782200f3c5a473 : FAILURE in 17m 39s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2e18dbe4f811486f97506ba4008c1773 : FAILURE in 17m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e372cd6725004e6aacd61ff34c933a43 : FAILURE in 18m 25s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6fe359fe5cc94c05ab62ba52a96be7c8 : FAILURE in 17m 33s\n- loci-nova https://zuul.opendev.org/t/openstack/build/31b85df3112f418895052fcea0290b8a : FAILURE in 19m 22s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ce0c185296c94a9ab01e72370da6ec5a : FAILURE in 15m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/bb98128a5fb748d1903da95080293a35 : FAILURE in 17m 26s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0e67a2cf6902492da0d5df58d0d20add : FAILURE in 16m 33s","accounts_in_message":[],"_revision_number":7},{"id":"10f988803b645468ec9273a00e4bfb7139a4aacf","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-25 03:43:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e6ae45d6282f5333587da6b096ecd0e94816e30d","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-25 03:53:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d824323c6396bd0d8c0b4aade94165f0f5ef0eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 04:57:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9b52c37a8bed42688528aec272efd853 : TIMED_OUT in 1h 01m 02s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cd08995ae32c47d8b89a169c4714a658 : TIMED_OUT in 1h 01m 08s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c4ac234df8904f278361f9a6f05ba46c : TIMED_OUT in 1h 01m 06s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/07d1fa9a9e344dad9d9ca17b40ff80f2 : TIMED_OUT in 1h 01m 00s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/38727497573e432a9d63b4585f773608 : SUCCESS in 39m 53s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/40954dd1fda3446bb82351c1eee8ad81 : SUCCESS in 43m 59s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8aae044974a24d7e8bf2cd19ce60bf33 : SUCCESS in 40m 49s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/9312e696a7014d3fb6b3712635324d0a : SUCCESS in 47m 48s\n- loci-nova https://zuul.opendev.org/t/openstack/build/de4723ae3cb343cebe9982d9e11fb274 : TIMED_OUT in 1h 00m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/780bbebf9d5c4c64ac0e3e30adf2a57e : SUCCESS in 45m 04s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4b006d8e4cc844bcb28b8590ffd5ebc5 : SUCCESS in 37m 01s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f9d1141e4c034e41a82181e9649b59ee : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":9},{"id":"a1e4e151895bd622056c6af56dddb89602b724a3","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-25 04:58:08.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"7cbdb51f286bfd76d4933fa110d76ddd9551b73e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 05:54:55.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/6b89868828eb4d49ab33c16ad0e060e0 : SUCCESS in 38m 07s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9f0344b078a04238b85a385478c74d7a : SUCCESS in 43m 53s\n- loci-heat https://zuul.opendev.org/t/openstack/build/49cdf7801a13405fa483d21b13dd33f0 : SUCCESS in 50m 43s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/14279653ce764105a2aab7f17d918211 : SUCCESS in 44m 33s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/90d978b405a746ccafed331dfba929f9 : SUCCESS in 43m 26s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/30e0ed90297f46c38070a548a547ea2c : SUCCESS in 42m 16s\n- loci-manila https://zuul.opendev.org/t/openstack/build/eec3da9e863741668978f7f103fc40e9 : SUCCESS in 44m 14s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/0551892134af4c7c854943a0c2c67f8c : SUCCESS in 43m 31s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bdf57700b0804d2e8ae45b6292a4d712 : SUCCESS in 55m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/1ea5ebded45b405cba2533fc8e979222 : SUCCESS in 54m 02s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4e531eb00b5b478bbb1b63d7952e59d6 : SUCCESS in 36m 03s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/617c3f79b6b84d3eb6528713a383f24e : TIMED_OUT in 40m 56s","accounts_in_message":[],"_revision_number":9},{"id":"049921b55b197ead416fb15e9652e54246dc1107","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-25 06:03:59.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"c968beb34e20a01ed942dc2d9b425d6f30e38b05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 07:12:21.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/de05502a799a40b7afcaf2da64333605 : SUCCESS in 52m 57s\n- loci-glance https://zuul.opendev.org/t/openstack/build/34d4dbffef5340b98cdfcf8963b61294 : SUCCESS in 46m 11s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ae39a73feb5d47dc90cd717d11b6519f : SUCCESS in 58m 52s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/7bfef63e9e8948beb9816202f9a454a7 : SUCCESS in 48m 50s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/6b1be2802a3844edb0aff8957eb11abf : SUCCESS in 49m 09s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a35713a2dc4246149ea4bc8a45b3bb08 : SUCCESS in 48m 24s\n- loci-manila https://zuul.opendev.org/t/openstack/build/92fafc6bf4054398954507caaf83b35d : SUCCESS in 49m 56s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/0c5100fa62b3450caf0fa4acc7bdd45f : SUCCESS in 49m 11s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c76201e0ea414e4697e988f237fcf7a9 : TIMED_OUT in 1h 01m 01s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/fa45f52050ea4ac09ce5dae8b2c881de : SUCCESS in 51m 27s\n- loci-placement https://zuul.opendev.org/t/openstack/build/271c28b9181f417c874422b3f9a45a97 : SUCCESS in 40m 37s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8f4abeb91d4f470ab0ef3d783a07de61 : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":9},{"id":"c0a5ef7363ac06935f73f695365462fca7eb5778","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-25 09:48:32.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"b380c5cd964edba0c6255cd08a01b06fe5b9f415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:48:10.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f907bbe489c3436aa5fa70d7f6c40ca3 : SUCCESS in 54m 21s\n- loci-glance https://zuul.opendev.org/t/openstack/build/17edfea66c1c436691ca019b1785c002 : SUCCESS in 39m 55s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e8fedc25736f418ba854aea5798dfb44 : SUCCESS in 43m 32s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f1ab370c648442eba7ea90606a1199eb : SUCCESS in 42m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ceb68304cec542b88254ad83b9c32a44 : SUCCESS in 50m 15s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1a2e4c6c3dd34b43a7587be8d0765249 : SUCCESS in 46m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0b37448ff247453dbe1aba2bf313e5a3 : SUCCESS in 48m 14s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/116424c982a3430faf67ec30bf61b033 : SUCCESS in 47m 03s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0b41469d5ef94048b077518ddaa38150 : SUCCESS in 51m 42s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/434b9967767f43a9a9a1636b906f3e3f : SUCCESS in 39m 15s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e729ba63daeb4495bc81b47b9b1e583c : SUCCESS in 35m 28s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/975ca1cd29d14dceaa5e88b5135520fb : SUCCESS in 29m 10s","accounts_in_message":[],"_revision_number":9},{"id":"e0f40dd486f9bac46677fa4fcc6faac7a12ecb87","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-02-25 16:49:13.000000000","message":"Patch Set 9:\n\nplease review","accounts_in_message":[],"_revision_number":9},{"id":"da300a84c646bd5c706dadcc182057b9f1377283","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2020-03-01 09:44:58.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"96a6510c12f1534abd60f62610d40371de55caf9","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-03-03 18:31:02.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"0ad5e0e4580f57443a32130fc0fb41aef42cacc7","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-03 18:56:13.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"63203a5ffeb8c3523f793675726b6aee7cc0583d","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-03-03 19:43:47.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"a527c995caf4e626ba668b6c96cf544d0a9f8c1f","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2020-03-04 04:35:37.000000000","message":"Patch Set 9: Code-Review+1\n\nworked for us, thanks!","accounts_in_message":[],"_revision_number":9},{"id":"6cb6cacd9a7311faec2cc360750397c56d0cf640","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2020-03-06 03:43:54.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"528637aa36241dabd9d6e21181e641827816c694","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 03:44:05.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"cc7f45b5f3a8b3d1cb6e5d9f98e3279c53cd9b8b","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-03-06 05:17:38.000000000","message":"Patch Set 9:\n\nwill fail due to https://review.opendev.org/#/c/694387/","accounts_in_message":[],"_revision_number":9},{"id":"fa4dfa3e40a16204861ffeeab6b6b0a9c1ec8726","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-03-06 05:26:13.000000000","message":"Patch Set 9:\n\nfix https://review.opendev.org/#/c/711581/\nwill retrigger gates once merged","accounts_in_message":[],"_revision_number":9},{"id":"c01d1446cbe3b469ba0d9cbc77a18561c1a94998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 05:38:19.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/73521f9c5fc24babadb4ab741cff0feb : SUCCESS in 51m 12s\n- loci-glance https://zuul.opendev.org/t/openstack/build/4177d10522a74664879a7e5facd9dbc5 : FAILURE in 42m 23s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0b6ff3ffd70c4208ad3912115853baed : SUCCESS in 41m 53s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/93bd49cda4114ddda5b7a30847ca220b : SUCCESS in 43m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3a67634483ab49b2b37075066ac36d1a : SUCCESS in 43m 52s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2e66c6a1b8124d47936047aff2655537 : SUCCESS in 53m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/1219e98c1b924b9d9daa386b02b3ceec : SUCCESS in 43m 43s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/16bcec5d9d0a4d9ca526d5c42d6f1c1b : SUCCESS in 49m 58s\n- loci-nova https://zuul.opendev.org/t/openstack/build/5c34e0f9129e4ed5bad67eabe97ca030 : SUCCESS in 48m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/43649da55c4146c391693fa10a75ee77 : SUCCESS in 43m 44s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/2a432d6570444216b835d65a79b7c616 : TIMED_OUT in 40m 46s","accounts_in_message":[],"_revision_number":9},{"id":"1a3a9e25e4c232d35b423e78bed4913cd400857d","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-03-06 09:47:57.000000000","message":"Patch Set 9:\n\nhttps://review.opendev.org/#/c/711581/ merged","accounts_in_message":[],"_revision_number":9},{"id":"61f26c797492c91fad783e3e6610382b494d8337","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-03-06 09:48:06.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"0218bf4436a03da9977671fea8be3840886eaab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 10:49:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d44752c93a5144b2aa2d3f7f89aa491e : SUCCESS in 55m 25s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cab77b7032f247a2b2fc96d5de53128e : SUCCESS in 40m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/4a2769dec4204efcb97a6bf7137a016e : SUCCESS in 55m 08s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4567e0ba792547ecb800841c79fc7d5e : SUCCESS in 43m 20s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/47b7cb309d1642dea5375fc1317ca53f : SUCCESS in 37m 02s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a06aec8ae5864d5caa621a4e252a9759 : SUCCESS in 54m 18s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5b814b2f365a4e5fb14a84964b329da7 : SUCCESS in 45m 58s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f371350392014e2496cc373a91df706b : SUCCESS in 48m 03s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fe44482ff4474ce79ec89c38886d58e6 : SUCCESS in 39m 39s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3bb7b2c468a246128e6cf9342cb780dd : SUCCESS in 44m 24s\n- loci-placement https://zuul.opendev.org/t/openstack/build/6161e664cd3c442a84e25ce2f3511534 : SUCCESS in 37m 55s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/649f6d37ccbf48c3a2c64964e6c0a209 : TIMED_OUT in 40m 52s","accounts_in_message":[],"_revision_number":9},{"id":"e8efec4d20f3a95a9a029b313091f87f4ba0659f","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-03-06 10:58:32.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"04c7b8ad43ca26f25a915670727b872df8e0eff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 11:53:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/af780cf4c1a5401ea1695b030c2dd2f7 : SUCCESS in 46m 25s\n- loci-glance https://zuul.opendev.org/t/openstack/build/fde9b308b6034d7fbb5c69bb237100fc : SUCCESS in 41m 27s\n- loci-heat https://zuul.opendev.org/t/openstack/build/621ebc797eeb44ce818fc83358738ad5 : SUCCESS in 44m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/cd295de1fbc347289cce8ea2614fa49e : SUCCESS in 47m 26s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/88b1deabdc774d5182b7b2707bbfa586 : SUCCESS in 44m 17s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1c6d764c2e9444f8a3789817bb094dc8 : SUCCESS in 49m 50s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f049565d97d5435db35d8b22d14a579e : SUCCESS in 45m 55s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/073fbe270410415bbeb73426064d61e1 : SUCCESS in 46m 29s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e8f647a2563a4e63ad80ca30b173cf39 : SUCCESS in 48m 40s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/46e3591bb73a402c9680c59c3e653f7b : SUCCESS in 46m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e42c9c8a5b18491a9cd1e58d122a62d8 : SUCCESS in 35m 04s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ccea628952a044dc8b50d4ea99e75146 : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":9},{"id":"932a93dbfbcd156a43c469166b61aacb5768befb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 11:53:55.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"2d00731eb71fbd23c479c8f0c50d7553d90a0a67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 12:53:37.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/098199179bd84c7ab1132f8d7e0b32f2 : SUCCESS in 52m 46s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0b8405410f3a43f7b6e68f4344217458 : FAILURE in 43m 14s\n- loci-heat https://zuul.opendev.org/t/openstack/build/cfd00210f8374834a30fc35762374c1f : SUCCESS in 44m 03s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a48c0446d90f491d862a7c7188915fe3 : SUCCESS in 36m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0a6a0a86807648afbc7e39779cb439ed : SUCCESS in 42m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2535417abb1a4f5f83e2254045d097d0 : SUCCESS in 36m 19s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f5972e35a53c4ea9a6c1ea1dc6a1c689 : SUCCESS in 35m 11s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/944cd1a5c16547be8d9554e03bac2e14 : SUCCESS in 48m 11s\n- loci-nova https://zuul.opendev.org/t/openstack/build/1e4f6c98334b40bfb15b3203ef4bc658 : SUCCESS in 49m 03s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3258fec519634359a23a32e5c41736f8 : SUCCESS in 45m 31s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6e3d821f2b2e4c7ea18a0bca8ad3865a : SUCCESS in 32m 51s","accounts_in_message":[],"_revision_number":9},{"id":"2cb91b20101cb36803d9d5b598c25d1b88a630ca","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-03-06 15:53:10.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"026ee5e4261a1437ab945f2ff4789296693f0116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 16:57:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a8fc0dcd873844d5998b71314995aac1 : SUCCESS in 43m 15s\n- loci-glance https://zuul.opendev.org/t/openstack/build/076d496f243849848fdb08cfcd3dfd08 : SUCCESS in 45m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e405c869b6ba42e29e0b4dc38a4900e9 : SUCCESS in 46m 59s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/6bdcb4aba71543d0ac8d6b77d7180f46 : SUCCESS in 46m 47s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/67d0cbb1bccf450b8565975998c14985 : SUCCESS in 41m 50s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a9720e5c6d9847eea31a18b92e2d8ccd : SUCCESS in 37m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/d5745f01e06f439daec4ed2edd1ee55a : SUCCESS in 47m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/bd1a1985b24c4ce68d9024f07893017f : SUCCESS in 50m 27s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ebd54a049de84c4eb0c2f5f7d35dab0a : SUCCESS in 51m 25s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/875eacb497934e51bc22107006ea0a84 : SUCCESS in 47m 11s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8392fcb2844f40c2ae4829d0c74b188a : SUCCESS in 41m 22s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/9b212f8d93b24efda28889c25890e9a1 : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":9},{"id":"cb3438115be61eaadf5fa1366308dff2444799e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 16:58:11.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"f860598e88706fac93601132836f2ad10becbf98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 18:02:55.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/91b4fa720cab43068f75e2e906e2bd15 : SUCCESS in 38m 58s\n- loci-glance https://zuul.opendev.org/t/openstack/build/29e4e6afb21a4020b883115bb3a52568 : SUCCESS in 37m 31s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a051de3916754889b40c1f9a11e4899c : SUCCESS in 42m 49s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/cbe64366ddbe417898d4a39a2bfc72ff : SUCCESS in 44m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/4ba70a8ee0c145239fd5e823e212e4be : SUCCESS in 37m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3f4ba6d1948148e2b0d711de650520dd : SUCCESS in 39m 34s\n- loci-manila https://zuul.opendev.org/t/openstack/build/4d209a7429c944689c48117239c6028b : SUCCESS in 48m 00s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d71e0230ad904a8cac04ccce16a4db52 : SUCCESS in 50m 14s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c2771f866e5d4fa196186d8c44cc65c0 : SUCCESS in 58m 38s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/631df91a5cd14d6298d5f59271f2b2ee : SUCCESS in 49m 20s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a1b68925f80b49a09b65f5589f76fb24 : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":9},{"id":"ed6be218cef46b357478252577a9273947001a11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 18:02:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"c94a86c32894f9d43ac008a7079fe7663fce3b80","revisions":{"0aa09b6f41c76edd1106d88af19cff9e2d3b4564":{"kind":"REWORK","_number":1,"created":"2020-02-10 23:58:34.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/32/707032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/707032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/707032/1"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"subject":"Add support for virtualenv \u003e 20","message":"Add support for virtualenv \u003e 20\n\nVirtualenv \u003e 20 no longer supports --no-site-packages option and does\nnot create no-global-site-packages.txt flag file.\nChanging pyvenv.cfg instead to achieve the same.\n\nChange-Id: I1023d6b2af4421ed014ba5b7c98e57294c5d46d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0aa09b6f41c76edd1106d88af19cff9e2d3b4564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0aa09b6f41c76edd1106d88af19cff9e2d3b4564"}]},"branch":"refs/heads/master"},"673a5a6fe19a7423e2ff35e44a4c022a281551c6":{"kind":"REWORK","_number":2,"created":"2020-02-11 00:19:00.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/32/707032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/707032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/707032/2"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-11 00:18:57.000000000","tz":0},"subject":"Add support for virtualenv \u003e 20","message":"Add support for virtualenv \u003e 20\n\nVirtualenv \u003e 20 no longer supports --no-site-packages option and does\nnot create no-global-site-packages.txt flag file.\nChanging pyvenv.cfg instead to achieve the same.\n\nChange-Id: I1023d6b2af4421ed014ba5b7c98e57294c5d46d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/673a5a6fe19a7423e2ff35e44a4c022a281551c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/673a5a6fe19a7423e2ff35e44a4c022a281551c6"}]},"branch":"refs/heads/master"},"9149d3601b4e1973ec5f08aedbeaede770d3149e":{"kind":"REWORK","_number":3,"created":"2020-02-11 01:37:13.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/32/707032/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/707032/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/707032/3"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-11 01:36:18.000000000","tz":-360},"subject":"Add support for virtualenv \u003e 20","message":"Add support for virtualenv \u003e 20\n\nVirtualenv \u003e 20 no longer supports --no-site-packages option and does\nnot create no-global-site-packages.txt flag file.\nChanging pyvenv.cfg instead to achieve the same.\n\nAlso install setuptools for py2 required for virtualenv \u003e20 as a\nfilelock installation dependency.\n\nChange-Id: I1023d6b2af4421ed014ba5b7c98e57294c5d46d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9149d3601b4e1973ec5f08aedbeaede770d3149e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9149d3601b4e1973ec5f08aedbeaede770d3149e"}]},"branch":"refs/heads/master"},"736333efa38b61d42508f395b0bbbd6d99e4131a":{"kind":"REWORK","_number":4,"created":"2020-02-11 01:51:19.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/32/707032/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/707032/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/707032/4"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-11 01:50:35.000000000","tz":-360},"subject":"Add support for virtualenv \u003e 20","message":"Add support for virtualenv \u003e 20\n\nVirtualenv \u003e 20 no longer supports --no-site-packages option and does\nnot create no-global-site-packages.txt flag file.\nChanging pyvenv.cfg instead to achieve the same.\n\nAlso setuptools in intermediate venv required for virtualenv \u003e20 as a\nfilelock installation dependency.\n\nChange-Id: I1023d6b2af4421ed014ba5b7c98e57294c5d46d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/736333efa38b61d42508f395b0bbbd6d99e4131a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/736333efa38b61d42508f395b0bbbd6d99e4131a"}]},"branch":"refs/heads/master"},"6b15afd339fc7962297b7bb23a05d423ee113bc0":{"kind":"REWORK","_number":5,"created":"2020-02-11 02:10:29.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/32/707032/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/707032/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/707032/5"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-11 02:10:17.000000000","tz":-360},"subject":"Add support for virtualenv \u003e 20","message":"Add support for virtualenv \u003e 20\n\nVirtualenv \u003e 20 no longer supports --no-site-packages option and does\nnot create no-global-site-packages.txt flag file.\nChanging pyvenv.cfg instead to achieve the same.\n\nAlso move setuptools installation to intermediate venv as it\u0027s not installed\nwith pip and required for virtualenv \u003e 20\n\nChange-Id: I1023d6b2af4421ed014ba5b7c98e57294c5d46d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6b15afd339fc7962297b7bb23a05d423ee113bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/6b15afd339fc7962297b7bb23a05d423ee113bc0"}]},"branch":"refs/heads/master"},"004e275dfe21851e698b86b7ebe60c64b6a9129e":{"kind":"REWORK","_number":6,"created":"2020-02-11 02:32:58.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/32/707032/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/707032/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/707032/6"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-11 02:32:57.000000000","tz":0},"subject":"Add support for virtualenv \u003e 20","message":"Add support for virtualenv \u003e 20\n\nVirtualenv \u003e 20 no longer supports --no-site-packages option and does\nnot create no-global-site-packages.txt flag file.\nChanging pyvenv.cfg instead to achieve the same.\n\nAlso move setuptools installation to intermediate venv as it\u0027s not installed\nwith pip and required for virtualenv \u003e 20\n\nChange-Id: I1023d6b2af4421ed014ba5b7c98e57294c5d46d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/004e275dfe21851e698b86b7ebe60c64b6a9129e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/004e275dfe21851e698b86b7ebe60c64b6a9129e"}]},"branch":"refs/heads/master"},"666bd5b4333f1f7a757940cab76a482b9034dbdc":{"kind":"REWORK","_number":7,"created":"2020-02-24 23:59:10.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/32/707032/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/707032/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/707032/7"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-24 23:58:23.000000000","tz":-360},"subject":"Add support for virtualenv \u003e 20","message":"Add support for virtualenv \u003e 20\n\nVirtualenv \u003e 20 no longer supports --no-site-packages option and does\nnot create no-global-site-packages.txt flag file.\nChanging pyvenv.cfg instead to achieve the same.\n\nChange-Id: I1023d6b2af4421ed014ba5b7c98e57294c5d46d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/666bd5b4333f1f7a757940cab76a482b9034dbdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/666bd5b4333f1f7a757940cab76a482b9034dbdc"}]},"branch":"refs/heads/master"},"06d8ebe846dbf9939e498adc24fe9798b229414c":{"kind":"REWORK","_number":8,"created":"2020-02-25 03:43:14.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/32/707032/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/707032/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/707032/8"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-25 02:16:06.000000000","tz":-360},"subject":"Add support for virtualenv \u003e 20","message":"Add support for virtualenv \u003e 20\n\nVirtualenv \u003e 20 no longer supports --no-site-packages option and does\nnot create no-global-site-packages.txt flag file.\nChanging pyvenv.cfg instead to achieve the same.\n\nChange-Id: I1023d6b2af4421ed014ba5b7c98e57294c5d46d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/06d8ebe846dbf9939e498adc24fe9798b229414c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/06d8ebe846dbf9939e498adc24fe9798b229414c"}]},"branch":"refs/heads/master"},"c94a86c32894f9d43ac008a7079fe7663fce3b80":{"kind":"REWORK","_number":9,"created":"2020-02-25 03:53:35.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/32/707032/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/32/707032/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/32/707032/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/32/707032/9"}}},"commit":{"parents":[{"commit":"690c3d7673ff1a48b2ac787505bffaf621d86eba","subject":"Remove an entry for building project from uc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/690c3d7673ff1a48b2ac787505bffaf621d86eba"}]}],"author":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-10 23:56:39.000000000","tz":-360},"committer":{"name":"Andrii Ostapenko","email":"aostapenko@mirantis.com","date":"2020-02-25 03:53:31.000000000","tz":-360},"subject":"Add support for virtualenv \u003e 20","message":"Add support for virtualenv \u003e 20\n\nVirtualenv \u003e 20 no longer supports --no-site-packages option and does\nnot create no-global-site-packages.txt flag file.\nChanging pyvenv.cfg instead to achieve the same.\n\nChange-Id: I1023d6b2af4421ed014ba5b7c98e57294c5d46d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
