)]}'
{"id":"openstack%2Floci~712891","triplet_id":"openstack%2Floci~master~I2b60a81168f5c47eab316d2be31a4fe3ba9c2254","project":"openstack/loci","branch":"master","topic":"for_ussuri","hashtags":[],"change_id":"I2b60a81168f5c47eab316d2be31a4fe3ba9c2254","subject":"Add an urlopen exception for ImcompleteRead error.","status":"MERGED","created":"2020-03-13 09:48:13.000000000","updated":"2020-04-24 05:13:22.000000000","submitted":"2020-04-24 05:13:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712891-1587705202324-792c2e80","meta_rev_id":"e791a75fb4655a8b1a38dcfc84a28c59f1bd36b8","_number":712891,"virtual_id_number":712891,"owner":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"date":"2020-04-23 15:30:28.000000000","_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 05:13:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":0,"_account_id":28431,"name":"Don Penney","email":"dpenney@redhat.com","username":"dpenney"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"recommended":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"all":[{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":1,"date":"2020-04-24 01:43:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-24 01:57:12.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":0,"_account_id":28431,"name":"Don Penney","email":"dpenney@redhat.com","username":"dpenney"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"all":[{"value":0,"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"value":0,"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-24 04:16:54.000000000","_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"value":0,"_account_id":28431,"name":"Don Penney","email":"dpenney@redhat.com","username":"dpenney"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":28431,"name":"Don Penney","email":"dpenney@redhat.com","username":"dpenney"},{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 09:49:55.000000000","updated_by":{"_account_id":28431,"name":"Don Penney","email":"dpenney@redhat.com","username":"dpenney"},"reviewer":{"_account_id":28431,"name":"Don Penney","email":"dpenney@redhat.com","username":"dpenney"},"state":"REVIEWER"},{"updated":"2020-04-09 06:57:01.000000000","updated_by":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"reviewer":{"_account_id":14119,"name":"Sam Yaple","email":"sam+git@yaple.net","username":"SamYaple"},"state":"REVIEWER"},{"updated":"2020-04-09 06:57:17.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-04-24 01:43:32.000000000","updated_by":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"reviewer":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"state":"REVIEWER"},{"updated":"2020-04-24 04:16:54.000000000","updated_by":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"reviewer":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"state":"REVIEWER"},{"updated":"2020-04-24 05:13:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d245981ea99430cf3c7460340bd02198391469f0","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-13 09:48:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa6a49e304c24fb1b0f939dac48e353ead033cd7","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-13 09:54:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6d20848ae64af216296ed81df03cb56d055a3873","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-13 10:20:16.000000000","message":"Patch Set 2: Code-Review-1\n\ntest","accounts_in_message":[],"_revision_number":2},{"id":"1c157ee1c25698b0f651d496630641a963c0a2f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 10:35:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/737366e20b17446bb10d09ffdd7b8e98 : FAILURE in 33m 16s\n- loci-glance https://zuul.opendev.org/t/openstack/build/04aa51af2d4d4b6eb9930633485484f0 : FAILURE in 30m 17s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c588b08788474650a228802b08edafe3 : FAILURE in 31m 10s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/40b9de00de9b44049f6ac6ec8322cb54 : FAILURE in 31m 35s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/eb3b77243a974d27b477056fc6e7251e : FAILURE in 30m 40s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7704454ecf82494eb44ff10275cd22e0 : FAILURE in 35m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e3f63eba33a4461895e32e046974bd3b : FAILURE in 31m 14s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/3779edee0adb4bbe925d2af9a32333b8 : FAILURE in 29m 28s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fee6b9bc45b641539ba5606136cf12e0 : FAILURE in 35m 35s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f17761f6c67649f0b97b04fd60cd3e3b : FAILURE in 28m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/04041b9e0fe34475bf1ddcce5db225ae : FAILURE in 33m 25s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5607d6ea3aa34ca69570426bc4e1e50b : FAILURE in 29m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e7c024ab3c3ef8003e9c2964063dc953920119a7","author":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"date":"2020-03-14 02:47:40.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e68f3eff6ad9e98eb9f577d0276a190431adf2c9","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-27 05:37:16.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c23820bd8c80ee32459de270e172f3b25b0f1676","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-27 07:24:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0b3ab40437fd116c151c53f4d2fdfd045efb99e9","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-27 07:30:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8e15125ed380c8914a0626adcafe9c039fb9575f","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-27 07:41:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b2f46f5cee247ce8ac3873bfe4013dfa63478fe4","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-27 07:43:56.000000000","message":"Topic changed from for_master to for_ussuri","accounts_in_message":[],"_revision_number":5},{"id":"88aafbc080dbe58960074e56d3eafc95e17baf9e","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-27 08:24:48.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"26661c24377bab3e65d0effd82e31a14225a3e29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 08:27:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/13f877babe544da78fec4a1676748a87 : FAILURE in 24m 25s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d860b7b5029e4d86a1ab0e5ff661d68c : FAILURE in 27m 51s\n- loci-heat https://zuul.opendev.org/t/openstack/build/87d79f802f9147b8a73edeb7c6ff1c0b : FAILURE in 28m 59s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/35fe115452614ba1a1ee91b4fbc22b7c : FAILURE in 29m 44s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/796199cd1c9545038a7781219ae9569f : FAILURE in 33m 29s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e8133e17241e473aad851a57c9a9fbfb : FAILURE in 27m 22s\n- loci-manila https://zuul.opendev.org/t/openstack/build/ec60978f4c9e4b68a6f7e9e680aca081 : FAILURE in 30m 31s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c814da9130444d3aa058245cf1ee8d74 : FAILURE in 34m 43s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ecb845309a3b4751b470c4578c560846 : FAILURE in 32m 02s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/d3cdf422da7247bdaf480b405fd61800 : FAILURE in 32m 08s\n- loci-placement https://zuul.opendev.org/t/openstack/build/1555679291b34db0b00d54861c44ee09 : SUCCESS in 40m 59s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/445bb50074b64ef497bc20116bbc5b80 : FAILURE in 33m 21s","accounts_in_message":[],"_revision_number":5},{"id":"814bc4c94697866c5143499993f3afad292fb273","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-27 14:39:01.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"80b11df662b874401a906f6bec464fa41da68626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 15:27:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e444ac3e538645fb87a86b3c0283932e : FAILURE in 33m 43s\n- loci-glance https://zuul.opendev.org/t/openstack/build/de5d37f8fda54d529fd277f1b3a0ac9f : FAILURE in 36m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fdb70e4ddf03414ea2d3de45e31cdd75 : FAILURE in 31m 06s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9b7029c9f5ee48fa8917521025b5e0e2 : FAILURE in 30m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9af386460b0048638d5f8be29c90d9af : FAILURE in 34m 44s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/76db572a16f84cff81802d14dd37da3c : FAILURE in 39m 01s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9b7ab4e946474e049bf093e56a95f3fe : FAILURE in 31m 35s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/802e54c75c7d4eb0942bcf908c6f6a71 : FAILURE in 29m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fd040c520ba747599df9541f995c307b : FAILURE in 33m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ec31005b0a904fe092f324d055e54776 : FAILURE in 29m 52s\n- loci-placement https://zuul.opendev.org/t/openstack/build/65dd0e0b5dd241fa81384809cb3896a9 : SUCCESS in 38m 54s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1ed71b2874994e83966a9e8b9998f694 : FAILURE in 36m 07s","accounts_in_message":[],"_revision_number":6},{"id":"261e740342c3dec67677bf5832afe9b734e8ca3d","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-30 15:35:48.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"7fb0241f9113529c7b5478faed15c348de465599","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-30 15:54:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7da40ef21f7ac9640baaa67ea90a99e50657a034","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-30 16:02:11.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"a13d8ec7ae08c7c823f567288d63ebb422ec94a3","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-30 16:04:20.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"0e14fe170d6fcd0dc612e303632ba202d6f2bbd3","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-30 16:41:19.000000000","message":"Patch Set 8: -Workflow","accounts_in_message":[],"_revision_number":8},{"id":"406cdf2ec04d7f7772dc67da73833d7df9fa54d2","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-03-30 16:43:57.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"0dbacfc0a9f4705a586810eaaf7eba8f4fa43312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 17:10:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/fe68d67868f745a8a5a7ae52d6334fd3 : SUCCESS in 45m 43s\n- loci-glance https://zuul.opendev.org/t/openstack/build/c1bcf19d2b534ae9bb6f17e9b3f0ed43 : SUCCESS in 47m 57s\n- loci-heat https://zuul.opendev.org/t/openstack/build/48ebfd1f3a8240f09f55aef66b49627b : SUCCESS in 46m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9058922f356a43baa95c58b542dcc81e : TIMED_OUT in 1h 02m 05s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/ad951e803a634b93b659d339abbb80ed : SUCCESS in 38m 34s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6eb3b0d12aea468487c0eca2814ddcaa : SUCCESS in 47m 09s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c964e0078d4e4f08928c72d364aa0f60 : SUCCESS in 51m 11s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c5d466db06df40aeb532d8ea372265f3 : SUCCESS in 50m 00s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a743a734c3d942e2a53f313837bb9326 : TIMED_OUT in 1h 01m 34s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ad16eff5b2f44f7c9aa4bde4849612a2 : SUCCESS in 48m 28s\n- loci-placement https://zuul.opendev.org/t/openstack/build/12c050e7656449148d1c7551a56637de : SUCCESS in 49m 08s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/9b474831fe7e44c687bf263e1477a796 : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":8},{"id":"345ec73afcbc504c5d0f671c6131e8dd6c6c5cce","author":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"date":"2020-04-22 14:58:10.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"93801935e860e75ae7a8a8bac0bba98660d9e09c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 16:04:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/26234b0656204139aab63101bf4693a1 : SUCCESS in 51m 29s\n- loci-glance https://zuul.opendev.org/t/openstack/build/adb391d5998a4001863cd9f8e935ff54 : SUCCESS in 42m 54s\n- loci-heat https://zuul.opendev.org/t/openstack/build/5c47559211fa4dc7ba4dbcb9822cd4c9 : SUCCESS in 42m 36s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/81503535067d494595ef030a4cd8892d : SUCCESS in 52m 15s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/246eb2805c884a1182759a33fa180f3a : SUCCESS in 53m 10s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/23712f85ad244425915897a91982692c : SUCCESS in 47m 11s\n- loci-manila https://zuul.opendev.org/t/openstack/build/50108840bdfd4e83ba0ebc88dfdc3bc7 : SUCCESS in 43m 26s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/0bad9dd755a54352b41bcbcf7e8c0358 : SUCCESS in 46m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/79c726f5ea284c38be81076ac78e0d7e : SUCCESS in 1h 00m 49s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/c53b782a97b04f17b9488c0cdbca4a3d : SUCCESS in 44m 29s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e83c94ec95b14333b585f1c9e6a1e824 : SUCCESS in 50m 21s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e67096848b0446e0b19a787750f8ba67 : SUCCESS in 39m 55s","accounts_in_message":[],"_revision_number":8},{"id":"38615fbf22f96adef83410ac61ef9d1676f0b0aa","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2020-04-22 16:24:34.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nWhy does the program continue if the read fails?","accounts_in_message":[],"_revision_number":8},{"id":"7cd94243fc4ed4e5a8c49964a906f6f45f642466","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-04-23 00:13:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dd27ea58cf07af1b6099586f19d2403506f2b2c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 01:17:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7d9db9cae6ed41e283557225aa3a357e : SUCCESS in 52m 51s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d5c02a873bb544a1bb3b89bf08bc1764 : SUCCESS in 45m 39s\n- loci-heat https://zuul.opendev.org/t/openstack/build/378f413ea016410d969d591ef8603e95 : TIMED_OUT in 1h 01m 09s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/693f44eab8f34ce5b7c06283a0e78a24 : TIMED_OUT in 1h 01m 04s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/389afdbb7cd849f79da4e139576615a4 : TIMED_OUT in 1h 01m 06s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2bd01cc849384bc59fbcfd2a6c7df996 : TIMED_OUT in 1h 01m 10s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8509e53d918e4b05b7a07970bfa19fdd : TIMED_OUT in 1h 01m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cf555610f8a84bd9a5521fc267c7b758 : SUCCESS in 59m 46s\n- loci-nova https://zuul.opendev.org/t/openstack/build/353d160007614479a4621b02eb9b96b4 : SUCCESS in 48m 27s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e94df513ca1742c6b089653b1f628f89 : TIMED_OUT in 1h 01m 06s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5844171061aa4a5fa251991cc74f0497 : TIMED_OUT in 1h 01m 01s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1dd6af3bd3e9459983944e0501b9d7f9 : TIMED_OUT in 41m 00s","accounts_in_message":[],"_revision_number":9},{"id":"1cf2d966788bfc0c19b5321d4fa79c6a81bf3536","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-04-23 15:23:27.000000000","message":"Patch Set 9:\n\nThe error log is came from \"http.client.IncompleteRead\". Which is an \"except\" process, the \"try\" process is _safe_read(). Based on its definition, it can\u0027t distinguish between EOF and an interrupt when zero bytes have been read, so IncompleteRead() will be raised in this situation. The exception doesn\u0027t mean it is wrong but unknown. So we catch the exception data, and make process keep going.","accounts_in_message":[],"_revision_number":9},{"id":"4418b2c0007f9350f6ba1719af57c1304bd646a5","author":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"date":"2020-04-23 15:30:28.000000000","message":"Patch Set 9:\n\nBesides, this issue happen always since using python3. Without this exception, it can\u0027t pass.","accounts_in_message":[],"_revision_number":9},{"id":"76d3c71ba1884c8c0ed1121808620ef23b3c015c","author":{"_account_id":28570,"name":"zhipeng liu","email":"zhipengs.liu@intel.com","username":"zhipengl"},"date":"2020-04-24 01:43:32.000000000","message":"Patch Set 9: Code-Review+1\n\n\u003e Besides, this issue happen always since using python3. Without this\n \u003e exception, it can\u0027t pass.\nHi Chris,\nAs we know, OpenStack Ussuri release switches to python3 completely,\nIf we do not merge this patch, it will block our upgrade to OpenStack Ussuri release.\nThanks!\nZhipeng","accounts_in_message":[],"_revision_number":9},{"id":"53872a873121b2f97fc00c24f240ca4c4355325c","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2020-04-24 01:57:12.000000000","message":"Patch Set 9: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"9406d87b63fbd1d4e325a6026240a47112a9d5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 02:56:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/294afc11ec644baca737b4478eb30b41 : SUCCESS in 57m 57s\n- loci-glance https://zuul.opendev.org/t/openstack/build/9b6a0767bfe34ea6bb09ee3c168ee030 : SUCCESS in 40m 55s\n- loci-heat https://zuul.opendev.org/t/openstack/build/daf6174d8c6949ce961b99d32b5efa1e : SUCCESS in 56m 17s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/70187f9411024d77b042a2cf69eefe4b : SUCCESS in 57m 17s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e3246c12a63e4e42860d19b20ef6775b : SUCCESS in 40m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8f5ae202fecf45d2bb878af086855083 : SUCCESS in 47m 12s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b5a1afdec4814fe9a0e533a3f9dd3c63 : SUCCESS in 56m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d77984d13aa2476a93da1e1b72ee3f52 : SUCCESS in 51m 14s\n- loci-nova https://zuul.opendev.org/t/openstack/build/4cc4f243658d43af979f3bf121605c83 : SUCCESS in 53m 54s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/9430e7bf28ca410ea174005582a82266 : SUCCESS in 41m 45s\n- loci-placement https://zuul.opendev.org/t/openstack/build/431a2d7ea58b412caaa4c6cb7cef941f : SUCCESS in 40m 02s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c2e5d42bca4e4af58ab97341be62989b : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":9},{"id":"a1cf14cc998b4a6cc8655329bb4dbc4195f837a4","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2020-04-24 04:16:54.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"c4d8ca2056b0dca40392ca550b7ee00011b062e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 04:17:06.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"df1856bdbd39bc30bc4e88318eb6eb12804c5826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 05:13:22.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4a39729d0d6040eea1b68897374049a3 : SUCCESS in 48m 31s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0e295f76c7e346ada5aa83c29393b103 : SUCCESS in 48m 28s\n- loci-heat https://zuul.opendev.org/t/openstack/build/3efe1cdfa949442bbda8b0457cb123bd : SUCCESS in 45m 17s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/650481a4a2c14699b8aa9d240be1b1f3 : SUCCESS in 50m 50s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/11dfe063b8d7464aa312627e0ce31bcc : SUCCESS in 45m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/28b6870351a1459b9ab4c5c62750915f : SUCCESS in 45m 48s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b3e8fa7b56c44efea87c1e631cb87501 : SUCCESS in 44m 28s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7d05598c1bff4cb499f057bdec171aa9 : SUCCESS in 48m 01s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e5f0794e96634b589289770038866ed0 : SUCCESS in 51m 33s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7183b6ac627a47869fdb9e6ee8ef01a3 : SUCCESS in 45m 06s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/48f71c64a5b4414491613d4a574015f1 : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":9},{"id":"b97ac7b6453a391a9225902ff27cacc9f676008e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 05:13:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"76a4b299ef10432479cd4566f1ec4dcc810600b0","revisions":{"ce97a02d43908c6321c05f8cdf0653f7e1966df0":{"kind":"REWORK","_number":1,"created":"2020-03-13 09:48:13.000000000","uploader":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"ref":"refs/changes/91/712891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/91/712891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/91/712891/1"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:44:20.000000000","tz":480},"committer":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:48:05.000000000","tz":480},"subject":"Change Loci Dockerfile build method from python to python3. Fix an urllib2.urlopen problem.","message":"Change Loci Dockerfile build method from python to python3.\nFix an urllib2.urlopen problem.\n\nChange-Id: I2b60a81168f5c47eab316d2be31a4fe3ba9c2254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ce97a02d43908c6321c05f8cdf0653f7e1966df0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ce97a02d43908c6321c05f8cdf0653f7e1966df0"}]},"branch":"refs/heads/master"},"cb7c268f2e1c675b6cc0b6ac8b7a650170054029":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-13 09:54:13.000000000","uploader":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"ref":"refs/changes/91/712891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/91/712891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/91/712891/2"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:44:20.000000000","tz":480},"committer":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:51:32.000000000","tz":480},"subject":"Change Loci Dockerfile ARG for Python3.","message":"Change Loci Dockerfile ARG for Python3.\n\nsiwtch to python3 to build docker image for master version of openstack\nand fix an urllib2.urlopen problem.\n\nChange-Id: I2b60a81168f5c47eab316d2be31a4fe3ba9c2254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cb7c268f2e1c675b6cc0b6ac8b7a650170054029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/cb7c268f2e1c675b6cc0b6ac8b7a650170054029"}]},"branch":"refs/heads/master"},"a515aaab90ca551c735146102732f48efebe34ce":{"kind":"REWORK","_number":3,"created":"2020-03-27 07:24:08.000000000","uploader":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"ref":"refs/changes/91/712891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/91/712891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/91/712891/3"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:44:20.000000000","tz":480},"committer":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-27 07:23:37.000000000","tz":480},"subject":"Add an urlopen exception for ImcompleteRead error.","message":"Add an urlopen exception for ImcompleteRead error.\n\nUsing urllib2.request.urlopen() from python3 will meet the IncompleteRead error.\nAdd an exception to avoid it.\n\nChange-Id: I2b60a81168f5c47eab316d2be31a4fe3ba9c2254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a515aaab90ca551c735146102732f48efebe34ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a515aaab90ca551c735146102732f48efebe34ce"}]},"branch":"refs/heads/master"},"83ed9af9444446111e3650531e474831e25e424a":{"kind":"REWORK","_number":4,"created":"2020-03-27 07:30:20.000000000","uploader":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"ref":"refs/changes/91/712891/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/91/712891/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/91/712891/4"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:44:20.000000000","tz":480},"committer":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-27 07:29:58.000000000","tz":480},"subject":"Add an urlopen exception for ImcompleteRead error.","message":"Add an urlopen exception for ImcompleteRead error.\n\nUsing urllib2.request.urlopen() from python3 will meet the IncompleteRead error.\nAdd an exception to avoid it.\n\nChange-Id: I2b60a81168f5c47eab316d2be31a4fe3ba9c2254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/83ed9af9444446111e3650531e474831e25e424a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/83ed9af9444446111e3650531e474831e25e424a"}]},"branch":"refs/heads/master"},"0476dd069ee982314b19ba922bcc15f388c0fdea":{"kind":"REWORK","_number":5,"created":"2020-03-27 07:41:44.000000000","uploader":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"ref":"refs/changes/91/712891/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/91/712891/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/91/712891/5"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:44:20.000000000","tz":480},"committer":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-27 07:36:23.000000000","tz":480},"subject":"Add an urlopen exception for ImcompleteRead error.","message":"Add an urlopen exception for ImcompleteRead error.\n\nUsing urllib2.request.urlopen() from python3 will meet the IncompleteRead error.\nAdd an exception to avoid it.\n\nChange-Id: I2b60a81168f5c47eab316d2be31a4fe3ba9c2254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0476dd069ee982314b19ba922bcc15f388c0fdea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0476dd069ee982314b19ba922bcc15f388c0fdea"}]},"branch":"refs/heads/master"},"5204057644a443af4da02e4158eb35f878fece23":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-03-27 14:39:01.000000000","uploader":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"ref":"refs/changes/91/712891/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/91/712891/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/91/712891/6"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:44:20.000000000","tz":480},"committer":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-27 14:37:13.000000000","tz":480},"subject":"Add an urlopen exception for ImcompleteRead error.","message":"Add an urlopen exception for ImcompleteRead error.\n\nUsing urllib2.request.urlopen() from python3 will meet the IncompleteRead error.\nAdd an exception to avoid it.\n\nStory: 2007474\nTask: 39180\n\nChange-Id: I2b60a81168f5c47eab316d2be31a4fe3ba9c2254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5204057644a443af4da02e4158eb35f878fece23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5204057644a443af4da02e4158eb35f878fece23"}]},"branch":"refs/heads/master"},"0dd498b24601d5471d6db8eaa15c74e42e76b8bf":{"kind":"REWORK","_number":7,"created":"2020-03-30 15:54:44.000000000","uploader":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"ref":"refs/changes/91/712891/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/91/712891/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/91/712891/7"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:44:20.000000000","tz":480},"committer":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-30 15:52:17.000000000","tz":480},"subject":"Add an urlopen exception for ImcompleteRead error.","message":"Add an urlopen exception for ImcompleteRead error.\n\nUsing urllib2.request.urlopen() from python3 will meet the IncompleteRead error.\nAdd an exception to avoid it.\n\nStory: 2007474\nTask: 39180\n\nChange-Id: I2b60a81168f5c47eab316d2be31a4fe3ba9c2254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0dd498b24601d5471d6db8eaa15c74e42e76b8bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/0dd498b24601d5471d6db8eaa15c74e42e76b8bf"}]},"branch":"refs/heads/master"},"e5809d5e753d55ef28a52d26c6bff404cadedb3e":{"kind":"NO_CHANGE","_number":8,"created":"2020-03-30 16:02:11.000000000","uploader":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"ref":"refs/changes/91/712891/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/91/712891/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/91/712891/8"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:44:20.000000000","tz":480},"committer":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-30 16:01:22.000000000","tz":480},"subject":"Add an urlopen exception for ImcompleteRead error.","message":"Add an urlopen exception for ImcompleteRead error.\n\nUsing urllib2.request.urlopen() from python3 will meet the IncompleteRead error.\nAdd an exception to avoid it.\n\nStory: 2007474\nTask: 39180\n\nChange-Id: I2b60a81168f5c47eab316d2be31a4fe3ba9c2254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e5809d5e753d55ef28a52d26c6bff404cadedb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/e5809d5e753d55ef28a52d26c6bff404cadedb3e"}]},"branch":"refs/heads/master"},"76a4b299ef10432479cd4566f1ec4dcc810600b0":{"kind":"REWORK","_number":9,"created":"2020-04-23 00:13:03.000000000","uploader":{"_account_id":30728,"name":"YU CHENGDE","display_name":"Chant Yu","email":"chengde.yu@intel.com","username":"yuchengde"},"ref":"refs/changes/91/712891/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/91/712891/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/91/712891/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/91/712891/9"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-03-13 09:44:20.000000000","tz":480},"committer":{"name":"yuchengde","email":"yu.chengde@99cloud.net","date":"2020-04-23 00:12:36.000000000","tz":480},"subject":"Add an urlopen exception for ImcompleteRead error.","message":"Add an urlopen exception for ImcompleteRead error.\n\nUsing urllib2.request.urlopen() from python3 will meet the IncompleteRead error.\nAdd an exception to avoid it.\n\nStory: 2007474\nTask: 39180\n\nChange-Id: I2b60a81168f5c47eab316d2be31a4fe3ba9c2254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76a4b299ef10432479cd4566f1ec4dcc810600b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/76a4b299ef10432479cd4566f1ec4dcc810600b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
