)]}'
{"id":"openstack%2Floci~714199","triplet_id":"openstack%2Floci~master~If180168fb2f60ecc4c7da3074409ea2ae3d354d8","project":"openstack/loci","branch":"master","topic":"stein/placement","hashtags":[],"change_id":"If180168fb2f60ecc4c7da3074409ea2ae3d354d8","subject":"Enable placement service gate","status":"MERGED","created":"2020-03-20 19:30:03.000000000","updated":"2020-07-25 21:56:04.000000000","submitted":"2020-07-25 21:56:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714199-1595714164449-c002e016","meta_rev_id":"87456662b2f1e4d4f3501e33e052f2d4ac9262b0","_number":714199,"virtual_id_number":714199,"owner":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 21:56:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"recommended":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":1,"date":"2020-07-25 20:25:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":1,"date":"2020-07-25 20:25:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":1,"date":"2020-07-25 20:25:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-25 20:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2020-07-25 20:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":1,"date":"2020-07-25 20:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 10:51:10.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-05-20 23:34:31.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2020-07-03 06:19:02.000000000","updated_by":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"reviewer":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"state":"REVIEWER"},{"updated":"2020-07-24 04:39:20.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-07-25 20:26:01.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2020-07-25 21:56:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca2108e2a40b88b2a8b92a6f6e181c1f65c4132c","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-20 19:30:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c22ac19a01f1aeadc77060d9026efbd19dcc8e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 20:17:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a3afb886bbd648bbbf84bdb3890a7d4e : FAILURE in 35m 44s\n- loci-glance https://zuul.opendev.org/t/openstack/build/59efe251aeae404aa7e50993420f030f : FAILURE in 37m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9b73268b26194599b72cc901091d7ce8 : FAILURE in 34m 30s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4d5893d6a4ff4dd28aa8daec054a90d5 : FAILURE in 34m 38s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d355151560de42dcb58cda7da33d6637 : FAILURE in 33m 59s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/17803e39b8e54cee847c11b418c3cac8 : FAILURE in 34m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/42eb107fe0a64ce4882f44d5bf8cb48e : FAILURE in 36m 13s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/cb0d0338e3ee4cf998dcf4a234bf5b3d : FAILURE in 34m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d9ab2b8bb230460ea47fa00da95af508 : FAILURE in 41m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a14db617cbc74d6e83aade96bc9c9083 : FAILURE in 33m 24s\n- loci-placement https://zuul.opendev.org/t/openstack/build/554122cf9f2944c887087d721020893c : FAILURE in 28m 34s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d631cf57011c448c821f6ccdc36416c0 : FAILURE in 32m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ee94c324c260cf7d91f4257d22a2324605bbd8bc","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-20 21:17:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"abf211eba3d4835fdfc604c44102e26008c32230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 22:06:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4389e786bdd14bfa94133d7d1406629f : FAILURE in 47m 49s\n- loci-glance https://zuul.opendev.org/t/openstack/build/79d801d656a04c59bbdcf47298bfaebc : FAILURE in 32m 37s\n- loci-heat https://zuul.opendev.org/t/openstack/build/05ec52ea33064d1ab398cf3d2091784e : FAILURE in 46m 23s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f7d435ab33224e95a0bc41c9b067d79c : FAILURE in 33m 03s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b71bb87a0a354a49858396002ae0e6e8 : FAILURE in 47m 06s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e02fe420b7e9445e8a436f574cfa6bb6 : FAILURE in 34m 05s\n- loci-manila https://zuul.opendev.org/t/openstack/build/099c24f744144871850a5a88c20f08d9 : FAILURE in 40m 48s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8764fb752fd54b1a8d98fa33d6b758b9 : FAILURE in 41m 14s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fd9b30331ce74561a117d6dfc9d3f8ab : FAILURE in 35m 58s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4dad170e253d41a48d16881fbb675537 : FAILURE in 46m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/40669260b4884acbb04999ad1541829b : FAILURE in 27m 24s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5af14675e725437780c65c94355c6023 : TIMED_OUT in 41m 00s","accounts_in_message":[],"_revision_number":1},{"id":"674dfe40ae1d9a701589095865e296607a0265da","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-21 01:41:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"406caa6abb5e4209b06aa05fc68f15a83ffe6054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 02:25:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/db8d2f37afae4f97a64d00e8f65b1917 : FAILURE in 38m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a88cb579ce244be0b573fcf4ea5f557d : FAILURE in 31m 47s\n- loci-heat https://zuul.opendev.org/t/openstack/build/a56a7d36258041968262c35d09c80a2d : FAILURE in 35m 29s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/0ce90462392d4a18b14dd8e0120fa761 : FAILURE in 31m 49s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/6b635a284ccd414d84b9948a58987e25 : FAILURE in 33m 35s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a0fa02e9b47f4acea4fd241316826b34 : FAILURE in 28m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/fddba3cc052645028e8440feefa26642 : FAILURE in 39m 14s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/866ea125b98142b898c6ab35354dfe12 : FAILURE in 29m 02s\n- loci-nova https://zuul.opendev.org/t/openstack/build/cdda8b68e4294cdbbd90b0580e4c3d0d : FAILURE in 39m 02s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3b9d10a9c80a45e19a91214480b66864 : FAILURE in 37m 08s\n- loci-placement https://zuul.opendev.org/t/openstack/build/fa69483ca89741968f4743bd7b0464d3 : FAILURE in 34m 36s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8467e32bc38d4795adca5a64bf5aa425 : FAILURE in 28m 47s","accounts_in_message":[],"_revision_number":1},{"id":"18ebc60f555c7f2eb80d11664d0619bd0f2ae1cb","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-21 17:25:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5a793f7b49222a7c7cb4c659afbe81c302a7284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 18:12:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/417d0ed2163b4626b02d6a158d94e16c : FAILURE in 44m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8b5622c468b34201b2c21ddfd36a6081 : FAILURE in 33m 29s\n- loci-heat https://zuul.opendev.org/t/openstack/build/798df7d33e774064b4e2fae7e98d6910 : FAILURE in 45m 32s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/421663cad5cf4f8bb6529cb313e19435 : FAILURE in 30m 13s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/3f7f32f521d744db905f4a3345043c50 : FAILURE in 45m 28s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6f69a20253f24f8da6ce90a2cd0381da : FAILURE in 44m 56s\n- loci-manila https://zuul.opendev.org/t/openstack/build/cd848d02f25d407588856334dc122471 : FAILURE in 46m 18s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/61082e27c139492ea98b5c4b57a545ef : FAILURE in 46m 16s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bdcede1e2c1c4b44a9cdb82257df3bb9 : FAILURE in 30m 02s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3b10ad79c3a64098a5dd97d8102bb38c : FAILURE in 29m 47s\n- loci-placement https://zuul.opendev.org/t/openstack/build/13673c54595443a4896a0f75fc8cb798 : FAILURE in 27m 52s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/76b31068a4094838b187055eed1302b7 : FAILURE in 33m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5350573be1d1832a069e952876e105b2f718eff8","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-23 18:12:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95e93605091e9b1ff511d1e141c5819a3ab383fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:04:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/7dc5ab4e18814753af2b1a29e91426fb : FAILURE in 30m 20s\n- loci-glance https://zuul.opendev.org/t/openstack/build/f15ccc878d8d47ebb4c658dadc042814 : FAILURE in 46m 38s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9883b894aea9434e87d9542149f2c265 : FAILURE in 34m 11s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2b0451a176ae443ea13a275ea41691d2 : FAILURE in 44m 59s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8c186e2a10f3489faec9e006ac70cceb : FAILURE in 40m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d45d1f64946543729560de1fe2c31ba7 : FAILURE in 46m 24s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a4b248aaafc94493ad15c163a6cd9cc3 : FAILURE in 43m 09s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/dcdc92790b334306a8fe56188a419b3a : FAILURE in 35m 02s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d57d953ad47345e9887779b0bd732390 : FAILURE in 33m 03s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/cb6c50ccd58d4a3687710ba9e0ed2a6c : FAILURE in 45m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/20c0df7daba34c8da550ef6ec06d7060 : FAILURE in 30m 36s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/34e58606f7024d849ae9ffcb93943f67 : FAILURE in 37m 55s","accounts_in_message":[],"_revision_number":1},{"id":"82fb797d30bb283334df68f1762c623d96defd7e","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-24 10:39:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e6eddbcd3cf1d47656b4b263e68fcb0d6f23b92f","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-24 10:40:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"57dfa8986eadd0041793e14b93b52a8af6004149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 11:46:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/109174628f2e4d559577355f4d73e012 : SUCCESS in 49m 54s\n- loci-glance https://zuul.opendev.org/t/openstack/build/695a169e09d64ae6a5028175ba8b4ad3 : SUCCESS in 42m 03s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e8cb4c0714ad49e2ac37230176af0fd1 : SUCCESS in 47m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/fdbd809c2b3c466d92b8bf706b7957fc : SUCCESS in 45m 34s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/48033e41bad34da48e74d209fb53292b : SUCCESS in 1h 00m 43s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c1c77964c98a488bbfa3e9d57afc0bbf : SUCCESS in 43m 57s\n- loci-manila https://zuul.opendev.org/t/openstack/build/dbb91e6cc5804bafb7b76a59d4b707d3 : SUCCESS in 43m 09s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/3fd64506b23449849bc50dc4d6bb0a23 : SUCCESS in 50m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/49c1538306e34effb06601b24f41ef1d : TIMED_OUT in 1h 00m 46s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3f606b3fb4f74ba7889c893ca3871e9e : SUCCESS in 42m 40s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3b1137a0cf004e2fb95ae0ff34c95d12 : SUCCESS in 40m 29s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/496cca4673114e9b8d74c475c62bdd97 : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":2},{"id":"85352716822e07ebe6355797eaa96f0bb496f089","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-24 11:51:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"66ec742a5bb34ce393fe9282fed2f1c51a71ad29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 12:48:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/cde7682fd2ff4fff9e1c432162da0993 : SUCCESS in 39m 06s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d400ee03864f461cbee7ee75ed65a9ae : SUCCESS in 40m 45s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9aa2f696e37d4888899b363d2fab577e : SUCCESS in 39m 23s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2f8a0a77bcb541dd81b0fba83d9ef976 : SUCCESS in 40m 20s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/6496bb5a74ed472782fbc3903f9056a7 : SUCCESS in 42m 18s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/040f806d0180407bac43dcc8acbe2d8e : SUCCESS in 45m 41s\n- loci-manila https://zuul.opendev.org/t/openstack/build/bc6ca14b3b0c4ffab541668b86eed6e6 : SUCCESS in 45m 50s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6b4e976b87fd490bbcf99d73436babe1 : SUCCESS in 52m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c9f727fcf9124a6489e798cdc10b50ed : SUCCESS in 49m 22s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e79b9767562447e69ffec10f1e5798c4 : SUCCESS in 40m 40s\n- loci-placement https://zuul.opendev.org/t/openstack/build/513001a334844960bb7b3e86dc2c3d11 : RETRY_LIMIT in 11m 57s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/99da3b6c453040efb9c8466f7cc865e7 : TIMED_OUT in 41m 11s","accounts_in_message":[],"_revision_number":2},{"id":"989d29f0b7b02b03f34c6264af695b47f6d714a5","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-03-24 13:53:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4a9a23a8f85b39f66072037cee546fbc3a05ac11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 14:46:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ccc35dea85b743c1bf56b157df9267a3 : SUCCESS in 46m 53s\n- loci-glance https://zuul.opendev.org/t/openstack/build/17bcd999240340eb90f9a340f979a6e9 : SUCCESS in 39m 16s\n- loci-heat https://zuul.opendev.org/t/openstack/build/fe434943ddd448eb8d6039ee6ec81be1 : SUCCESS in 41m 29s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4fe820236df64629a9a1e160aef6c1bc : SUCCESS in 49m 35s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8d84fee38e004a22add0259e30c8dd4c : SUCCESS in 47m 54s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e68dc4a689ab4139a640fee1809e3747 : SUCCESS in 46m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b8a5757926ea4ec59d8f9638ae0f135d : SUCCESS in 37m 54s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a4fc1ecf211a474ba0169da04e21a195 : SUCCESS in 40m 31s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bfd08f825aef4bb7a627686699c1028b : SUCCESS in 46m 49s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/605b20406dd84920a78f477f937b2c5c : SUCCESS in 41m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5f28ef7f9dcc4b3fbc21b943a832f132 : SUCCESS in 35m 00s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/3cb7e9bb5cd544d9bdc15f6a658205c6 : SUCCESS in 32m 08s","accounts_in_message":[],"_revision_number":2},{"id":"a0099d1566aca46200190ca1cfe76f3fdbbcf38c","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-04-01 04:50:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ea3f826f85bc2dfa73879cdf5364058ddfcbefac","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2020-04-22 17:05:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3fd285c65ffd484b7bee49225a736a1e9e5fd274","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-04-24 11:02:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"881db6c37f14756f5015ea2be6de20e06d01d4b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 12:02:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/81f0e5751bd84e7f935e3f1c14207404 : SUCCESS in 47m 10s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e8048b0774c54635952d9b9ae9219406 : SUCCESS in 44m 25s\n- loci-heat https://zuul.opendev.org/t/openstack/build/350a2c4810bd4373b0ba6e94ed23c971 : SUCCESS in 46m 59s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/8ed0eabb82b14fbb94671f270733e846 : SUCCESS in 46m 04s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a6b6d018ca184815844119ada77f2552 : SUCCESS in 41m 14s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/90f6022e15424d7783c6908c7ba6493a : SUCCESS in 42m 21s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5bdaf3922f2a41ac8f6aa3f05a29b61d : SUCCESS in 55m 37s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/fe0cfc18a215428eb6997e9dabbf45af : SUCCESS in 42m 50s\n- loci-nova https://zuul.opendev.org/t/openstack/build/184527e156174691898a59f4664c54a1 : SUCCESS in 58m 51s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/d5398045692a4a6dab25ed475ef28d04 : SUCCESS in 48m 29s\n- loci-placement https://zuul.opendev.org/t/openstack/build/d7be41a5649042ae94662bbed696e9c9 : SUCCESS in 47m 02s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/000d06535baf47e688a0d2bc851c41fd : TIMED_OUT in 41m 05s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2d9a47ab02e145918cf9f79f65345f2f : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":2},{"id":"fc39448bcd5e6d64916f5e06352f53fdb3613e3d","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-05-01 12:31:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5d27c1ddb2e0cdf34a3ec762d6b3452b2309e871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 13:33:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d4abe37837e54261817e7a78095c5644 : SUCCESS in 56m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8202a99945f04ca8b9a08a5244e120c6 : TIMED_OUT in 1h 01m 01s\n- loci-heat https://zuul.opendev.org/t/openstack/build/694a5ee7a0b640eab8b62e4e0ff680c9 : SUCCESS in 49m 59s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/99d959c967014c74a31a92ccf18f1370 : TIMED_OUT in 1h 00m 58s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/685264961abf428ea84aa1930d7c0be4 : TIMED_OUT in 1h 00m 53s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6629a1e0bb1446cd9e19d2d7389e8e07 : TIMED_OUT in 1h 01m 20s\n- loci-manila https://zuul.opendev.org/t/openstack/build/51f1be7e80324a98bc3d8ce9a583f458 : TIMED_OUT in 1h 01m 04s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4ed5f13cc3d14ba8a771e95fd108ea00 : SUCCESS in 53m 03s\n- loci-nova https://zuul.opendev.org/t/openstack/build/63cd253e91a04b85ab8028eda77d1709 : SUCCESS in 53m 38s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/6ea8087213f44988a268ded7d846c9be : SUCCESS in 49m 42s\n- loci-placement https://zuul.opendev.org/t/openstack/build/fd0c13aa5dc14a25ada929d70cc31034 : SUCCESS in 49m 40s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a55f7bce5498493a931f4fc229b3e55d : SUCCESS in 39m 42s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c9580b0e73d0436b8bec847b1a0dff5b : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":2},{"id":"dc8341b5d027f170e6900d2b7555cc4eda05191a","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-05-01 13:34:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d50796c2f89cfc3106436cbb6880b83925c2db85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 14:34:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/441e7c42bb09400d96f8747f0872fde9 : SUCCESS in 53m 08s\n- loci-glance https://zuul.opendev.org/t/openstack/build/275e14687f454da68423965c06f00e35 : SUCCESS in 50m 39s\n- loci-heat https://zuul.opendev.org/t/openstack/build/85e98653217d4690b56c0c23c9df97d5 : SUCCESS in 53m 34s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/061acd10c5ab454ab74e0813040a6e38 : SUCCESS in 51m 14s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/cd0b0272d63f4a19860ed19560257667 : SUCCESS in 55m 10s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a27c6096fb9449a9863c50ff715efd27 : SUCCESS in 51m 29s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7040fccd8f674dec9007e05a25d5f387 : SUCCESS in 52m 00s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f71e4e4f35c14c4cb31d1c27a3aa9a9a : SUCCESS in 52m 05s\n- loci-nova https://zuul.opendev.org/t/openstack/build/cce5b705a0d34a06ba999a2e4e24d400 : SUCCESS in 53m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/d7f75d3f676148ee90e626d48dd4a81c : SUCCESS in 50m 56s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8a0c491f4a5748e2a4d8eba10d19cc5f : SUCCESS in 43m 07s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ae34c52bcb2e45afb92e143ea0482628 : TIMED_OUT in 41m 18s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c5c58758bb4442a19273a7aa805fde3f : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2e2cbe525a274eb974181af64eb4209ba5da1a89","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-05-03 09:08:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a7e5962fe8265a9dc02bd14ded42363c56a107db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 10:04:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d58067f8169d4eae90ec2f7ac24c7c03 : SUCCESS in 44m 52s\n- loci-glance https://zuul.opendev.org/t/openstack/build/aea2fd6859cc451c9715ffed7f83c00a : SUCCESS in 41m 06s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d3049d1171624ceb9f71a2e35a178d64 : SUCCESS in 51m 20s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/fae4571ad5e3484d842d0b2917d92ecf : SUCCESS in 50m 57s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/c24d99910e9540fe83e2840e0c460438 : SUCCESS in 44m 25s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f6f2bbd394b84f1a9e03cc9cd59257d6 : SUCCESS in 40m 04s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0d108989184e424a9bb8c6b48801adf2 : SUCCESS in 43m 25s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/1e6b35c272ca44119efde12023605d7e : SUCCESS in 44m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ec559632f9b9469db0c0ae65e4d9248c : SUCCESS in 51m 18s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e3918930cce446209127128a5ca939ef : SUCCESS in 46m 06s\n- loci-placement https://zuul.opendev.org/t/openstack/build/920904239b0d41ce9e904c181b0b623a : SUCCESS in 40m 03s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d5ed8f4fe76247bfa7fa878c4eea09f2 : SUCCESS in 39m 14s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/87c3013e69cd4276895bc9c9a617739b : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":2},{"id":"91f1b82482ebe69a9ed71f4cfc15d560c5a8f64b","author":{"_account_id":7822,"name":"Chris Hoge (not involved in upstream development)","email":"chris@hogepodge.com","username":"hogepodge"},"date":"2020-05-14 08:44:02.000000000","message":"Removed reviewer Chris Hoge (not involved in upstream development) with the following votes:\n\n* Code-Review+2 by Chris Hoge (not involved in upstream development) \u003cchris@hogepodge.com\u003e\n* Workflow+1 by Chris Hoge (not involved in upstream development) \u003cchris@hogepodge.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"642b03ff253df119fdfb0e0c928d233f9a36b112","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-15 10:51:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cc6ffcf03d24128c98fc6d0fa6a6a5ee5acf81a2","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-05-20 18:04:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8a1605730aa93c08b1733ebea88af9b47ebc62dc","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-05-20 23:34:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a5737e846427055b194d62f270f73d0cc8db8ab7","author":{"_account_id":29911,"name":"Dantali0n","email":"info@dantalion.nl","username":"Dantali0n"},"date":"2020-07-03 06:19:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"313bc907526a8cb80471a22d07550501792d3777","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-07-13 18:33:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a6033d939fb8873639efc998c00f4c8be73dfc72","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-07-24 04:39:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"47b37b3d2c8ef0b1848ee4fb3c4b5e296bdbdb74","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-07-25 02:31:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"52d6ce165c9ec91d2ca85302f1ee1b44b3911fbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 03:19:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c798985c27d143979b0c726541fdb991 : SUCCESS in 39m 44s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bde0778241084ed6aa4de86764558034 : SUCCESS in 36m 19s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9771a684f5654ab680995e966d404434 : SUCCESS in 38m 23s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a9babf4cf3f84f3bbec0776857aeb44d : SUCCESS in 40m 30s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/7513fb2938474c819cc2d125a8f813ba : SUCCESS in 37m 17s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e4743cd6f6bc42b8949282161f345976 : SUCCESS in 36m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/7a3738d80de74eb989f3d009bea29597 : POST_FAILURE in 37m 18s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/443b4796b4204af290f5263c1bf26ffc : SUCCESS in 34m 40s\n- loci-nova https://zuul.opendev.org/t/openstack/build/621c2bd53e4a413ba472a1eda127d292 : SUCCESS in 42m 48s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/fcc2bb8e2cf24bc0a67c5949c634499d : SUCCESS in 37m 23s\n- loci-placement https://zuul.opendev.org/t/openstack/build/6caaff59e52c493a806df1b819d47c9b : SUCCESS in 31m 35s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ecba5acbfa80425bb7278a17f49c91de : SUCCESS in 26m 01s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/c1c6260a02774d30bb43904ccfd75611 : SUCCESS in 37m 57s","accounts_in_message":[],"_revision_number":2},{"id":"02d4ec9ed72722dbb23340326c262d453dd2ba46","author":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"date":"2020-07-25 17:05:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6c305d31f4f7491db34ee782e5a4edb77c051ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 17:51:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4508054f8c8b4f2bb34bebf2de3b1e6c : SUCCESS in 40m 53s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7fcba8268152421c8ecdf7d3b0d06d99 : SUCCESS in 40m 07s\n- loci-heat https://zuul.opendev.org/t/openstack/build/cd073ca021c3452cb5e608d3ab948d95 : SUCCESS in 38m 37s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/644477a748ef42e6932866fb3a3356b1 : SUCCESS in 43m 44s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/fa1a7970e17d4683a2270de4024cd692 : SUCCESS in 40m 48s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/33020cd44c7544efafc42d2bd38b59a9 : SUCCESS in 36m 14s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c41fac78a7934763bcc6d879946678dd : SUCCESS in 34m 38s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/98ade0ae4ed64716a9826c1f29a024f3 : SUCCESS in 41m 29s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0e13d1017f1b4a2398f97de93b896909 : SUCCESS in 41m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5c64a1b2e1e74d498cee8e398501f62f : SUCCESS in 39m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/fba9dfc780314e6982917c5549064954 : SUCCESS in 31m 08s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1a3934a9ba9342b1a2910444cc7e3cb1 : SUCCESS in 26m 30s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d4d980d8b1a843f0aa80abc1b41df514 : SUCCESS in 35m 00s","accounts_in_message":[],"_revision_number":2},{"id":"32b336b91a071a8f321e13195f5d58fb50a9cda8","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-07-25 20:24:56.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"02fcd8ed863d58467d1a235b72f0e3e91dfa4300","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-07-25 20:25:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d550056c4c838b31e0f718daba19e8b62fa8b28b","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-07-25 20:25:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f74638fd72b2ec9ace299ae32dcf225d82926d3a","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-07-25 20:26:01.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"392b63b1fce8f349505dc8adb5de8685511868dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 21:09:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ffa70a6a89a0488d93510a052c009dde : SUCCESS in 39m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7c061f6b825b40528fbbaceedbe4ef86 : SUCCESS in 39m 03s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9ab1d43351634db6a2ef96815122f358 : SUCCESS in 33m 27s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3013c8bcd85749eabba4e78bbb201756 : SUCCESS in 36m 41s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9b8210c51cd849fb86861310f2e05967 : SUCCESS in 33m 25s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/7630d80a0a5d4bb4b003fc84b96980f6 : SUCCESS in 36m 25s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9ca1e577190249eeb7315129a66d7dea : SUCCESS in 36m 23s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6c3d393afbef40119bf22b516ade9efa : SUCCESS in 42m 31s\n- loci-nova https://zuul.opendev.org/t/openstack/build/900335a613b043b9b9158142b36847aa : SUCCESS in 36m 53s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/91309b82b39444d6b304024bfe3aab74 : SUCCESS in 32m 09s\n- loci-placement https://zuul.opendev.org/t/openstack/build/febafd564b6148ae8a8fa95bf1f5703c : SUCCESS in 30m 13s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1435c8c27c90414e8e17860d3b99da8a : SUCCESS in 29m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/7d50b24e1ee24ee0ba454812fe7fa4ba : SUCCESS in 38m 18s","accounts_in_message":[],"_revision_number":4},{"id":"1d4c92f7303b8cdcd437cfafc6b0f7284b50feb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 21:09:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ab3d747e7cd5211cffdc80ebf8b166f511e7c70d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 21:56:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"f25f27c7c8d07cb73c6c7e04b1e0eeca2d766ff5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 21:56:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/89411f9a512347ac8ce352cc5b6fb07c : SUCCESS in 40m 31s\n- loci-glance https://zuul.opendev.org/t/openstack/build/efe35f7f5db6442786f5f333dc40136a : SUCCESS in 38m 18s\n- loci-heat https://zuul.opendev.org/t/openstack/build/cddca54b0d594516ad3a0914249cddac : SUCCESS in 39m 54s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ff50edd253fe4ca8aa3194613336df8b : SUCCESS in 38m 30s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/7d0d1a2212be4c3d835fd2ce9cd8ab9f : SUCCESS in 28m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/330fe77c2cee4c39922722315b41380c : SUCCESS in 36m 35s\n- loci-manila https://zuul.opendev.org/t/openstack/build/6b1958b5dd9846699c9f1e5fac941724 : SUCCESS in 30m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ce76809eda824ef1922ae284a7ee8684 : SUCCESS in 41m 01s\n- loci-nova https://zuul.opendev.org/t/openstack/build/8a4ebec179d14f8e86d48e44cc2509a2 : SUCCESS in 38m 09s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5e0dbdc0db52472ab255be05f56c503c : SUCCESS in 39m 13s\n- loci-placement https://zuul.opendev.org/t/openstack/build/9618ba31ec8c4887a5a2a9234cd47fd1 : SUCCESS in 24m 09s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ac9ce73431e6459f98809a46e62dc418 : SUCCESS in 24m 52s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/9829ace5cede4dbeac941fc8c9c0cc66 : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"baa006471c7bc4c929ddce79370997a8d48ac767","revisions":{"4bf72c75ead88256d62f6c8dbbff543a6903dd40":{"kind":"REWORK","_number":1,"created":"2020-03-20 19:30:03.000000000","uploader":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"ref":"refs/changes/99/714199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/99/714199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/99/714199/1"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-03-20 19:28:58.000000000","tz":-300},"committer":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-03-20 19:30:02.000000000","tz":-300},"subject":"Enable placement service gate","message":"Enable placement service gate\n\nThis patch set makes the placement service voting and publish the\nplacement image for OpenStack releases: stein, train and master.\n\nChange-Id: If180168fb2f60ecc4c7da3074409ea2ae3d354d8\nSigned-off-by: Tin Lam \u003ctin@irrational.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4bf72c75ead88256d62f6c8dbbff543a6903dd40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4bf72c75ead88256d62f6c8dbbff543a6903dd40"}]},"branch":"refs/heads/master"},"b44e859caad1c5d7a59fb2b88113f3bbff36169b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-24 10:40:09.000000000","uploader":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"ref":"refs/changes/99/714199/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/99/714199/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/99/714199/2"}}},"commit":{"parents":[{"commit":"5710603d631857ab290f10098b069a03f534ca70","subject":"Limit setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5710603d631857ab290f10098b069a03f534ca70"}]}],"author":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-03-20 19:28:58.000000000","tz":-300},"committer":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-03-24 10:40:09.000000000","tz":0},"subject":"Enable placement service gate","message":"Enable placement service gate\n\nThis patch set makes the placement service voting and publish the\nplacement image for OpenStack releases: stein, train and master.\n\nChange-Id: If180168fb2f60ecc4c7da3074409ea2ae3d354d8\nSigned-off-by: Tin Lam \u003ctin@irrational.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b44e859caad1c5d7a59fb2b88113f3bbff36169b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/b44e859caad1c5d7a59fb2b88113f3bbff36169b"}]},"branch":"refs/heads/master"},"d5e4051dd9e9b09a1b32704b98894c7f442f49a0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-25 20:25:34.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/99/714199/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/99/714199/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/99/714199/3"}}},"commit":{"parents":[{"commit":"c94a86c32894f9d43ac008a7079fe7663fce3b80","subject":"Add support for virtualenv \u003e 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c94a86c32894f9d43ac008a7079fe7663fce3b80"}]}],"author":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-03-20 19:28:58.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-07-25 20:25:34.000000000","tz":0},"subject":"Enable placement service gate","message":"Enable placement service gate\n\nThis patch set makes the placement service voting and publish the\nplacement image for OpenStack releases: stein, train and master.\n\nChange-Id: If180168fb2f60ecc4c7da3074409ea2ae3d354d8\nSigned-off-by: Tin Lam \u003ctin@irrational.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5e4051dd9e9b09a1b32704b98894c7f442f49a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d5e4051dd9e9b09a1b32704b98894c7f442f49a0"}]},"branch":"refs/heads/master"},"baa006471c7bc4c929ddce79370997a8d48ac767":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-25 20:25:55.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/99/714199/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/99/714199/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/99/714199/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/99/714199/4"}}},"commit":{"parents":[{"commit":"a08b229441d6d4b92acb1cefaf807c81386bdd5a","subject":"Merge \"Avoid storing cache in pip install for requirements image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a08b229441d6d4b92acb1cefaf807c81386bdd5a"}]}],"author":{"name":"Tin Lam","email":"tin@irrational.io","date":"2020-03-20 19:28:58.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-07-25 20:25:55.000000000","tz":0},"subject":"Enable placement service gate","message":"Enable placement service gate\n\nThis patch set makes the placement service voting and publish the\nplacement image for OpenStack releases: stein, train and master.\n\nChange-Id: If180168fb2f60ecc4c7da3074409ea2ae3d354d8\nSigned-off-by: Tin Lam \u003ctin@irrational.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/baa006471c7bc4c929ddce79370997a8d48ac767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/baa006471c7bc4c929ddce79370997a8d48ac767"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
