)]}'
{"id":"openstack%2Floci~737592","triplet_id":"openstack%2Floci~master~Ieea70140ba95deb6443bedc0d078a24afee4d023","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"Ieea70140ba95deb6443bedc0d078a24afee4d023","subject":"Bump version of eventlet for Rocky","status":"ABANDONED","created":"2020-06-23 18:14:53.000000000","updated":"2020-06-24 14:20:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cfceb64c8325c4c628656108aadfaab5b6130489","_number":737592,"virtual_id_number":737592,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-23 19:04:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 19:04:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13b9dbb743d81ab4f1866e082153cb256f288c57","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-06-23 18:14:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79b8229a7e487cc61ca4fbcf167a507ec674bd8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 19:04:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9961e7457296463b9966a6e80095714e : SUCCESS in 37m 20s\n- loci-glance https://zuul.opendev.org/t/openstack/build/87f75df296b84bf7b2fcd2ba6d014562 : SUCCESS in 39m 30s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ba946f54fac34364ae8e5c8fcdba2c7d : SUCCESS in 43m 13s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2f2f6295c51b449ba723802738cf9556 : SUCCESS in 40m 46s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/acae91bd569f4bbba2faafb16eecb5e7 : SUCCESS in 36m 27s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/cc5a27ff3ac942a0a2646fd2176b7a0e : SUCCESS in 34m 56s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a7287e8468e148fdb3e10e74fca79b61 : SUCCESS in 33m 00s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e8e5798775124d59911a0d97b789ada2 : SUCCESS in 35m 32s\n- loci-nova https://zuul.opendev.org/t/openstack/build/82a6a883645b457282c010a3db3d15bd : SUCCESS in 22m 50s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4ee18992036b4fbe91b67768f2a88820 : SUCCESS in 18m 27s\n- loci-placement https://zuul.opendev.org/t/openstack/build/fcd7635725de49bb9be471d239b93d14 : SUCCESS in 30m 24s (non-voting)\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f0f940d818834e1ba2af0a5eaabe9543 : SUCCESS in 28m 00s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/502d913dec714a37bb39183994f42880 : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":1},{"id":"cfceb64c8325c4c628656108aadfaab5b6130489","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-06-24 14:20:48.000000000","message":"Abandoned\n\nwill be waiting for https://review.opendev.org/#/c/737651/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"732beb1834bf581dcc606a7ca3d2fe3eda2286db","revisions":{"732beb1834bf581dcc606a7ca3d2fe3eda2286db":{"kind":"REWORK","_number":1,"created":"2020-06-23 18:14:53.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/92/737592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/92/737592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/92/737592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/92/737592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/92/737592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/92/737592/1"}}},"commit":{"parents":[{"commit":"d6b2e7aaee270e16d1de9f14d3714cb726d1e645","subject":"Allow to specify eol release for project and fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d6b2e7aaee270e16d1de9f14d3714cb726d1e645"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-06-23 18:13:17.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-06-23 18:13:17.000000000","tz":-300},"subject":"Bump version of eventlet for Rocky","message":"Bump version of eventlet for Rocky\n\nTemporary fix to address https://github.com/eventlet/eventlet/issues/371\nfor Rocky and py36. While https://review.opendev.org/#/c/737584/ or\nalternative is not available.\n\nChange-Id: Ieea70140ba95deb6443bedc0d078a24afee4d023\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/732beb1834bf581dcc606a7ca3d2fe3eda2286db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/732beb1834bf581dcc606a7ca3d2fe3eda2286db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
