)]}'
{"id":"openstack%2Floci~745209","triplet_id":"openstack%2Floci~master~Ia1c4774aede69f83e1f43c7b511e176edc10cc9e","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"Ia1c4774aede69f83e1f43c7b511e176edc10cc9e","subject":"Fix KEEP_ALL_WHEELS flag usage","status":"MERGED","created":"2020-08-06 19:44:41.000000000","updated":"2020-08-07 16:38:53.000000000","submitted":"2020-08-07 16:38:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745209-1596818333425-480d4010","meta_rev_id":"df4bce84c96883ae179ec9d18a8bc4d15464ff91","_number":745209,"virtual_id_number":745209,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-07 16:38:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":0,"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":2,"date":"2020-08-07 15:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":2,"date":"2020-08-07 01:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"all":[{"value":0,"date":"2020-08-06 19:45:24.000000000","_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"value":1,"date":"2020-08-07 15:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"date":"2020-08-06 19:45:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-06 19:45:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 19:45:16.000000000","updated_by":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"REVIEWER"},{"updated":"2020-08-06 19:45:24.000000000","updated_by":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"reviewer":{"_account_id":20466,"name":"Tin Lam","email":"tin@lam.wtf","username":"tinlam"},"state":"REVIEWER"},{"updated":"2020-08-06 19:45:26.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2020-08-07 01:02:08.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-08-07 15:48:21.000000000","updated_by":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2020-08-07 16:38:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae73072b6e2260cc15a721e0500893edd4600977","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-08-06 19:44:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac0fe9d6eb34af13091a7e439114e9355bb85573","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-08-06 19:45:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3590a96f5f37da7bf7bc24e376d75a4bc8dc8e63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 20:34:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4bb6a4aab1024994a39c957403cfa3ed : SUCCESS in 31m 39s\n- loci-glance https://zuul.opendev.org/t/openstack/build/333d618b1caa441e9bdbf4fb860339a7 : SUCCESS in 42m 44s\n- loci-heat https://zuul.opendev.org/t/openstack/build/d17e686f6b3b42b4a6c03d1e2f66dcc2 : SUCCESS in 45m 38s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4458ee7483e847b88cedf085ae851d34 : SUCCESS in 47m 22s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b1df6248e85a45ee8f63a26332e20849 : SUCCESS in 44m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6aebbe92a401495b83fd8b5dc891a7e0 : SUCCESS in 45m 06s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a1914eaec66843a8bba566ad7d1a51cf : SUCCESS in 42m 13s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/75b4b3679a1c4fb1bdadd4a2cafb2c0a : SUCCESS in 48m 01s\n- loci-nova https://zuul.opendev.org/t/openstack/build/7b429acf5d0f4974b90edaae0c00d00c : SUCCESS in 40m 13s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e4bf5a7e53ed47f7b78c45d8810c621d : SUCCESS in 38m 23s\n- loci-placement https://zuul.opendev.org/t/openstack/build/a43d12a3653e482eaa4ff9056a515ce3 : SUCCESS in 25m 14s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/54a386a86a4545fea9398d0f09c62126 : SUCCESS in 27m 14s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/97037b0dd88c477bab7a6f3ccbf12c74 : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":2},{"id":"cd691cdb64d0f753333d4f0b11bfdf455705c6c1","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-08-07 01:02:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f7bf6bf4e40b1a0a594071d3d626c2b6dabf2842","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2020-08-07 15:48:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6e7b0dab2e5cf763d0d1d9c49b6197fd73c253f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 15:48:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"005c760b25fae1b684bdbb19878dc67678549500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 16:38:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/989edc8d2a8a4cc8b789e0c33210bced : SUCCESS in 39m 50s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ff0ae383dd094ae18b52bb81f3b170b6 : SUCCESS in 41m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/734adc66271744adb8e88043c6488841 : SUCCESS in 36m 18s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/942a8c9b46314406b25a17cddb89c58d : SUCCESS in 35m 01s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0518c6334ca6479dbebdec018fe04a7e : SUCCESS in 41m 58s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/170935158f4c43c6b23f886e15b5df0b : SUCCESS in 44m 15s\n- loci-manila https://zuul.opendev.org/t/openstack/build/302cfa94ee76461b8640d539f17c7fa6 : SUCCESS in 33m 08s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/ad13fd9d7612497a98a5e057047a73a4 : SUCCESS in 20m 51s\n- loci-nova https://zuul.opendev.org/t/openstack/build/1c212c8b7c454e97b87e3031039c33e0 : SUCCESS in 22m 12s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a768d98ba5474399a772543482ca7571 : SUCCESS in 40m 18s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e39026ae61924f688ef9aa6b64a5fa24 : SUCCESS in 32m 37s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/48dcf5ae9df14eb8ab0c4ad1d0fb1a13 : SUCCESS in 28m 12s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/3f43bbadd45e42d89d5f12f96a468b95 : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":2},{"id":"055e103ee7a83080c725506845dfd672f3f2d0c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 16:38:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f041966a8c6bb365a721dee016875893aa3925db","revisions":{"42727b0e261d5147011f446b389762091dfbbe7e":{"kind":"REWORK","_number":1,"created":"2020-08-06 19:44:41.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/09/745209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/09/745209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/745209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/745209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/745209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/09/745209/1"}}},"commit":{"parents":[{"commit":"20a7872551d789aa897fd1abecdb666c068f6baf","subject":"Merge \"Add purestorage to requirements image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/20a7872551d789aa897fd1abecdb666c068f6baf"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-08-06 19:44:27.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-08-06 19:44:27.000000000","tz":-300},"subject":"Fix KEEP_ALL_WHEELS flag usage","message":"Fix KEEP_ALL_WHEELS flag usage\n\nChange-Id: Ia1c4774aede69f83e1f43c7b511e176edc10cc9e\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/42727b0e261d5147011f446b389762091dfbbe7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/42727b0e261d5147011f446b389762091dfbbe7e"}]},"branch":"refs/heads/master"},"f041966a8c6bb365a721dee016875893aa3925db":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-06 19:45:06.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/09/745209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/09/745209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/745209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/745209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/09/745209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/09/745209/2"}}},"commit":{"parents":[{"commit":"3e78a46caf700aa6ba0bc176b3702c7ccfbf290c","subject":"Workaround unicode issue with py2 and scandir\u003d\u003d1.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/3e78a46caf700aa6ba0bc176b3702c7ccfbf290c"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-08-06 19:44:27.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-08-06 19:45:06.000000000","tz":0},"subject":"Fix KEEP_ALL_WHEELS flag usage","message":"Fix KEEP_ALL_WHEELS flag usage\n\nChange-Id: Ia1c4774aede69f83e1f43c7b511e176edc10cc9e\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f041966a8c6bb365a721dee016875893aa3925db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f041966a8c6bb365a721dee016875893aa3925db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
