)]}'
{"id":"openstack%2Floci~750482","triplet_id":"openstack%2Floci~master~Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4","project":"openstack/loci","branch":"master","hashtags":[],"change_id":"Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4","subject":"Add gating for Debian buster","status":"ABANDONED","created":"2020-09-08 19:54:10.000000000","updated":"2020-09-12 18:06:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a943cd02376c6574dd1733220337e82ae5016fe0","_number":750482,"virtual_id_number":750482,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-11 19:50:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 19:50:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a90cc11487a5ea934f328f7ca8d96bf62772f08","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-08 19:54:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87f8c723674e15b9b2d59beddb5e42aceafd4fba","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-08 19:54:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e044b90b3968aab09edded3711aa8070434758c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 20:36:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c5026849b897453ca54f94124ebec57b : FAILURE in 29m 38s\n- loci-glance https://zuul.opendev.org/t/openstack/build/8e4e50f24dfa45bc8d2f59728dc54c67 : FAILURE in 29m 14s\n- loci-heat https://zuul.opendev.org/t/openstack/build/0510fcb429754493a1f80c0bc2890988 : FAILURE in 29m 49s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/67017246c15045d6a125bcabffed3818 : FAILURE in 29m 04s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/27c93318d8ab48558ef5a50178683315 : FAILURE in 25m 55s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/692954fb8c4e4063bdf16c387d4f18f7 : FAILURE in 25m 48s\n- loci-manila https://zuul.opendev.org/t/openstack/build/aee9904b58224252b5f6e25be04f9eb0 : FAILURE in 26m 27s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/54f0037893694d9fb266cd83ca739396 : FAILURE in 26m 08s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c5ee85441f19428891dd8548942bc5de : FAILURE in 26m 40s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b1142190f7c94ff288c26f269561916f : FAILURE in 25m 18s\n- loci-placement https://zuul.opendev.org/t/openstack/build/3bc076c8aadb41c4a1efd3bc351940dc : FAILURE in 27m 05s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/3c789cd6f3d84d5b962b60b6112de84e : FAILURE in 24m 32s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b481cdf614414517a17c47fff27ccb40 : FAILURE in 28m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c89e5d095368caca0048aa110b63023443c7281f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-08 20:49:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f88cdcc9be1c9e4ac922975f8c08dd3ef508fab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 21:04:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/05a19e4e214e460891b6c04d3863da4a : FAILURE in 4m 57s\n- loci-glance https://zuul.opendev.org/t/openstack/build/18910ca90dd9430aa8f5a3cd24a699b2 : FAILURE in 5m 31s\n- loci-heat https://zuul.opendev.org/t/openstack/build/45aae7a59cbe4b779d3b759cd3d3318a : FAILURE in 5m 15s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a9b5ed16969d45d59a22290b74cf0887 : FAILURE in 6m 01s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/10e4cdd4e0ff453bbf4c57e86db6754b : FAILURE in 5m 18s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/abe541e9cb154fe88ab3ac1421a6167b : FAILURE in 4m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/db25580e9ffc465c92a482b7641c97d1 : FAILURE in 6m 01s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a05c0141ad9e44cca6d7347b9329c78c : FAILURE in 5m 41s\n- loci-nova https://zuul.opendev.org/t/openstack/build/689929838fb0440dad07e74e2b91c9c3 : FAILURE in 6m 20s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/6ba403cfd015421992b4c259f786ef6a : FAILURE in 5m 30s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b2488dca23d14980a90668e40292664d : FAILURE in 5m 29s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/2e1782799c24498c844a3ad48f5a592f : FAILURE in 5m 27s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/56877b6046b24c1fa574d9a3fd9ef94d : FAILURE in 7m 39s","accounts_in_message":[],"_revision_number":3},{"id":"280dc901299a86f14a90403ca308e24ebf2f542c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-08 21:41:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"623b6ccb7f6fbd7e24c4073ae43fd8cbffe97495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 22:27:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/ba160c56951843b7a75a3fed910f376d : FAILURE in 27m 09s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6204de2254d84e809f75da3d8778a488 : FAILURE in 25m 44s\n- loci-heat https://zuul.opendev.org/t/openstack/build/17ca66b1327e40798a114674ce8e8586 : FAILURE in 31m 25s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/fd65773736394f0b9427e9ce138fff9f : FAILURE in 16m 30s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/06dc56573ef94c22bf674c7a7ea519be : FAILURE in 24m 45s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8a6e75850bab4fd8a168aee6d4b53aed : FAILURE in 27m 07s\n- loci-manila https://zuul.opendev.org/t/openstack/build/906cee480f4a4f97b34b7a5e2755db58 : FAILURE in 31m 16s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/3119c01b10564a57be28bca455686bf0 : FAILURE in 26m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d4948d77e3ac4b1b8f8dd7ca4a829a0e : FAILURE in 29m 30s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/84946dba18e644ceb505721f6f93483b : FAILURE in 26m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cc582c79b9d64afb95e806104b0a4476 : FAILURE in 21m 34s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0b82f9e2bb0a4631a209404a8f941b9e : FAILURE in 23m 53s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/17160d585e1a43fb90b3a5ad9416a958 : FAILURE in 14m 54s","accounts_in_message":[],"_revision_number":4},{"id":"15ff64338e5aee2f18e1c8e0d028f4ca5950a87d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-08 22:39:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"41d4e8c635f6f82ed6e304263c6c10b82f3e4bbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 23:35:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9e69c7f61a9c4f6bbd941d21763d0fa5 : FAILURE in 28m 17s\n- loci-glance https://zuul.opendev.org/t/openstack/build/aa6e3cc018ed48a4b5b652ca52d9d9e8 : FAILURE in 27m 29s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e98b273be0a2405b80e9578e96de97e9 : FAILURE in 27m 32s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/1b96822780a042bb863ff02580c8e346 : FAILURE in 29m 33s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a6cf99bcd1c9429ab71eb636a601f87a : FAILURE in 28m 32s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/361a7e80e3974d47afa7e7cfbb6392a3 : FAILURE in 28m 30s\n- loci-manila https://zuul.opendev.org/t/openstack/build/8f40ed718a8c4522bd49472682daf0e3 : FAILURE in 34m 31s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/e579682869b94c66928fe4a13a75e467 : FAILURE in 34m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/abb4ece0f1e94ee8ba61382d601b87a6 : FAILURE in 34m 37s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/91fd872dc40745d4909e8906e91a2780 : FAILURE in 34m 20s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2b61d6a76c964b548773465e228e4e02 : FAILURE in 27m 42s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e41e9371a51946bfab8576f3c7d1dcec : FAILURE in 26m 36s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/dc15aa2fc8d14d8eb13a88bcdaf739ca : FAILURE in 26m 36s","accounts_in_message":[],"_revision_number":5},{"id":"5a8cbc8144527b42085489c036a7028268a1fe23","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-11 16:51:01.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"2321093a14d0c467a580d6a4ecdb670c90bd87dc","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-11 17:16:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9f98dc06cf226204a9471fdd4503e609ce34058a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-11 17:20:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a6f6e060579f599e3c7c266fffe4a9ce0adf5b70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 18:19:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/451fc1c3b73545e4a7d00ccdf630f3b4 : SUCCESS in 43m 35s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a5409361c57141d2992dff1708eb26af : SUCCESS in 40m 32s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8fb731f3cb374071b0c64b8f5708dc45 : SUCCESS in 42m 44s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/878bf0400efd45269f4c012eb51be940 : SUCCESS in 39m 00s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/eaa7011bf1664b689f92d75ee8117178 : SUCCESS in 37m 08s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9bf805f8e4f14dbdbd30f374f54957b4 : SUCCESS in 38m 14s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a24a2cc42a054bdeb27d0ab2d4cb29f1 : SUCCESS in 35m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d630b695a17e4c22b705258a030abb44 : SUCCESS in 45m 46s\n- loci-nova https://zuul.opendev.org/t/openstack/build/4ed3d8f51a664991bb682e574973f7ff : SUCCESS in 51m 36s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/0b478172959f4fcb8dd1e978ef66a71c : SUCCESS in 43m 36s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e806edcee20a4a11bc5a278605f04843 : FAILURE in 35m 47s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/089eb4e640a7402ca8210d05de1c2284 : SUCCESS in 36m 50s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/7db202020a6b4f9e8c7646093f2c95ab : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":7},{"id":"4b78aa5707b6f72867c5657d9405f313b4754942","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-11 18:31:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6982c15349fb1811df34c65a6abe51f542054b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 18:49:00.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0342471b9b324214a2b77611bc3d363a : FAILURE in 4m 24s\n- loci-glance https://zuul.opendev.org/t/openstack/build/51eab996b9de4dd8be26e7bfe601630d : FAILURE in 2m 44s\n- loci-heat https://zuul.opendev.org/t/openstack/build/84fe8ba5e03a44cda98e08aaecc48d38 : FAILURE in 2m 50s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/81ebf67e8fbc49e0a7f6761ed4d67d71 : FAILURE in 3m 01s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0f44268920954e7a88f72eb183172168 : FAILURE in 3m 17s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/ec135b5389344d3b80367c97e9f34dcd : FAILURE in 2m 50s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e7523b0a14cd48b9a355700e5b018280 : FAILURE in 3m 07s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/3627d5a1553c4dd99611fcdd59c45d09 : FAILURE in 3m 04s\n- loci-nova https://zuul.opendev.org/t/openstack/build/78c82939846148c1ad5e58cba973776c : FAILURE in 2m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3bb08850ab5848e89faa5cb41f79f566 : FAILURE in 4m 31s\n- loci-placement https://zuul.opendev.org/t/openstack/build/366f5e5d890e43388624270bff3e941c : FAILURE in 3m 22s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1680b04883854753a4a56eb0a7f23857 : FAILURE in 3m 02s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5adcf8b819064f7fbc1631ab201375a0 : FAILURE in 4m 22s","accounts_in_message":[],"_revision_number":8},{"id":"17b4008429e2be27bc954298ddcd621a02487d02","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-11 18:56:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2b55bc7f877af681ff977b82a371cc125f55e104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 19:50:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a1f6d0e4ceee41aeae828121e24819d1 : SUCCESS in 43m 41s\n- loci-glance https://zuul.opendev.org/t/openstack/build/579a07188d32401ebd9dc9ef41b9b3e1 : SUCCESS in 40m 38s\n- loci-heat https://zuul.opendev.org/t/openstack/build/4e36218bcd944dd9a4df648ac05921be : SUCCESS in 41m 57s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/45af89b3fbe243b6beca321bfc6dd72d : SUCCESS in 43m 31s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/23490122bdd148f59a5de7525d3cec93 : SUCCESS in 39m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/426694a3671f452f977b207816efd9be : SUCCESS in 40m 13s\n- loci-manila https://zuul.opendev.org/t/openstack/build/1894274020d44f90af82fe8656aa3bf6 : SUCCESS in 47m 14s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f6107f4644ca429b8f49bde1cd8cff33 : SUCCESS in 47m 10s\n- loci-nova https://zuul.opendev.org/t/openstack/build/48844ea976e0468f88806afb9b5b5073 : SUCCESS in 47m 50s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/65745ee5e98d4eb69f2afeeb57b02a3a : SUCCESS in 41m 18s\n- loci-placement https://zuul.opendev.org/t/openstack/build/43edffd47f0440c58e57f728efa6021e : SUCCESS in 35m 46s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6b4b10105b644c62a9c8baf65910c7e0 : SUCCESS in 30m 20s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/96659ed9cbc94ec29eca213455055173 : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":9},{"id":"a943cd02376c6574dd1733220337e82ae5016fe0","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-12 18:06:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"dfe6d8b6e4f7f98dcd0553eec6b9f1b27da3a6b0","revisions":{"5c930c6d43c933cf061db0891f6f32675d2c04b2":{"kind":"REWORK","_number":1,"created":"2020-09-08 19:54:10.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/82/750482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/82/750482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/82/750482/1"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:53:52.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:54:08.000000000","tz":-240},"subject":"Add gating for Debian buster","message":"Add gating for Debian buster\n\nChange-Id: Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c930c6d43c933cf061db0891f6f32675d2c04b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5c930c6d43c933cf061db0891f6f32675d2c04b2"}]},"branch":"refs/heads/master"},"d30bd434780cbf21f9f496e016b38e907d3c2cf4":{"kind":"REWORK","_number":2,"created":"2020-09-08 19:54:58.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/82/750482/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/82/750482/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/82/750482/2"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:53:52.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:54:53.000000000","tz":-240},"subject":"Add gating for Debian buster","message":"Add gating for Debian buster\n\nChange-Id: Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d30bd434780cbf21f9f496e016b38e907d3c2cf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d30bd434780cbf21f9f496e016b38e907d3c2cf4"}]},"branch":"refs/heads/master"},"1beae81672c6c5dc74b88c26d1c6cf8f019f7d62":{"kind":"REWORK","_number":3,"created":"2020-09-08 20:49:12.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/82/750482/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/82/750482/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/82/750482/3"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:53:52.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 20:49:06.000000000","tz":-240},"subject":"Add gating for Debian buster","message":"Add gating for Debian buster\n\nChange-Id: Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1beae81672c6c5dc74b88c26d1c6cf8f019f7d62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1beae81672c6c5dc74b88c26d1c6cf8f019f7d62"}]},"branch":"refs/heads/master"},"8e0505443252add50f2b96a74b94f041631fa9f6":{"kind":"REWORK","_number":4,"created":"2020-09-08 21:41:47.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/82/750482/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/82/750482/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/82/750482/4"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:53:52.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 21:41:41.000000000","tz":-240},"subject":"Add gating for Debian buster","message":"Add gating for Debian buster\n\nChange-Id: Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8e0505443252add50f2b96a74b94f041631fa9f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8e0505443252add50f2b96a74b94f041631fa9f6"}]},"branch":"refs/heads/master"},"67269ec058638304c0d45be5845690d7d38bbb0f":{"kind":"REWORK","_number":5,"created":"2020-09-08 22:39:08.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/82/750482/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/82/750482/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/82/750482/5"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:53:52.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 22:39:02.000000000","tz":-240},"subject":"Add gating for Debian buster","message":"Add gating for Debian buster\n\nChange-Id: Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67269ec058638304c0d45be5845690d7d38bbb0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67269ec058638304c0d45be5845690d7d38bbb0f"}]},"branch":"refs/heads/master"},"278270b65710087519bfb0a7637e25d650bb1a9f":{"kind":"REWORK","_number":6,"created":"2020-09-11 17:16:38.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/82/750482/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/82/750482/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/82/750482/6"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:53:52.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-11 17:16:33.000000000","tz":-240},"subject":"Add gating for Debian buster","message":"Add gating for Debian buster\n\nChange-Id: Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/278270b65710087519bfb0a7637e25d650bb1a9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/278270b65710087519bfb0a7637e25d650bb1a9f"}]},"branch":"refs/heads/master"},"d968f1766b397c3782f2c0e99012fb11c4ec5dd8":{"kind":"REWORK","_number":7,"created":"2020-09-11 17:20:46.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/82/750482/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/82/750482/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/82/750482/7"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:53:52.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-11 17:20:41.000000000","tz":-240},"subject":"Add gating for Debian buster","message":"Add gating for Debian buster\n\nChange-Id: Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d968f1766b397c3782f2c0e99012fb11c4ec5dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/d968f1766b397c3782f2c0e99012fb11c4ec5dd8"}]},"branch":"refs/heads/master"},"f40f1a3809ee8d4fe2a6b4279b96832634a62180":{"kind":"REWORK","_number":8,"created":"2020-09-11 18:31:14.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/82/750482/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/82/750482/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/82/750482/8"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:53:52.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-11 18:31:08.000000000","tz":-240},"subject":"Add gating for Debian buster","message":"Add gating for Debian buster\n\nChange-Id: Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f40f1a3809ee8d4fe2a6b4279b96832634a62180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/f40f1a3809ee8d4fe2a6b4279b96832634a62180"}]},"branch":"refs/heads/master"},"dfe6d8b6e4f7f98dcd0553eec6b9f1b27da3a6b0":{"kind":"REWORK","_number":9,"created":"2020-09-11 18:56:23.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/82/750482/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/82/750482/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/82/750482/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/82/750482/9"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-08 19:53:52.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-11 18:56:12.000000000","tz":-240},"subject":"Add gating for Debian buster","message":"Add gating for Debian buster\n\nChange-Id: Id0d1f622ecb0be6f64cbdcbe6feb3462f16a85f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfe6d8b6e4f7f98dcd0553eec6b9f1b27da3a6b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/dfe6d8b6e4f7f98dcd0553eec6b9f1b27da3a6b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
