)]}'
{"id":"openstack%2Floci~753424","triplet_id":"openstack%2Floci~master~Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f","project":"openstack/loci","branch":"master","topic":"repo-patch4","hashtags":[],"change_id":"Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f","subject":"Add openssh client to binary deps","status":"MERGED","created":"2020-09-22 18:34:30.000000000","updated":"2021-01-26 20:05:24.000000000","submitted":"2021-01-26 20:05:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753424-repo-patch4","meta_rev_id":"0292afd07794e4ef202a688ca3c6a33e7fee03bf","_number":753424,"virtual_id_number":753424,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 20:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"recommended":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"all":[{"value":1,"date":"2021-01-08 00:42:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":2,"date":"2021-01-26 15:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-26 15:23:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2021-01-08 01:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-26 15:23:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 22:56:08.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-09-28 18:18:34.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-10-11 23:59:38.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2020-10-19 23:26:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 19:18:08.000000000","updated_by":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"reviewer":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"state":"REVIEWER"},{"updated":"2021-01-26 15:27:32.000000000","updated_by":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"reviewer":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"state":"REVIEWER"}],"messages":[{"id":"ba5e99d539fefc5fc0190ed21315268b60dbb2e2","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-22 18:34:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48387a40e6da54d19d8be38b9ef650f04742a519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 19:03:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4e3df5d89582451e91ba37e7c51e1339 : RETRY_LIMIT in 2m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2b296caf83ec4660bcf6f5df6236e424 : RETRY_LIMIT in 2m 17s\n- loci-heat https://zuul.opendev.org/t/openstack/build/54487ea584ef40369fbf08a6fb6ff67e : RETRY_LIMIT in 2m 17s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4e24e90ba2ea4ee5a2ba05ea0ec4a778 : RETRY_LIMIT in 2m 19s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/0f9913a915f64377a49ebdc8aa0ddc18 : RETRY_LIMIT in 2m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/35b464cf54d84e93bd46d87e49e2c478 : RETRY_LIMIT in 2m 19s\n- loci-manila https://zuul.opendev.org/t/openstack/build/35cb96b8b235443aa22a402f57767df9 : RETRY_LIMIT in 2m 15s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/a59f96d856f94c05add2aeb5e19321f0 : RETRY_LIMIT in 2m 21s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e321ef253ef9495d9f69ba9da5582d6d : RETRY_LIMIT in 2m 22s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/37fd9398699e4be6b3671146a3002355 : RETRY_LIMIT in 2m 28s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ca1626fdecd94f64b944a5b800915863 : RETRY_LIMIT in 3m 56s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/06fbfed0b351483d9421cb7eb27b3eb0 : RETRY_LIMIT in 2m 59s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/891a85d826db4276a85e35212de32912 : RETRY_LIMIT in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"42265e45470e55e3d419a59a1c5edb64c1fceb9f","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-22 19:10:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f718ab5e051cb27e1ab9249ecce2cb5476d14014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 19:51:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/114d1775eca04df8a2829526beb2f69d : RETRY_LIMIT in 3m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/55a4f464c6f84767beb92fff4b4e8a98 : RETRY_LIMIT in 2m 35s\n- loci-heat https://zuul.opendev.org/t/openstack/build/212897e0afc047adb2631e8aa4a8070e : RETRY_LIMIT in 2m 43s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4c59c493a57a4ba28b6fd89f9241df41 : RETRY_LIMIT in 2m 34s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/47713adb92b948269698417e0a2df66a : RETRY_LIMIT in 3m 55s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/fbd11b7709d1451aa18cbfd99f755c4a : RETRY_LIMIT in 2m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0dc7b4601b6545d987125c86a82801b9 : RETRY_LIMIT in 3m 01s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2cfc272a21374b3d9eb17eeb600624eb : RETRY_LIMIT in 2m 29s\n- loci-nova https://zuul.opendev.org/t/openstack/build/dcd0992405c84cbb988f254c301b846e : RETRY_LIMIT in 4m 46s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/02fe53bd21844d91beb32dc4c4d8c88e : RETRY_LIMIT in 2m 12s\n- loci-placement https://zuul.opendev.org/t/openstack/build/8e5f4888c6df4356bc8a7745c3811687 : RETRY_LIMIT in 2m 55s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4e28c4dd738a4c8da3c43b1c1dda757c : RETRY_LIMIT in 3m 23s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/69d89fb28f4b4413ae3099153868f26b : RETRY_LIMIT in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c63ac84588314c8a3dd7fe650159be9aa25a3233","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-22 21:19:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"be04282cc718ea59644d6c9f9b2ad17eb9b77ee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 21:45:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9a90ea5baf3a46b092c140029255a865 : RETRY_LIMIT in 2m 24s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d001d857fb0546e19c8a322014046b3f : RETRY_LIMIT in 2m 33s\n- loci-heat https://zuul.opendev.org/t/openstack/build/619578daa7ef473fae8facf657dad42d : RETRY_LIMIT in 1m 59s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9383ecc301e846268240996ef3699cba : RETRY_LIMIT in 2m 24s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b0b3d00b3b3545229b73d1dba9ebe710 : RETRY_LIMIT in 3m 18s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/8eb637abd84e4dd89c66bef32260bdd9 : RETRY_LIMIT in 3m 15s\n- loci-manila https://zuul.opendev.org/t/openstack/build/838557af338842abb058b63685609acd : RETRY_LIMIT in 2m 02s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/47fcadc5b3b34494a644f1e3fed6b692 : RETRY_LIMIT in 2m 28s\n- loci-nova https://zuul.opendev.org/t/openstack/build/2e4cae9456374c739476bf0968149df1 : RETRY_LIMIT in 2m 19s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3b1ab6ddf95c4654b92800a63c3d82ed : RETRY_LIMIT in 3m 47s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b33e330a1e0046bb8316c6171464efad : RETRY_LIMIT in 3m 09s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/e12f7f5bb54c44f197f80d1cffe5c6eb : RETRY_LIMIT in 1m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/1e0011577e154ec29dec48bb85aed312 : RETRY_LIMIT in 2m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8a1bbe0798dd6b1549223740012007b74ed54355","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-22 21:51:07.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nplease rebase on top of https://review.opendev.org/#/c/753452/ :)\n\nthat should be the gate fix","accounts_in_message":[],"_revision_number":2},{"id":"7543a2da326ddc443972ff5e91d44643cf6fed72","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-22 22:06:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50eb2b3226efe0c4da1873b5bc075ebb66d3642f","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-22 22:09:12.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"21f3e06e5acaa1269c58ffcda756dbaf951c27ec","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-22 22:56:08.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b9e7277b9bb81f873443233482b736953702a68e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 22:56:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/dd9a67dfb85847ca810456674d4a1c49 : SUCCESS in 33m 26s\n- loci-glance https://zuul.opendev.org/t/openstack/build/121ffe1b3f8a4b2b81cbce958523e8d4 : SUCCESS in 36m 15s\n- loci-heat https://zuul.opendev.org/t/openstack/build/70bc79ec48474c68b9a77b7d1df60980 : SUCCESS in 40m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/9287e5061e6e449db4edb8cb1f78471c : SUCCESS in 38m 08s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2a1e222b355a40b98817f9ec3da2218b : SUCCESS in 36m 56s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6883e259e5aa471392edd382b27a2c1b : SUCCESS in 33m 47s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9a28f8c1458f4497a7292be94e24aecc : SUCCESS in 33m 37s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8318578331394882847034a5fb3f5965 : SUCCESS in 36m 26s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c718ef917eb1474590142809d8e1d9a7 : FAILURE in 34m 08s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/0525249eacf74e5d82579fd106368432 : SUCCESS in 33m 28s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e413f90de737407893e996c485a71502 : SUCCESS in 26m 49s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/59afe070898e4c3aa560303ce8aa4bb3 : SUCCESS in 23m 20s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2f721e97b0394dc1aa3f7c71adbaaf0c : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":4},{"id":"4db04c6c01cde83e8eb07febee47ced59befde7f","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-23 13:37:00.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"c24490a270e4f42723f1cd293131c601de677ea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 14:21:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a715365de5174ef4a680af57940d468a : RETRY_LIMIT in 2m 37s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6b90befc067149208c8b3750dcc5c979 : RETRY_LIMIT in 3m 57s\n- loci-heat https://zuul.opendev.org/t/openstack/build/54c2254b5d824b19be8e0975558fb087 : RETRY_LIMIT in 2m 55s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4cb9ea6a3c414abd9e9d8d9ab1805964 : RETRY_LIMIT in 4m 06s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/1363b9d581aa419093d5329c2e6109c5 : RETRY_LIMIT in 3m 36s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e0c7eecaf3954f97811b519a48cbffd8 : RETRY_LIMIT in 2m 34s\n- loci-manila https://zuul.opendev.org/t/openstack/build/1e91198234664c3ebf52a9e1daa91225 : RETRY_LIMIT in 2m 45s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/10d951a3fa264fa789b02e34a9395635 : RETRY_LIMIT in 3m 25s\n- loci-nova https://zuul.opendev.org/t/openstack/build/53c6f3c124cb4b53bf070d755180fb6d : RETRY_LIMIT in 2m 57s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/e1e1f71120164231b6339e0925353f14 : RETRY_LIMIT in 2m 43s\n- loci-placement https://zuul.opendev.org/t/openstack/build/d1755002b68943129e0d6d9521ef812b : RETRY_LIMIT in 4m 00s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/bf1fb9cb3330459d93723ea7575a8a19 : RETRY_LIMIT in 4m 13s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/6f020dbcfc9648018aad49179306a31d : RETRY_LIMIT in 3m 19s","accounts_in_message":[],"_revision_number":5},{"id":"9555e0d47f2e158b5f94b15a3d95ca05ea0b4379","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-23 14:59:38.000000000","message":"Patch Set 5:\n\nplease rebase on top of gate fix","accounts_in_message":[],"_revision_number":5},{"id":"82ace84cd6006e1ecb2decff4984612b03399a64","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-23 15:39:27.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"80c8b65d7af298388b68e5f62f4f6737280da5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 16:16:24.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/2559b13075b94a9cb360267efa623c43 : RETRY_LIMIT in 2m 46s\n- loci-glance https://zuul.opendev.org/t/openstack/build/efbadbaacbee4a44a1fa7d9d72d5c4e1 : RETRY_LIMIT in 2m 34s\n- loci-heat https://zuul.opendev.org/t/openstack/build/70adc60f5dfc432fb133c84df2f23f4c : RETRY_LIMIT in 3m 38s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f64daf584328489c9aea3791acebf9ff : RETRY_LIMIT in 3m 53s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2ab130a5841045019256bdfe90330129 : RETRY_LIMIT in 3m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/eeeae37bcbe84b78bf3f9a638a792d0b : RETRY_LIMIT in 2m 29s\n- loci-manila https://zuul.opendev.org/t/openstack/build/d836511bda3d4c53b3f48653a3379b4d : RETRY_LIMIT in 4m 39s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5cf97fd71a144e5288acd5965e4fa6be : RETRY_LIMIT in 2m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c5fcc76c910c4aa2b9b84d984a10a6e3 : RETRY_LIMIT in 2m 44s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5324b48956e848d3bee4841c51b5ba37 : RETRY_LIMIT in 2m 44s\n- loci-placement https://zuul.opendev.org/t/openstack/build/35ef0c02bfba408b9460ed2d40807e9d : RETRY_LIMIT in 2m 43s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/3b50219ada7d44dfad6066956f3a372f : RETRY_LIMIT in 2m 56s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/2d8ba1697d174fd3b85bfa8eb83895ed : RETRY_LIMIT in 2m 45s","accounts_in_message":[],"_revision_number":6},{"id":"9f84d9fb93442552f073dc7b22763cfcb46c86e7","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-24 12:47:19.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"24afaf313fdac2efe7cea105c14e0e45398849ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:41:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/b8e5a90efe174fe995d5dcaed1905c6a : SUCCESS in 45m 12s\n- loci-glance https://zuul.opendev.org/t/openstack/build/494d3ffc76a4472396bcc4c6b70bf504 : SUCCESS in 31m 59s\n- loci-heat https://zuul.opendev.org/t/openstack/build/398c6e8d63b84a06a80872a1c6a16790 : SUCCESS in 36m 16s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/2a772ff95f4f45469bf1e0da7e84a231 : SUCCESS in 32m 18s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b857e4238ea5417abd1f6bc56654fbd8 : SUCCESS in 31m 23s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c1e6751497064a4485a0fdffdc964e20 : SUCCESS in 34m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/255f12832007456e98ee180ff5b7d28a : SUCCESS in 34m 34s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4ac556c07a4747f0ad40e57391240191 : SUCCESS in 43m 55s\n- loci-nova https://zuul.opendev.org/t/openstack/build/124ea73121f4494e96b2d69a9a701663 : FAILURE in 35m 00s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3ea42c341de84f64a9dff7d54ae8321c : SUCCESS in 33m 12s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4cf0fa0c865b4625ba3299d98a9e070d : SUCCESS in 26m 14s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a477b77cd78b40f68799ce57ba0d1973 : SUCCESS in 24m 33s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/5050d179f67841d6949c0f00cb9ce896 : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":7},{"id":"94b692d0fb14cf3e3a5ddf2ef702c0c04126d159","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-09-24 21:06:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a47ed6188ed45880a6401f0c7d2d07e52c090704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 21:56:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/4d39121f50be45ecaf1d7aaa5cd6a3de : SUCCESS in 32m 43s\n- loci-glance https://zuul.opendev.org/t/openstack/build/bfcbbd003fc94fcd874b2cc73f9a0394 : SUCCESS in 20m 23s\n- loci-heat https://zuul.opendev.org/t/openstack/build/9c542b37244448ae8e456721051d107c : SUCCESS in 31m 53s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f79d72d0aaeb4c7fae66383b04a963b3 : SUCCESS in 34m 53s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/fbf7cd63dc7e4444a624dbe37b51cee2 : SUCCESS in 37m 03s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3abf980ee7dc4053b382031b5da21dbd : SUCCESS in 36m 24s\n- loci-manila https://zuul.opendev.org/t/openstack/build/2fd51a4b001149c292b1d81113aae9e3 : SUCCESS in 19m 51s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/041ba31954f84106a1314a18d8c6aa29 : SUCCESS in 34m 16s\n- loci-nova https://zuul.opendev.org/t/openstack/build/0785c5078f504f29bafa3e8ad2d2da30 : SUCCESS in 41m 12s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f4af8b09cb584651b8dad723f3149fe2 : SUCCESS in 40m 14s\n- loci-placement https://zuul.opendev.org/t/openstack/build/73f5618f4f4b432fb7963c17db42430d : SUCCESS in 29m 22s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ae0b36f617b64130a4b5d7b6eba6c459 : SUCCESS in 23m 01s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/150124b762e7489ea16f5cd5f99852dd : SUCCESS in 19m 14s","accounts_in_message":[],"_revision_number":8},{"id":"948987a798e9cff3acd3f411b96976db94df1b69","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-09-28 18:18:34.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"f24ff5019068db8f4f5f992efff3d6dec646ac7d","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-11 23:56:54.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"cfa50bc8340f5b7aaafbcd8086b4ff0701663e9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-11 23:57:06.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"e7400c6e32d522573269f61a9f3ae513b01f6bc3","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-11 23:59:38.000000000","message":"Patch Set 8: Code-Review-1 Workflow-1\n\nlets add another profile for ssh, e.g. \u0027migration\u0027 or \u0027ssh\u0027. or just add it through DIST_PACKAGES separately. We need this to out by default","accounts_in_message":[],"_revision_number":8},{"id":"1cdcf15fa217b1e553ab2662356ddd18f7850e8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 01:03:14.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/3e2cd35daf9745bdb111b66c9587fd20 : TIMED_OUT in 1h 00m 56s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cec50ccf3ce64001ab9fdc546a72bf2f : SUCCESS in 32m 48s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ea47e9e677cc4810a4c4908edfcb4574 : TIMED_OUT in 1h 00m 57s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/04ccfed66c2f4eca9c3bab6ef59a775e : TIMED_OUT in 1h 00m 52s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/522451d0c7f44a198627abd51811adae : TIMED_OUT in 1h 00m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/e8f014c34fcf4799bd5f391e4b300c29 : TIMED_OUT in 1h 00m 55s\n- loci-manila https://zuul.opendev.org/t/openstack/build/81c4313387bf46698ee1b5bf78ead2ef : TIMED_OUT in 1h 00m 46s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/da70e322c4bf42c19f9d816e9d2cadfb : TIMED_OUT in 1h 00m 49s\n- loci-nova https://zuul.opendev.org/t/openstack/build/66cc233b710f49219cc0abf355b5c978 : TIMED_OUT in 1h 01m 07s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/b1aa4b80f17b4490b88ffc020e8de949 : TIMED_OUT in 1h 00m 50s\n- loci-placement https://zuul.opendev.org/t/openstack/build/92a21d846c874c37883200b6e5cf08dc : TIMED_OUT in 1h 00m 51s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a73624db2c6742a0808074caa86acaa1 : TIMED_OUT in 40m 51s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/675fd4202cab44babb15efb43ba7b9b2 : TIMED_OUT in 1h 00m 47s","accounts_in_message":[],"_revision_number":8},{"id":"d42d8f8ce7927523f94ff9eccc7aa6171f2290ca","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-12 10:32:35.000000000","message":"Removed Workflow-1 by Andrii Ostapenko \u003canost1986@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":8},{"id":"7c6863c4da1d305ad82cf8e8c42413ef742548ac","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-12 10:32:50.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"33ddfc3d18f8f76bcfd14c185446fcd43534cd92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 11:22:50.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d6cd8e84091b4db8aea918a15daf460f : SUCCESS in 37m 13s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6a1d4dcf839542c093a161886ea4517a : SUCCESS in 38m 02s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ac0a24e4e46348beb1e75cd92e923035 : SUCCESS in 40m 40s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f6de7111c575441f8703fe2eb5629878 : SUCCESS in 41m 09s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9672a744d0bf44a39f4af095fa7cb9d7 : SUCCESS in 37m 57s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/41004b00120b4cc0b9d4dfee3a2886db : SUCCESS in 35m 45s\n- loci-manila https://zuul.opendev.org/t/openstack/build/478178c1073249c8a56d273a217edd93 : SUCCESS in 35m 44s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/c7c3e2a9f7534af5852a2bc1d22fb7e4 : SUCCESS in 38m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/06f058ac1c92490fb977e5d90cc415fc : SUCCESS in 42m 29s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f9ab45e1643145a8b0584d2c0b5722c3 : SUCCESS in 42m 16s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ca8ab2fc466c4b0da4a1b95f4d3673d9 : SUCCESS in 31m 15s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/1027b5e07ab047b7a3b1e620860dda52 : TIMED_OUT in 41m 05s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f302a29b7e374df9a00f9f48992d92c0 : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":8},{"id":"e5876a5837bdfb157d59da9982a99b5db1d4c6fd","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-10-19 17:39:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"879ec4bfd5a0af8e124c799fb55c204b2e6cf53d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 18:06:02.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f066d350024946118b6fef6761f56c77 : FAILURE in 15m 02s\n- loci-glance https://zuul.opendev.org/t/openstack/build/ada9877743bd4bf29fab77b3668d3e07 : FAILURE in 17m 35s\n- loci-heat https://zuul.opendev.org/t/openstack/build/2c9c04c5794a47c1a46078869a7c2c7a : FAILURE in 17m 48s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c772aea8ef9c431397a6594d2729f94c : FAILURE in 17m 54s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b7fefcef5527476ca14b06276d40cc24 : FAILURE in 18m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/17bbadf6f1f24c938154171cd3e0fab7 : POST_FAILURE in 16m 22s\n- loci-manila https://zuul.opendev.org/t/openstack/build/de43367dead440df85bf5d2513b151d9 : FAILURE in 19m 09s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7d69a431597c44e9ab6b4885fa5c98c1 : FAILURE in 18m 33s\n- loci-nova https://zuul.opendev.org/t/openstack/build/ce789932f86f44378faf6624926fcc5e : FAILURE in 17m 56s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3d077c2b244346a0b4ba0379932c218a : FAILURE in 15m 54s\n- loci-placement https://zuul.opendev.org/t/openstack/build/88c7c64205cc43019b7ba74441adaba4 : FAILURE in 13m 46s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/d936d9d5620f40bfab6eb09d724f3533 : FAILURE in 17m 31s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b3b929676725437a991fb005d33acb36 : FAILURE in 19m 03s","accounts_in_message":[],"_revision_number":9},{"id":"65818bf680d67d2ed071e5c0276a5f1c3f289103","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-10-19 18:17:41.000000000","message":"Patch Set 9:\n\nsorry i was not specific enough. What i suggested to add an extra image type (e.g. nova-compute-ssh) to openstack-helm-images and pass DIST_PACKAGES from there. No change is needed to loci.","accounts_in_message":[],"_revision_number":9},{"id":"fa7d4e58ad6e6ac37ca97930d5e65ce7ad4c496a","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-10-19 18:18:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"98f51a3d1a9519ecffb583b57793c848767cfa8e","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-10-19 18:25:16.000000000","message":"Patch Set 10:\n\nhttps://review.opendev.org/#/c/753358/","accounts_in_message":[],"_revision_number":10},{"id":"426103055f67b09ccf3637aea05af20b43fa2a5c","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-10-19 18:25:58.000000000","message":"Patch Set 10:\n\nAndrii: Already made a PS for that","accounts_in_message":[],"_revision_number":10},{"id":"39b7fc826e808fb7063fb9ab5a00696962ea5fb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 19:56:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/31abfefbfa704eb7b3c762d17f6ba3cb : SUCCESS in 42m 53s\n- loci-glance https://zuul.opendev.org/t/openstack/build/70944fe2d29c4fe4b202396ebcff7f9d : SUCCESS in 37m 14s\n- loci-heat https://zuul.opendev.org/t/openstack/build/f18bff07c17b4f27a76e93e2725313ce : SUCCESS in 40m 54s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/fc4f669b2284468eb8a60b1cbefe9ebb : SUCCESS in 41m 12s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f63a35cbde904a068b97a4221e08ca69 : SUCCESS in 36m 35s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/75b5dee1b9194559a5438f7266bf5a28 : SUCCESS in 36m 50s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a379a4c1396b42d7bbdc58f7179a73cf : SUCCESS in 36m 58s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/766a2ce4b2e6439baa571c5dc7197d45 : SUCCESS in 37m 16s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a062db7d38784141883896c30f7daf42 : SUCCESS in 44m 20s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4926d827f5de4fc691dbb3562f299383 : SUCCESS in 33m 29s\n- loci-placement https://zuul.opendev.org/t/openstack/build/4630ef72a5c94bb680b600e4bc3572ed : SUCCESS in 33m 39s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/9c264bc0cfde4705af2aa0f228fd4abe : SUCCESS in 28m 02s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/1c4e64dc5d9f4cee919c99f53cbc7181 : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":10},{"id":"1262e68292c86a6efb441fd67d833701729e1c9f","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-10-19 20:39:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"34ef1599e16c528e7f11126f3c9537075aef8a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 22:09:25.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/842df1e0a217490ba84588b3455e130d : SUCCESS in 36m 45s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e24f5affbe254d248161a06159a950f8 : SUCCESS in 33m 14s\n- loci-heat https://zuul.opendev.org/t/openstack/build/227619affc5d4115a788e4fd65eb1ed9 : SUCCESS in 39m 16s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e7849b8b5a914eb59b68c0c17c6ef999 : SUCCESS in 34m 20s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/8d68907781d24f509623a5e7ba98c98b : SUCCESS in 33m 30s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/954de203532e49558b315e5e70d17184 : SUCCESS in 40m 27s\n- loci-manila https://zuul.opendev.org/t/openstack/build/e97eb55372784df68130d468afd7eb48 : SUCCESS in 40m 44s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b2dfc4677893437091bd513fc2425e50 : SUCCESS in 37m 13s\n- loci-nova https://zuul.opendev.org/t/openstack/build/b948601d98ec460b85c327a3086cfad9 : SUCCESS in 37m 29s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/12361bbe8fc7449dabb3f60b6500cf15 : SUCCESS in 41m 09s\n- loci-placement https://zuul.opendev.org/t/openstack/build/ce4d230ef40249a2a5a3c3c09848be21 : SUCCESS in 32m 36s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/a3c1f16ec4ea4d5bb96bf31663eeeab0 : SUCCESS in 24m 43s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/fe6b8430542547869c5575ffcbbc578c : RETRY_LIMIT in 22m 41s","accounts_in_message":[],"_revision_number":11},{"id":"45b766b5da4847169705ec1d2839365683a8dd53","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-10-19 22:37:58.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"0712e1fa88ac673490ec841eb408543cf73e1585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 23:26:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/17b011daebcc4675a11c4ab94edffbf5 : SUCCESS in 39m 08s\n- loci-glance https://zuul.opendev.org/t/openstack/build/1c043a551d484c7f9eb4d39ab9681476 : SUCCESS in 36m 32s\n- loci-heat https://zuul.opendev.org/t/openstack/build/919d5a7120e2465c96875932aff10858 : SUCCESS in 38m 18s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/f7325486042e47bfa6191e245b493a33 : SUCCESS in 38m 27s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/318c7cf0588744a69c44aa1e347ab99c : SUCCESS in 38m 21s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/c403b063e1c547839fd50e42b73511a6 : SUCCESS in 40m 46s\n- loci-manila https://zuul.opendev.org/t/openstack/build/071ad65eaa58402db9549c81f0b7ffcc : SUCCESS in 37m 31s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b8f2c1e993cb4c7cb36a3a9344eb80fd : SUCCESS in 39m 50s\n- loci-nova https://zuul.opendev.org/t/openstack/build/dab62931646c4aafa9c7d012a426b044 : SUCCESS in 43m 23s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/9699d9bf85c44d178ee911be37dbea77 : SUCCESS in 39m 32s\n- loci-placement https://zuul.opendev.org/t/openstack/build/1edcb04715284fae9f222f267023e8ac : SUCCESS in 32m 16s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/5e1b55be75754600aae1a0c9fdbcb56e : SUCCESS in 28m 13s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/f621ae9a06954353aa9032fd075797e6 : SUCCESS in 38m 34s","accounts_in_message":[],"_revision_number":11},{"id":"c03bf02933ff7ec27d1b3b7a0f026833322a600e","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-12-09 19:33:59.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"bf2ae33c3c224a55d56fb7397c4b21a503c4fc12","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-09 20:23:26.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"272502d0918005ae7b96e0a1b5ae607a1a942aec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 21:01:58.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/e6352aabc2d54059bb8d30c2bafe79fe : FAILURE in 31m 18s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6b05351311114b2ab75901816ebb8ea8 : FAILURE in 31m 19s\n- loci-heat https://zuul.opendev.org/t/openstack/build/6181e5de84ed4f00becdf13542709009 : FAILURE in 35m 25s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ab8e4b4161f4477abb52cb5e22a44e5d : FAILURE in 23m 07s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f9eec3cc7cbf48058f24d67bc75903a5 : FAILURE in 30m 00s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/d7d82c59655e4c10bc94df16eb0c0634 : FAILURE in 23m 04s\n- loci-manila https://zuul.opendev.org/t/openstack/build/150a552db7bd42da97c3404cbc3a9d25 : FAILURE in 14m 52s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d768d25840cb4c81af9a739ae65d60c5 : FAILURE in 13m 29s\n- loci-nova https://zuul.opendev.org/t/openstack/build/26b78d6f137f431bb3a22d4485a93e18 : FAILURE in 34m 51s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/a5780c543ae442808cd668e01708f01c : FAILURE in 28m 27s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f4d0c95f910e4e3f8d1a6091c672b84e : FAILURE in 23m 08s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/0fc88ad7297d41e78e96f23562356031 : FAILURE in 21m 48s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/eca9595f592740b8b84df5c0a4a8709c : FAILURE in 25m 08s","accounts_in_message":[],"_revision_number":11},{"id":"319dd1eec4676356ec3ea665293e86ba5205b025","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-10 11:00:07.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"0254502749e79ff69ec6f8173a1046cec6990b98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 11:33:03.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/5031101e2fa04a9592ecca32be31bcf8 : FAILURE in 22m 27s\n- loci-glance https://zuul.opendev.org/t/openstack/build/79298fedaf78481aaa1b8dc44754437a : FAILURE in 21m 16s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8dfe7e86f73f419a9bcecfa5428c27df : FAILURE in 21m 26s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/22aa96e62de4458fb6ad70bade490884 : FAILURE in 20m 56s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/f1760e4bf2cf415999b7708b5c685fae : FAILURE in 26m 34s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1c64e89ba8724e459555d698bad96643 : FAILURE in 25m 49s\n- loci-manila https://zuul.opendev.org/t/openstack/build/a3513c5971724fd487d799a112d1ad25 : FAILURE in 21m 38s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/3803a5ec269b400384ca1ba3199ac12c : FAILURE in 21m 54s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a9020d63a1f14e8c931e43f328a9453d : FAILURE in 15m 13s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/baea6cc4afb249ccbb15a1cad2db9672 : FAILURE in 21m 19s\n- loci-placement https://zuul.opendev.org/t/openstack/build/9ece42b64a304092834b76bfdc05c21d : FAILURE in 21m 56s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/4ce7fdd70f7c40a28832e3cab7fc177b : FAILURE in 20m 12s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b400b91ac42d42569060a6fff869a0a6 : FAILURE in 21m 08s","accounts_in_message":[],"_revision_number":11},{"id":"a9db62d350636aeec1254aa9ce2965b22193bdc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-10 11:55:43.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"c74517b649e8677de710ca9b4999007804433ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 12:29:46.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/9323fdb1a74a4974895f820fd4b703e4 : FAILURE in 24m 19s\n- loci-glance https://zuul.opendev.org/t/openstack/build/cd8e7be569f445999b670db74260697e : FAILURE in 25m 45s\n- loci-heat https://zuul.opendev.org/t/openstack/build/aba9cc12a31b4bc9ba2899f6265b3250 : FAILURE in 25m 44s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a94ea32cacac473f92c53ff892cbbf75 : FAILURE in 21m 32s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e216d62c89bd4da19a59f3f5a8cb79a9 : FAILURE in 21m 13s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/046ea1afa708493ca9d1ad28821da715 : FAILURE in 22m 16s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5601949699b74148882253764e218f9b : FAILURE in 26m 24s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/95a9fadebc8d4da6839a7c9f74f131b9 : FAILURE in 21m 22s\n- loci-nova https://zuul.opendev.org/t/openstack/build/72b7661c81b44e899cd7121356545200 : FAILURE in 21m 53s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/31a20ef2130843ddb2391ebb7c8d1f0a : FAILURE in 22m 53s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e2094f2581c24bab87ad54d595f385e5 : FAILURE in 23m 59s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/21b4f9c13275441ca70697b18eae9da9 : FAILURE in 21m 14s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d65a6c4b2bd448ca959c7e979085ff98 : FAILURE in 20m 11s","accounts_in_message":[],"_revision_number":12},{"id":"7223946ae536676e4adf459df34faec185dba6db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-10 19:00:24.000000000","message":"Uploaded patch set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"aa77edbe70d5177bbfef99118725a7480ecf3402","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-10 19:07:55.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"b25587482f6978683b2a8a5cb264c904d03f0e4b","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2020-12-10 19:18:08.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"25550a5a29769ed9c5b1a02cfe9f7e39791d7169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-12-10 19:33:31.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"126b9c28fe546993f0118e520190e8e9c464d475","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2020-12-10 20:20:02.000000000","message":"Removed Code-Review+1 by Hemachandra Reddy \u003chemachandra.reddy@att.com\u003e\n","accounts_in_message":[],"_revision_number":15},{"id":"6cb0c7d7f609d87c853c19dda889289364459417","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2020-12-10 20:20:26.000000000","message":"Patch Set 15: Code-Review-1\n\nDoes not it need ssh server too?","accounts_in_message":[],"_revision_number":15},{"id":"8feed0a55e1e297ec14c835656b5939835be4546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 20:30:50.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d07231646c6946869ddff21b15c2af2f : SUCCESS in 34m 56s\n- loci-glance https://zuul.opendev.org/t/openstack/build/6f033c89c4384f59b14889955862f502 : SUCCESS in 32m 44s\n- loci-heat https://zuul.opendev.org/t/openstack/build/572aaaa3b9bd436a8974d08852f99516 : SUCCESS in 38m 19s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ad0956f1b6c945b1bc01aafa2ba768d2 : SUCCESS in 22m 10s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/da118dfbc3254dcbab18eddf4a6e84fd : SUCCESS in 38m 46s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/76a9dd907a24423bac9edf5bcf37755c : SUCCESS in 37m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/b279a6ffd4cc4418b768b34cbf99a189 : SUCCESS in 43m 42s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/29436a51471242d3abf19bd7e4ba6817 : SUCCESS in 37m 49s\n- loci-nova https://zuul.opendev.org/t/openstack/build/f7f7239a527047348c409b20263bd6ec : SUCCESS in 44m 48s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/944ae74a12fb4cd48d21000b0663c1c8 : SUCCESS in 39m 15s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2c9a313b7d2a4cb08ed979913d4ec621 : SUCCESS in 30m 27s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/35c8d55a608e4a9a9de8c8e08255344e : SUCCESS in 28m 30s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/44cfb702ffad4b7ea875f8c24803c758 : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":15},{"id":"7fd918a2ad43d5551865195cacf742c798994cf5","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2020-12-10 20:55:28.000000000","message":"Patch Set 15:\n\nAlso, does \u0027nova\u0027 user needs a login shell for ssh to work? I\u0027m looking at this instructions: https://docs.openstack.org/nova/latest/admin/ssh-configuration.html#cli-os-migrate-cfg-ssh","accounts_in_message":[],"_revision_number":15},{"id":"8dfc88a7b250cb1d23c8ec7093552a1f66d0d937","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-11 12:50:06.000000000","message":"Patch Set 15:\n\n\u003e Patch Set 15:\n\u003e \n\u003e Also, does \u0027nova\u0027 user needs a login shell for ssh to work? I\u0027m looking at this instructions: https://docs.openstack.org/nova/latest/admin/ssh-configuration.html#cli-os-migrate-cfg-ssh\n\nno need in nova image.\nwe can use any ssh server for images.tags.nova_compute_ssh in openstack-helm repo, nova chart","accounts_in_message":[],"_revision_number":15},{"id":"1bb376e84ae2582abbdc04882749ff153df9f73e","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-11 12:51:41.000000000","message":"Patch Set 15:\n\n\u003e Patch Set 15:\n\u003e \n\u003e Also, does \u0027nova\u0027 user needs a login shell for ssh to work? I\u0027m looking at this instructions: https://docs.openstack.org/nova/latest/admin/ssh-configuration.html#cli-os-migrate-cfg-ssh\n\nalso look this, https://review.opendev.org/c/openstack/openstack-helm/+/753413","accounts_in_message":[],"_revision_number":15},{"id":"a21ae29a45f623190331815d292d533742f62658","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-11 12:54:34.000000000","message":"Patch Set 15:\n\n\u003e Patch Set 15: Code-Review-1\n\u003e \n\u003e Does not it need ssh server too?\n\nno need in nova image.\nwe can use any ssh server for images.tags.nova_compute_ssh in openstack-helm repo, nova chart","accounts_in_message":[],"_revision_number":15},{"id":"76bf9320a494393c03a1e713ec1b56f266ec73e4","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-14 13:02:44.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"424ca9729fda2de41f87bf15d0d589e1b7b568f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-12-14 13:12:23.000000000","message":"Uploaded patch set 16: New patch set was added with same tree, parent, and commit message as Patch Set 15.","accounts_in_message":[],"_revision_number":16},{"id":"bf34f2ea157a7b2a118d24bc8ebf277f2026a5ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 14:10:34.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/c851dd78eab84304aff204e7e027dbf3 : SUCCESS in 52m 03s\n- loci-glance https://zuul.opendev.org/t/openstack/build/b4d0524d28694910a7b87b411ab35cef : SUCCESS in 49m 21s\n- loci-heat https://zuul.opendev.org/t/openstack/build/8474cf05e2e249a58c67e75f124e3fb5 : SUCCESS in 51m 14s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/09c051ddeb3e428aa2f04ba24a08871e : SUCCESS in 37m 57s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/95d0fb599e5544c48f0974e9ef8a0126 : SUCCESS in 34m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/2ff6a7e95c2c4927b5487c2084319f8b : SUCCESS in 36m 15s\n- loci-manila https://zuul.opendev.org/t/openstack/build/218e4227fe6345d9b1911c499a691386 : SUCCESS in 34m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/6455f8d190a244d39802e48e7ec14aed : SUCCESS in 43m 30s\n- loci-nova https://zuul.opendev.org/t/openstack/build/63c04e6d726d4ef689e95a1b7e7511bd : SUCCESS in 44m 45s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/5e9a502b15bb48499d4a1dea3955f60f : SUCCESS in 40m 40s\n- loci-placement https://zuul.opendev.org/t/openstack/build/97fc35d4363a422281a70c7377839011 : SUCCESS in 36m 38s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/cb0057d395b7447db5e30acf277068ba : SUCCESS in 27m 13s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/e9e10238be424638818c9abd835a72a8 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":16},{"id":"93519be2bb6c104b8e3f324a73ed6b7dcfb89da1","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2021-01-05 00:43:30.000000000","message":"Patch Set 16:\n\nI tried this PS along with https://review.opendev.org/c/openstack/openstack-helm/+/753413. However, it does not work without the below changes...\n\na. runAsUser:0 for nova-compute-ssh container. Without it, it fails to write host keys due to \"permission denied\" error.\nb. SSH daemon expects /run/sshd directory.\nc. It also needed \"usermod -s /bin/bash nova\"","accounts_in_message":[],"_revision_number":16},{"id":"6a82c82b9b965a875eb7a9786afe80c3c7721f65","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2021-01-05 17:33:15.000000000","message":"Patch Set 16:\n\n\u003e Patch Set 16:\n\u003e \n\u003e I tried this PS along with https://review.opendev.org/c/openstack/openstack-helm/+/753413. However, it does not work without the below changes...\n\u003e \n\u003e a. runAsUser:0 for nova-compute-ssh container. Without it, it fails to write host keys due to \"permission denied\" error.\n\u003e b. SSH daemon expects /run/sshd directory.\n\u003e c. It also needed \"usermod -s /bin/bash nova\"\n\nAlso, can you restore this change: https://review.opendev.org/c/openstack/openstack-helm-images/+/753358?","accounts_in_message":[],"_revision_number":16},{"id":"da5b44502aa00c47971a9b7d41a4a53cc4c7e6e9","author":{"_account_id":18256,"name":"Hemachandra Reddy","email":"hemachandra.reddy@att.com","username":"hr858f"},"date":"2021-01-08 00:42:02.000000000","message":"Patch Set 16: Code-Review+1\n\n\u003e Patch Set 16:\n\u003e \n\u003e I tried this PS along with https://review.opendev.org/c/openstack/openstack-helm/+/753413. However, it does not work without the below changes...\n\u003e \n\u003e a. runAsUser:0 for nova-compute-ssh container. Without it, it fails to write host keys due to \"permission denied\" error.\n\u003e b. SSH daemon expects /run/sshd directory.\n\u003e c. It also needed \"usermod -s /bin/bash nova\"\n\nI did not have to do these when I used https://review.opendev.org/c/openstack/openstack-helm-images/+/753358.","accounts_in_message":[],"_revision_number":16},{"id":"627368b82103d2aacc944d79cb89f4de815df87c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-01-08 01:19:39.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"34b18d601826901e6e3867e81796cfc78b0865de","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-01-26 15:23:16.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"07a7f87e1b54b4631699ff9b7445846029eda42e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 15:23:54.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"e63dcc5926e9c91e91f871a588216c2fdb04da7f","author":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"},"date":"2021-01-26 15:27:32.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"855211ed5d3b57a31fa8c4d60b5cf15b2c0b2d83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 16:28:19.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/d0c0b41cb39f4583b7f633afb6b220fb : SUCCESS in 43m 40s\n- loci-glance https://zuul.opendev.org/t/openstack/build/2adeb6238beb41a7beeb2bfe3085150b : SUCCESS in 52m 51s\n- loci-heat https://zuul.opendev.org/t/openstack/build/76cbe2f5173c4cee99055c6c3febb506 : SUCCESS in 35m 49s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/94f5c720b7fd4426b0058423e95b228c : POST_FAILURE in 38m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b6adad16622d42eb8331b1a227fa7960 : SUCCESS in 51m 07s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/906d96cbb7b94338a3337f8fd64a5ceb : TIMED_OUT in 1h 01m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/c464489885864970913cad13182fb1c6 : TIMED_OUT in 1h 00m 55s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d25901149b1344d390d51018d5655d76 : TIMED_OUT in 1h 01m 15s\n- loci-nova https://zuul.opendev.org/t/openstack/build/bdf05ac91ce14601b3a55377205bcefc : TIMED_OUT in 1h 01m 59s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/62021aa7b4604982aac0bca25ecaaaa7 : TIMED_OUT in 1h 01m 19s\n- loci-placement https://zuul.opendev.org/t/openstack/build/5a0622fc3ef24972b006f59004891920 : SUCCESS in 30m 39s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/447a6f63c8314706b119f37e4660d027 : SUCCESS in 27m 15s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/de19e1c5335e4d9c9e8ee5c0f10e551f : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":16},{"id":"77a9ee698704aa1375c1b9c8484926b64d62b38c","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2021-01-26 18:08:15.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"b07a3c0b7a31f62712d82786fa51c28585ca97c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 19:16:27.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f84dadfad6d04fceb2121fc558771b18 : SUCCESS in 41m 23s\n- loci-glance https://zuul.opendev.org/t/openstack/build/90d63392547b4b5cacdafa10dcdd9fcf : SUCCESS in 40m 11s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c09f9922b0064de1a1408a45e216cc95 : SUCCESS in 35m 32s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ae9ffea78e074c4bba0534c4b8f991da : SUCCESS in 34m 25s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/70463c7e22be4389878097c482843dc9 : SUCCESS in 34m 28s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/115365f86fab49e9b18f290d15ef3caf : SUCCESS in 42m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/0df8ca00e3414b93b96cb769d8d3415b : SUCCESS in 36m 06s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/48ce59bc1aeb49f0bad64d77604471f7 : SUCCESS in 46m 24s\n- loci-nova https://zuul.opendev.org/t/openstack/build/940863ead1014a0ea788db7315435b9c : SUCCESS in 49m 53s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f02257a113244e35b8d186502a9f9763 : SUCCESS in 43m 53s\n- loci-placement https://zuul.opendev.org/t/openstack/build/255d6d9e577149a8859962b0093becf3 : SUCCESS in 38m 00s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/97c7af18862a482d8d610d8c7cadd842 : SUCCESS in 32m 21s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/9bbc621b80894bae88ad049bbf52f930 : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":16},{"id":"93ef50f0ef0a7dbcbf4752e8b982df3b10aa39fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 19:16:41.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"68505bcc3417e0654d9d42a9bf7a79459f19605f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:05:23.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/2543450efde645dcaddb611593f7a5c4 : SUCCESS in 40m 55s\n- loci-glance https://zuul.opendev.org/t/openstack/build/24968b9f3f384554837a49bc58470157 : SUCCESS in 37m 42s\n- loci-heat https://zuul.opendev.org/t/openstack/build/60c36fff08fb4642a0f91700bc7d316d : SUCCESS in 44m 24s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/cbb74422f72a424ca991b37112ef3340 : SUCCESS in 38m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/b55eca87bbcd4df992bd4ef900f30708 : SUCCESS in 41m 29s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/a2be468687ba434484973a53de64e769 : SUCCESS in 37m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/436e3ce4adfb459483252bda141cd411 : SUCCESS in 36m 17s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/f0242ee5619a44708d5153c8423dff67 : SUCCESS in 38m 46s\n- loci-nova https://zuul.opendev.org/t/openstack/build/6f60bf50fc2440c88c2f958881461ce8 : SUCCESS in 43m 14s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/8f659655f3ad49f5b5c28c1068814046 : SUCCESS in 38m 28s\n- loci-placement https://zuul.opendev.org/t/openstack/build/cfe0e3b6094b4d3da4dc3b8461339fad : SUCCESS in 33m 33s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c82a13f1e672405c990524b1cbdc114e : SUCCESS in 25m 56s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b8ee3f654b034554849db3132d695358 : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":16},{"id":"0292afd07794e4ef202a688ca3c6a33e7fee03bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 20:05:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"2652595c2cc031b58efef6e62132b5e4d0b5ee6a","revisions":{"49301e6bc05d06436beabac7bb0542daba1d9125":{"kind":"REWORK","_number":1,"created":"2020-09-22 18:34:30.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/1"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:34:26.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/49301e6bc05d06436beabac7bb0542daba1d9125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/49301e6bc05d06436beabac7bb0542daba1d9125"}]},"branch":"refs/heads/master"},"ac31df7ed5f803d1e0917c3eea0fc090837c99d8":{"kind":"REWORK","_number":2,"created":"2020-09-22 21:19:16.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/2"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 21:19:02.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac31df7ed5f803d1e0917c3eea0fc090837c99d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/ac31df7ed5f803d1e0917c3eea0fc090837c99d8"}]},"branch":"refs/heads/master"},"1c21d4a7df4845e60c18c278a0a6746d1d83d3a2":{"kind":"REWORK","_number":3,"created":"2020-09-22 22:06:54.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/3"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 22:06:40.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1c21d4a7df4845e60c18c278a0a6746d1d83d3a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1c21d4a7df4845e60c18c278a0a6746d1d83d3a2"}]},"branch":"refs/heads/master"},"7878641b75c547bf21068c72b6b3547534a95250":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-09-22 22:09:12.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/4"}}},"commit":{"parents":[{"commit":"409c784dc524be641370094f3353d6579ab54603","subject":"gate: fix repository url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/409c784dc524be641370094f3353d6579ab54603"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 22:08:49.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7878641b75c547bf21068c72b6b3547534a95250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/7878641b75c547bf21068c72b6b3547534a95250"}]},"branch":"refs/heads/master"},"95240eabb8068e2a66b2cd1c06455421f0707c47":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-23 13:37:00.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/5"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"Oleksandr Kozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-23 13:37:00.000000000","tz":0},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/95240eabb8068e2a66b2cd1c06455421f0707c47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/95240eabb8068e2a66b2cd1c06455421f0707c47"}]},"branch":"refs/heads/master"},"1d1abb88be8c9b1f7cdac4fe75fd1773e6298e48":{"kind":"NO_CHANGE","_number":6,"created":"2020-09-23 15:39:27.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/6"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-23 15:39:05.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1d1abb88be8c9b1f7cdac4fe75fd1773e6298e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1d1abb88be8c9b1f7cdac4fe75fd1773e6298e48"}]},"branch":"refs/heads/master"},"4be1e1323f0aca2b36cce226a63facded2efbc04":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-24 12:47:19.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/7"}}},"commit":{"parents":[{"commit":"409c784dc524be641370094f3353d6579ab54603","subject":"gate: fix repository url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/409c784dc524be641370094f3353d6579ab54603"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-24 12:47:00.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4be1e1323f0aca2b36cce226a63facded2efbc04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/4be1e1323f0aca2b36cce226a63facded2efbc04"}]},"branch":"refs/heads/master"},"564ffad179cbec13ead1d84e1ad9babf98495824":{"kind":"REWORK","_number":8,"created":"2020-09-24 21:06:42.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/8"}}},"commit":{"parents":[{"commit":"409c784dc524be641370094f3353d6579ab54603","subject":"gate: fix repository url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/409c784dc524be641370094f3353d6579ab54603"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-24 20:39:20.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/564ffad179cbec13ead1d84e1ad9babf98495824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/564ffad179cbec13ead1d84e1ad9babf98495824"}]},"branch":"refs/heads/master"},"321b4741ceb174c9c011d1a11bea669df8af6e7e":{"kind":"REWORK","_number":9,"created":"2020-10-19 17:39:56.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/9"}}},"commit":{"parents":[{"commit":"409c784dc524be641370094f3353d6579ab54603","subject":"gate: fix repository url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/409c784dc524be641370094f3353d6579ab54603"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-10-19 17:39:42.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/321b4741ceb174c9c011d1a11bea669df8af6e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/321b4741ceb174c9c011d1a11bea669df8af6e7e"}]},"branch":"refs/heads/master"},"449741b84aa44879ae2ebda64cf1799039739db7":{"kind":"REWORK","_number":10,"created":"2020-10-19 18:18:26.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/10"}}},"commit":{"parents":[{"commit":"409c784dc524be641370094f3353d6579ab54603","subject":"gate: fix repository url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/409c784dc524be641370094f3353d6579ab54603"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-10-19 18:18:14.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/449741b84aa44879ae2ebda64cf1799039739db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/449741b84aa44879ae2ebda64cf1799039739db7"}]},"branch":"refs/heads/master"},"67b07e11af74afcfa998e02b75a5c0fef18e6242":{"kind":"REWORK","_number":11,"created":"2020-10-19 20:39:57.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/11"}}},"commit":{"parents":[{"commit":"409c784dc524be641370094f3353d6579ab54603","subject":"gate: fix repository url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/409c784dc524be641370094f3353d6579ab54603"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-10-19 20:39:46.000000000","tz":180},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67b07e11af74afcfa998e02b75a5c0fef18e6242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/67b07e11af74afcfa998e02b75a5c0fef18e6242"}]},"branch":"refs/heads/master"},"8a4910dc26f4cc72513f3e59d7ad7efa770aa52f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-12-10 11:55:43.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/12"}}},"commit":{"parents":[{"commit":"115396b0aae975d2cd0778ff47fd8c63b0b2e608","subject":"aarch64: enable building multi-arch images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/115396b0aae975d2cd0778ff47fd8c63b0b2e608"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-12-10 11:55:38.000000000","tz":120},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8a4910dc26f4cc72513f3e59d7ad7efa770aa52f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/8a4910dc26f4cc72513f3e59d7ad7efa770aa52f"}]},"branch":"refs/heads/master"},"1736432a0b426243bd862ece47130b1098617736":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2020-12-10 19:00:24.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/13"}}},"commit":{"parents":[{"commit":"115396b0aae975d2cd0778ff47fd8c63b0b2e608","subject":"aarch64: enable building multi-arch images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/115396b0aae975d2cd0778ff47fd8c63b0b2e608"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-12-10 18:59:50.000000000","tz":120},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nDepends-On: https://review.opendev.org/766510\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1736432a0b426243bd862ece47130b1098617736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/1736432a0b426243bd862ece47130b1098617736"}]},"branch":"refs/heads/master"},"af8952ed6d704756a3516d1d632130eecf891ab3":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2020-12-10 19:07:55.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/14"}}},"commit":{"parents":[{"commit":"115396b0aae975d2cd0778ff47fd8c63b0b2e608","subject":"aarch64: enable building multi-arch images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/115396b0aae975d2cd0778ff47fd8c63b0b2e608"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-12-10 19:07:37.000000000","tz":120},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/af8952ed6d704756a3516d1d632130eecf891ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/af8952ed6d704756a3516d1d632130eecf891ab3"}]},"branch":"refs/heads/master"},"404af847a5b52bc3cd53f1fd0857a74e7e20398e":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-12-10 19:33:31.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/24/753424/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/15"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-12-10 19:33:31.000000000","tz":0},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/404af847a5b52bc3cd53f1fd0857a74e7e20398e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/404af847a5b52bc3cd53f1fd0857a74e7e20398e"}]},"branch":"refs/heads/master","description":"Rebase"},"2652595c2cc031b58efef6e62132b5e4d0b5ee6a":{"kind":"NO_CHANGE","_number":16,"created":"2020-12-14 13:12:23.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/24/753424/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/24/753424/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/24/753424/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/24/753424/16"}}},"commit":{"parents":[{"commit":"5e3fa8dd2c2cecec46401a640e24fe1f86795597","subject":"Fix gates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/5e3fa8dd2c2cecec46401a640e24fe1f86795597"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-09-22 18:33:38.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-12-14 13:12:07.000000000","tz":120},"subject":"Add openssh client to binary deps","message":"Add openssh client to binary deps\n\nChange-Id: Ie6a87868dda1f2d5b7746a53b5b9e8c60869167f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2652595c2cc031b58efef6e62132b5e4d0b5ee6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/2652595c2cc031b58efef6e62132b5e4d0b5ee6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":18250,"name":"Roy Tang","email":"roy.s.tang@att.com","username":"rt7380"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
