)]}'
{"id":"openstack%2Floci~753452","triplet_id":"openstack%2Floci~master~I93ea021810568a7998c99958c622aacb3d39e908","project":"openstack/loci","branch":"master","topic":"refactor-setup-gate","hashtags":[],"change_id":"I93ea021810568a7998c99958c622aacb3d39e908","subject":"gate: fix repository url","status":"MERGED","created":"2020-09-22 21:50:22.000000000","updated":"2020-09-25 17:14:03.000000000","submitted":"2020-09-25 17:14:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753452-1601054043478-84033bed","meta_rev_id":"2cc4747b029ccd77c276e596d8396702e33809ca","_number":753452,"virtual_id_number":753452,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 17:14:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":2,"date":"2020-09-25 14:02:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-25 02:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":1,"date":"2020-09-25 14:02:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 02:58:14.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2020-09-25 14:02:10.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-09-25 17:14:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da585ccff9252de720909f34dcb3ccd4c7a69e7b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-22 21:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c46489c978bf84c2ae60f15885edcc4101c9420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 22:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/b814c12180e6474697342eacaf90c692 : SUCCESS in 38m 09s\n- loci-glance https://zuul.opendev.org/t/openstack/build/246a3a19c0cf46f198235a9c9cc4589a : SUCCESS in 34m 52s\n- loci-heat https://zuul.opendev.org/t/openstack/build/ca8061e839f54043b5d8c654c1572141 : SUCCESS in 20m 05s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/fb590ae4ffd8425c8920c8411dafbc47 : SUCCESS in 38m 56s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e020c6e9f109496e90bd754e9361a40a : SUCCESS in 37m 38s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/3e00ed2e8d314e078334b5b4e65ee2c9 : SUCCESS in 35m 42s\n- loci-manila https://zuul.opendev.org/t/openstack/build/9327dee5bf4e49f7953903670a6e7063 : SUCCESS in 35m 15s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7630c13be724447b92a7ae174a804af4 : SUCCESS in 38m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/9f7bddd704a34bb1a706721e26af6873 : SUCCESS in 40m 47s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/76f15f59ce0d42ed8f5c666ad7399b1c : SUCCESS in 42m 01s\n- loci-placement https://zuul.opendev.org/t/openstack/build/29ca6522b9724a8da8c2514fb76161b3 : SUCCESS in 32m 31s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/c4b539e8f6ed44c5bd0e61c30886395a : SUCCESS in 27m 00s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/dab07f6880bc431682e9248a3b57b4b8 : SUCCESS in 35m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ff7e44466fd46c9e4c025a061f9c0ffdd7382bc7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-09-23 00:43:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7720a3b81ae617580ee72fe0b48aa5b8fab49ad0","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-24 23:16:40.000000000","message":"Patch Set 1:\n\nlets try to go with https://review.opendev.org/#/c/754228","accounts_in_message":[],"_revision_number":1},{"id":"3efc1eb0e648246eb9a7017965f0f253980face7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-09-24 23:18:54.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"955ba18e96d07f84045beeec38226c390cf02c99","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-24 23:24:07.000000000","message":"Patch Set 1: Code-Review+2\n\nGood for me. Missed it\u0027s setting up a registry also here","accounts_in_message":[],"_revision_number":1},{"id":"476e4e0b3aa0f8f184733d6491717106f3e6d53d","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-24 23:41:02.000000000","message":"Patch Set 1: -Code-Review\n\nlets still go with https://review.opendev.org/#/c/754228","accounts_in_message":[],"_revision_number":1},{"id":"8b85404983dd7972ba72a7b105ccb5031d985eae","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-25 02:58:14.000000000","message":"Patch Set 1: Code-Review+2\n\nok i figured there\u0027s a bunch of logic we can\u0027t remove and using ensure-docker is an overhead as there\u0027s a lot of stuff we don\u0027t need.","accounts_in_message":[],"_revision_number":1},{"id":"4744930686863bf2cf1e24cbd1b844a4349dfead","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-09-25 14:02:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9bae4a41172fb0360358e33d5ba0b5dcd39274f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 14:02:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b793ff6fe1bdb0614961f5e01595bb9436e6673","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 15:08:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/fdeceaefd28b471bb263daab7f73325a : SUCCESS in 36m 46s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e9b29680806d46a68d8cde37b20a5dd2 : SUCCESS in 35m 51s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e35966ac2bb24d66bad04d673106f44d : SUCCESS in 38m 49s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ce0fcf2fc82e4995b9ee4f1d320e0757 : SUCCESS in 40m 19s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/5033abe5f24f4826bdcb410470c4161f : POST_FAILURE in 59m 42s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/4323c51becce4601817f3f72b9d17b7e : SUCCESS in 34m 36s\n- loci-manila https://zuul.opendev.org/t/openstack/build/5690e29be8294f3980649f3713d11518 : SUCCESS in 40m 02s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/824768fce6734c56845f33beaae8d19d : SUCCESS in 40m 14s\n- loci-nova https://zuul.opendev.org/t/openstack/build/412aec14caeb48c79cf01a23fc7410d2 : SUCCESS in 42m 24s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/7e585ce4745e4123a0ff01f613694e9b : SUCCESS in 33m 34s\n- loci-placement https://zuul.opendev.org/t/openstack/build/e8bf44a90a544380a7ae8492ab45ab08 : SUCCESS in 27m 31s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/f8dc3c901ed848638c5ce97455c7bd32 : SUCCESS in 25m 56s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/7ef4b7dbedd3495ca0db997c1d8fb765 : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b1507b4c889912a151f8a6dac9303bb44f28dc20","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-09-25 15:20:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8190b31b0f9ca43caf0bf5c2cecc7affb305c1ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 16:24:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/1940fb13924147368f580bf30f00deaf : SUCCESS in 34m 09s\n- loci-glance https://zuul.opendev.org/t/openstack/build/341d879a54654c1eadee96770a5c973c : SUCCESS in 33m 21s\n- loci-heat https://zuul.opendev.org/t/openstack/build/beebd745545b49429c90bd22fb379235 : SUCCESS in 35m 13s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/4ef22b52638c4450a2f190d7c3eb96d7 : SUCCESS in 40m 26s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/be22ac61589b43e0977d6f98e93e321d : SUCCESS in 36m 14s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/5e86a9fbb41846fcaa3bfe57c1fb2c09 : SUCCESS in 37m 54s\n- loci-manila https://zuul.opendev.org/t/openstack/build/39817b0bd40c4a3b8d922de09a4cde11 : SUCCESS in 43m 12s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/5660cc9edb744972b17c5d2a342511a7 : SUCCESS in 39m 53s\n- loci-nova https://zuul.opendev.org/t/openstack/build/3afd13becf014e45bd698dbc93a1ee42 : SUCCESS in 48m 13s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/722c6a62f79d4328a04801231e770ec7 : SUCCESS in 39m 05s\n- loci-placement https://zuul.opendev.org/t/openstack/build/2405f7c00fb64e6ab798f625f06d4fb5 : SUCCESS in 30m 01s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/94bbfd5ed7a94d4e822d3477484bdefa : SUCCESS in 26m 31s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/4d05d2176cc34ccf858ed841f1764db0 : SUCCESS in 34m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a3755d713db263b76fa6c442845b2ce5a1405262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 16:24:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa3c85ca55596cab61a92ae60906da696b400d75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 17:14:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/252182b6806d4dc790b24a8b62e740bb : SUCCESS in 37m 36s\n- loci-glance https://zuul.opendev.org/t/openstack/build/e1968ba42207435ca2fa77e2fafa198d : SUCCESS in 20m 04s\n- loci-heat https://zuul.opendev.org/t/openstack/build/53314da3dd534406bf99a558fe9974ad : SUCCESS in 19m 33s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/3670a83220b349c4b1f1cfb4ec69ec42 : SUCCESS in 42m 23s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/d0c7ef2cf2944669abc00106ba88a539 : SUCCESS in 36m 50s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/613967ac2ea7432ba4003f3bd41e896b : SUCCESS in 38m 29s\n- loci-manila https://zuul.opendev.org/t/openstack/build/60d7c7b86e394ca6958046e3ae2b507a : SUCCESS in 37m 28s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/8bb1c52175c34f58b8dfec251d1b8210 : SUCCESS in 40m 20s\n- loci-nova https://zuul.opendev.org/t/openstack/build/a719efcc968d4b1ea4e9c3cc6452823b : SUCCESS in 36m 35s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/4ae384ed265a405da08f4d2b305f5672 : SUCCESS in 36m 03s\n- loci-placement https://zuul.opendev.org/t/openstack/build/6ef8d8e01586442f8c5a87f6b785f2d7 : SUCCESS in 27m 23s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/56255586445c46f09a2ff71f26462a4b : SUCCESS in 25m 07s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/199e7a0394a641b1a6a1c6c887210979 : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":1},{"id":"38a2d939bb941c1860a30532733c9d107c44fe26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 17:14:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"409c784dc524be641370094f3353d6579ab54603","revisions":{"409c784dc524be641370094f3353d6579ab54603":{"kind":"REWORK","_number":1,"created":"2020-09-22 21:50:22.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/52/753452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/52/753452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/753452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/753452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/52/753452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/52/753452/1"}}},"commit":{"parents":[{"commit":"9481282d23a8a9fd224bcadf2a1467595cf31719","subject":"Merge \"Add support for unnamed constraints to loci with new pip resolver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/9481282d23a8a9fd224bcadf2a1467595cf31719"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-22 21:49:51.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-22 21:49:51.000000000","tz":-240},"subject":"gate: fix repository url","message":"gate: fix repository url\n\nThe repository was changed by the OpenDev infrastructure in order to\nproperly mirror systems, this change reflects that modification.\n\nChange-Id: I93ea021810568a7998c99958c622aacb3d39e908\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/409c784dc524be641370094f3353d6579ab54603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/409c784dc524be641370094f3353d6579ab54603"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
