)]}'
{"id":"openstack%2Floci~753455","triplet_id":"openstack%2Floci~master~If0a90919cdddaf91a67ce96646dea87cf0b9632c","project":"openstack/loci","branch":"master","topic":"refactor-setup-gate","attention_set":{},"removed_from_attention_set":{"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2022-02-21 08:01:04.000000000","reason":"Oleksandr Kozachenko replied on the change"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2022-03-01 23:37:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0a90919cdddaf91a67ce96646dea87cf0b9632c","subject":"aarch64: stop trying to build confluent-kafka","status":"MERGED","created":"2020-09-22 22:01:22.000000000","updated":"2022-03-01 23:37:05.000000000","submitted":"2022-03-01 23:37:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753455-refactor-setup-gate","meta_rev_id":"969f440065b2f740a15ca13b424b558ce1929d4d","_number":753455,"virtual_id_number":753455,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 23:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"recommended":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"all":[{"value":1,"date":"2022-03-01 23:37:05.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-01 21:41:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2022-03-01 23:37:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-01 21:41:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 22:55:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 21:35:41.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-01 15:58:29.000000000","updated_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"reviewer":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"state":"REVIEWER"},{"updated":"2022-02-17 15:11:50.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"CC"},{"updated":"2022-02-21 08:01:04.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"REVIEWER"}],"messages":[{"id":"c454d39845ad74d82e02cd2bfb56d1a0012cbda0","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-22 22:01:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fd953ba3694dd9b74ecdde9d911509c1223b224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 22:55:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0061666aedb94a57b9d1018ade09d0d8 : SUCCESS in 40m 17s\n- loci-glance https://zuul.opendev.org/t/openstack/build/15bca0ae510544e8bea35a9585336cfb : SUCCESS in 33m 05s\n- loci-heat https://zuul.opendev.org/t/openstack/build/67f75beeab404f91abd7c443e638f085 : SUCCESS in 32m 59s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/754c76e081704ccd88d0fe788ff2320e : SUCCESS in 35m 26s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/e2603312658640d093f7cfc7d41b674e : SUCCESS in 43m 22s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b89ed28d53d14696a84f066abc9717a5 : SUCCESS in 34m 02s\n- loci-manila https://zuul.opendev.org/t/openstack/build/f5747398208c42a7b027fb0780cce700 : SUCCESS in 35m 32s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4199782e46df4afdb544968b09f23a7d : SUCCESS in 36m 16s\n- loci-nova https://zuul.opendev.org/t/openstack/build/d695216d3acc4fc584398c8ce6a0c660 : SUCCESS in 42m 09s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/cf394e3fded643f4a58e60b9a7b59765 : SUCCESS in 33m 47s\n- loci-placement https://zuul.opendev.org/t/openstack/build/de4e5ac7cd544a1d96594724cd7f3b8c : SUCCESS in 26m 17s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8ad93a6d63cb4370a656e00f074c7e66 : SUCCESS in 26m 21s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/73a6240328534cf1aba78dcabe7ce703 : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":1},{"id":"955cc86ed6fbb51779991258c5f71a1a9fd6aa3f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-09-30 16:05:10.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this or create a new change.","accounts_in_message":[],"_revision_number":1},{"id":"7d882de5db5821c8777a42649a381b00ab5334c0","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-10-22 02:16:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d0cd8883b87dc2c4b5f7058e70b976b6cb41521","tag":"autogenerated:gerrit:restore","author":{"_account_id":33519,"name":"Tin Lam","email":"tinlam@outlook.com","username":"tlam"},"date":"2021-11-12 16:48:34.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a7db79b24ee5761293a0aa9fbb33941580a5e304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 17:34:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/92b9d2c296114e63a6feef4a4556dc67 : FAILURE in 37m 51s\n- loci-glance https://zuul.opendev.org/t/openstack/build/59ecad5b5720446ba1fd523ea9a3da9f : FAILURE in 28m 54s\n- loci-heat https://zuul.opendev.org/t/openstack/build/7a1c79e606094fe3bbbb6e7bf0397702 : FAILURE in 28m 43s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a6d8826f5bfa477c9ad36dceedee85eb : FAILURE in 29m 02s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/4686e4c260784e39857d10e2b404d47b : FAILURE in 29m 39s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/9b72f3b737584c15bfe51b5162c3bb55 : FAILURE in 28m 46s\n- loci-manila https://zuul.opendev.org/t/openstack/build/fb6bb957f8e54f21bcda39ae88e77432 : FAILURE in 30m 04s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/9546cb46eb62463eb117438a2a81e19c : FAILURE in 31m 18s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/2ca6990712ad4ebe8da2586b5314222e : FAILURE in 29m 27s\n- loci-nova https://zuul.opendev.org/t/openstack/build/723d8d786774411591d6d2acbc7873ab : FAILURE in 31m 40s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/113686bab8324353abd7b828eff64cff : FAILURE in 30m 10s\n- loci-placement https://zuul.opendev.org/t/openstack/build/05c2cc394e584a688ea7b157ae062830 : FAILURE in 29m 59s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/3aef9edd327945b6b0089bf35aaef587 : FAILURE in 29m 33s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/384511eb11cb4c30905c0c18cb0f452d : FAILURE in 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a03b1373ea12c20d1424456ba745592ecd7fda56","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2021-11-25 13:17:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6e51f5aa1bea9ef978b96d08eea6c864e5e3ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 13:51:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/2e1f75bc9f1c4425ba3ba9c8f91ab450 : FAILURE in 24m 39s\n- loci-glance https://zuul.opendev.org/t/openstack/build/196cae7f27b0428cbe1c934f476f41ef : FAILURE in 29m 19s\n- loci-heat https://zuul.opendev.org/t/openstack/build/226fe927d4ed4bb79a405e355d7d92e9 : FAILURE in 26m 21s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/a5176b57ec9e46979c91dbf81e9e7a57 : FAILURE in 28m 22s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9800c991ae864241908fa3defa88ec78 : FAILURE in 23m 47s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/fc2f5d6b7bf942488ef2ca55fbd94d97 : FAILURE in 26m 16s\n- loci-manila https://zuul.opendev.org/t/openstack/build/815e7cfd09d447bca4bd21704e248e74 : FAILURE in 31m 18s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/be831e6ad7da44a4b2541c63b1cc0220 : FAILURE in 23m 03s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/7d8586e2406a42038d113dd8b304c9c7 : FAILURE in 30m 02s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c4e73ee257a34a9b86e567b006713eba : FAILURE in 23m 21s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/062f3d1f41ba43a08a71dc7621aa8ad9 : FAILURE in 29m 00s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f11dcab19f6049218e279e2d8c43bc62 : FAILURE in 21m 21s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/ea73e3337b4a449195c20b735fe1f106 : FAILURE in 23m 29s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/fd52fbf6d70f41c186f953d311f0c16f : FAILURE in 30m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6b27b4a3267d9e48ba9b9b6b21c6bbfb4ef33d47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-01-18 08:37:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5884dde1b158bf0995185f1f742ada94d3e0b6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 09:27:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a32429867b9a411c8f66e71ff0c59d97 : SUCCESS in 38m 15s\n- loci-glance https://zuul.opendev.org/t/openstack/build/77a8524eb90b48e7adf13121db5487db : SUCCESS in 31m 12s\n- loci-heat https://zuul.opendev.org/t/openstack/build/187e5d6fab6a4a298f6e53534b1689ab : SUCCESS in 36m 21s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/e1aebdf664c240c98c7a0f2da7dbaa46 : SUCCESS in 33m 44s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/9e8554a8e28443cfb35f013abff37bd0 : SUCCESS in 37m 01s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/1a7f26fe00b34c438c8a82ce9fd61152 : SUCCESS in 32m 14s\n- loci-manila https://zuul.opendev.org/t/openstack/build/89cac2c05267402eb99053b6410080a3 : SUCCESS in 45m 37s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/f9632e843ab4454daa72505dc0ea0971 : SUCCESS in 31m 29s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/b5d15ff0da1f4904b6f9e1c0987fed33 : SUCCESS in 36m 06s\n- loci-nova https://zuul.opendev.org/t/openstack/build/c530c771e36244dab348e281014bceeb : SUCCESS in 35m 39s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/ff9e3301db0c4454846562d319212907 : SUCCESS in 36m 42s\n- loci-placement https://zuul.opendev.org/t/openstack/build/f67dd37b8f624cb7b7182db1697332eb : SUCCESS in 19m 20s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6e6ffa57be994ee8b146ab61a0e28119 : SUCCESS in 28m 46s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/468edc9ad9af4a4a8b143a3e05a070bc : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c05a9f173a980b74cf44eb8bb3ee89fcea4a6d75","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-21 21:35:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"27554f029911bb305c0d854a9ed4c427e9e8dad2","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2022-02-01 15:58:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a4ced2770a897177d7bbc8a2cef59caff1cc7ab1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 15:59:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc5650affd42200d9d484c8bc2f6de67c0591cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:47:11.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/a6bad2413e27401abcc3855645275798 : FAILURE in 5m 22s\n- loci-glance https://zuul.opendev.org/t/openstack/build/895bfe0f7d904d82bebd160a3681b9c4 : FAILURE in 5m 28s\n- loci-heat https://zuul.opendev.org/t/openstack/build/1b810f63f62c496c9be3897c50455bae : FAILURE in 5m 41s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/35961b2299c04b99832eea5c18b1a1f2 : FAILURE in 4m 15s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/2fd2088aae924bca8df8d836937c17b6 : FAILURE in 4m 26s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/996e38d7f33649eaa364861cfbd78818 : FAILURE in 4m 27s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3853e5b5fbbd459d8ae121176886e4db : FAILURE in 5m 38s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/5cdaccc84fc14baa8c2db701b4b664ac : FAILURE in 4m 33s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/d46a7e5fc4774b4e91ffb5dd7e51bea5 : FAILURE in 4m 21s\n- loci-nova https://zuul.opendev.org/t/openstack/build/e5fc84d4a3904a198d6f65c6c0bde8d3 : FAILURE in 4m 34s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/2ce9b5969ece401a9c21f4bb1b21c42a : FAILURE in 5m 07s\n- loci-placement https://zuul.opendev.org/t/openstack/build/05700750ff2b431fa4f03c32a95c0353 : FAILURE in 4m 23s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/8d3cdcc98d25440fa7a1c3baaeed180e : FAILURE in 3m 32s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/b9c2b15f5e5e45b78ed26a4959038f48 : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"180c36c83ab8d06d640c70ec79a786295e5526e7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-01 16:48:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26f8f77717cbfce510521463c3f337dce7aeed82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:58:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/0ad28a0c5da84ed2a87c372454aba498 : FAILURE in 3m 30s\n- loci-glance https://zuul.opendev.org/t/openstack/build/d7de584fca6f47ba8c3fecb3854409d0 : FAILURE in 3m 36s\n- loci-heat https://zuul.opendev.org/t/openstack/build/c98739f0245e49e0a27ac7bca9cd4762 : FAILURE in 3m 44s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/ae43baf62797485c8923d3d2b0206ed8 : FAILURE in 3m 29s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/a153f38fc8944909a7d131a24ef9fdeb : FAILURE in 4m 21s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/b44913a308f74e098e995e5e518c97f6 : FAILURE in 3m 34s\n- loci-manila https://zuul.opendev.org/t/openstack/build/3aca2b0c448547d6908fa619087e5e5e : FAILURE in 3m 38s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/aa01f3369a354c0e8ce4bb168763e1fb : FAILURE in 3m 31s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/da27458678884bc385464037a55d8706 : FAILURE in 3m 34s\n- loci-nova https://zuul.opendev.org/t/openstack/build/81b11c48f1cc49b29c7bce29148b603e : FAILURE in 3m 45s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/666390ef1abb4cb6822a60c06b8868da : FAILURE in 3m 52s\n- loci-placement https://zuul.opendev.org/t/openstack/build/0028d71b2e02481aa34d5270a293e0f8 : FAILURE in 3m 12s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/12c4af3bbb4f4a64ae9e19c54fe23440 : FAILURE in 3m 34s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/d5ad80150f374f86bf118859e1bff2dc : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5b2bdd294d9093ab3cf501b0fc9f3b81d49b2e8b","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-02-17 15:11:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17552d45e27ac2256f44567a6241252bf114048c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 15:24:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/f2d1889ea1114acb8ccf782aa55d1d3f : FAILURE in 4m 01s\n- loci-glance https://zuul.opendev.org/t/openstack/build/7f28c489d8954cae9bd277fd4946ac75 : FAILURE in 3m 56s\n- loci-heat https://zuul.opendev.org/t/openstack/build/34648978d9b9478c8c65e96dbf1caa5d : FAILURE in 3m 31s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/89b99b8f8ea24dcaba6ce34da52e15b9 : FAILURE in 3m 51s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/766af3da7f514cf9b892471579d18cf5 : FAILURE in 3m 49s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/f27b4379b7c24988ac25e4d76c6b886c : FAILURE in 3m 52s\n- loci-manila https://zuul.opendev.org/t/openstack/build/79adcc29e93e4e96a6c3e1d0e7120eea : FAILURE in 3m 34s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/684aceaeb49048dab3f7c1060be5bca3 : FAILURE in 3m 54s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/78548877c85a4e1683f76dc75e3ffcd1 : FAILURE in 3m 33s\n- loci-nova https://zuul.opendev.org/t/openstack/build/63372e9e57e74ffba99205f4e3a99e7f : FAILURE in 4m 03s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/3556b6b1836e4b31aa363cc8fd2f96fe : FAILURE in 3m 59s\n- loci-placement https://zuul.opendev.org/t/openstack/build/444588e929224d7fa9e12b5722c0b19c : FAILURE in 3m 26s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/6cb7e1f157e941d594ca758a8a571a89 : FAILURE in 3m 42s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/be6035bdaa80447582010fcab4cf3668 : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"bb7f62555ff573f3c3a62ee256f2062d5768313d","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-02-21 08:01:04.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63c5fbaa5d4270e7fd266e60f8e850f180dbc968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-03-01 21:41:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"877b36b946123375508872891aeda2b279a7ceac","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-03-01 21:41:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b016cf1a52fa3134037103389f2a2f1d91ed10d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:26:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/776093952f2648a1b6b4008beee08141 : SUCCESS in 25m 39s\n- loci-glance https://zuul.opendev.org/t/openstack/build/a43d6220525e422da8e3bf729bd21049 : SUCCESS in 24m 24s\n- loci-heat https://zuul.opendev.org/t/openstack/build/e07919380eb04c288273a646990c5cbf : SUCCESS in 24m 39s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/c66d75aa071346b8bb399d19e52f0067 : SUCCESS in 32m 23s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/652254b8861f47ddb4dd39b714fb051f : SUCCESS in 26m 12s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/6e14bce439b146f8a6942811a6184719 : SUCCESS in 25m 53s\n- loci-manila https://zuul.opendev.org/t/openstack/build/d732911beae54739bb6f81d7963e6c04 : SUCCESS in 29m 58s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/42fdec2167b7446385dd5fa12cf03e79 : SUCCESS in 25m 37s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/4c8a7e891dd548408107a4fa8a498ad0 : SUCCESS in 27m 10s\n- loci-nova https://zuul.opendev.org/t/openstack/build/fae0ec8e082745348a0f605713a61241 : SUCCESS in 37m 39s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/f0999814c98d472cb739be01003fae87 : SUCCESS in 25m 58s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b7e956ff63e8484a89e9cc5dc984f0d1 : SUCCESS in 17m 23s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/cd468ce8ea4a4e3184570a2c7f382d1e : SUCCESS in 21m 35s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/46619337fcfd48e087541eb784a74238 : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":3},{"id":"594fef7297c5019d9bc6ee567679cab1e0b18031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 22:27:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"384ea20a48935e809b9c869d498b98a0ece87d02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 23:37:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- loci-cinder https://zuul.opendev.org/t/openstack/build/76d6f116ba084c90bbba35a8886c5e27 : SUCCESS in 1h 01m 17s\n- loci-glance https://zuul.opendev.org/t/openstack/build/0d8dc2b6979f42ecac21fc87f32978be : SUCCESS in 58m 25s\n- loci-heat https://zuul.opendev.org/t/openstack/build/32e3c728b679498cadc5369603adaff8 : SUCCESS in 31m 36s\n- loci-horizon https://zuul.opendev.org/t/openstack/build/8fb59194c63f4cfc80ff2240dbde733a : SUCCESS in 26m 19s\n- loci-ironic https://zuul.opendev.org/t/openstack/build/73566f970df846ddb1f3f8320e3248bd : SUCCESS in 28m 52s\n- loci-keystone https://zuul.opendev.org/t/openstack/build/99c934e4065b431a9a882b72a19c9cc9 : SUCCESS in 27m 34s\n- loci-manila https://zuul.opendev.org/t/openstack/build/70058f2347a94a8fab3486b1ed5c6ea0 : SUCCESS in 46m 20s\n- loci-monasca-api https://zuul.opendev.org/t/openstack/build/2f3290e952984cf9b242e09e1d9ac408 : SUCCESS in 27m 10s\n- loci-neutron https://zuul.opendev.org/t/openstack/build/11b1c6d831fa4fe0b5de44b0b38b0095 : SUCCESS in 31m 16s\n- loci-nova https://zuul.opendev.org/t/openstack/build/cd36402c202a45fb8148457740a8cb17 : SUCCESS in 31m 20s\n- loci-octavia https://zuul.opendev.org/t/openstack/build/53a5158f82fb430790d8339a86480e5c : SUCCESS in 46m 30s\n- loci-placement https://zuul.opendev.org/t/openstack/build/b35198d7073d4d7a9d865ae43c40955b : SUCCESS in 19m 25s\n- loci-requirements https://zuul.opendev.org/t/openstack/build/7ab3719517af4dd6b6873e25bc970860 : SUCCESS in 21m 55s\n- loci-watcher https://zuul.opendev.org/t/openstack/build/813c692b0126401eaed6632e0a39b414 : SUCCESS in 27m 16s","accounts_in_message":[],"_revision_number":3},{"id":"49fe1ecbae20bbbc62700dae93f870dc902e0bc2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 23:37:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a7a31b18c44de2c44b2995af819bb77ee34522bc","revisions":{"27ac34f7fc541cb7b23295a4798eda7af45f05f6":{"kind":"REWORK","_number":1,"created":"2020-09-22 22:01:22.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/55/753455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/55/753455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/753455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/753455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/753455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/55/753455/1"}}},"commit":{"parents":[{"commit":"409c784dc524be641370094f3353d6579ab54603","subject":"gate: fix repository url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/409c784dc524be641370094f3353d6579ab54603"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-22 22:00:25.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-22 22:00:25.000000000","tz":-240},"subject":"aarch64: stop trying to build confluent-kafka","message":"aarch64: stop trying to build confluent-kafka\n\nThis package requires a version of libfdkafka-dev that is too old inside\nthe distributinos to be able to build the wheel.  In amd64, it has\nwheels on PyPI so it doesn\u0027t attempt to build it but it fails to build\nit inside an aarch64 system.\n\nChange-Id: If0a90919cdddaf91a67ce96646dea87cf0b9632c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/27ac34f7fc541cb7b23295a4798eda7af45f05f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/27ac34f7fc541cb7b23295a4798eda7af45f05f6"}]},"branch":"refs/heads/master"},"feaced9ceaf34c01edf3088eeac45e63965a3b88":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-18 08:37:53.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/55/753455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/55/753455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/753455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/753455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/753455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/55/753455/2"}}},"commit":{"parents":[{"commit":"efccd0a853879ac6af6066eda09792d0d3afe9c0","subject":"Specify a different source for novnc/spice-html5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/efccd0a853879ac6af6066eda09792d0d3afe9c0"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-22 22:00:25.000000000","tz":-240},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-01-18 08:37:39.000000000","tz":660},"subject":"aarch64: stop trying to build confluent-kafka","message":"aarch64: stop trying to build confluent-kafka\n\nThis package requires a version of libfdkafka-dev that is too old inside\nthe distributinos to be able to build the wheel.  In amd64, it has\nwheels on PyPI so it doesn\u0027t attempt to build it but it fails to build\nit inside an aarch64 system.\n\nChange-Id: If0a90919cdddaf91a67ce96646dea87cf0b9632c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/feaced9ceaf34c01edf3088eeac45e63965a3b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/feaced9ceaf34c01edf3088eeac45e63965a3b88"}]},"branch":"refs/heads/master"},"a7a31b18c44de2c44b2995af819bb77ee34522bc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-01 21:41:48.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/55/753455/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/loci","ref":"refs/changes/55/753455/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/753455/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/753455/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/loci refs/changes/55/753455/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/loci refs/changes/55/753455/3"}}},"commit":{"parents":[{"commit":"c488166421e868c3b348208ad8af8a4922942f66","subject":"Add neutron vpn packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/c488166421e868c3b348208ad8af8a4922942f66"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-09-22 22:00:25.000000000","tz":-240},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-03-01 21:41:48.000000000","tz":0},"subject":"aarch64: stop trying to build confluent-kafka","message":"aarch64: stop trying to build confluent-kafka\n\nThis package requires a version of libfdkafka-dev that is too old inside\nthe distributinos to be able to build the wheel.  In amd64, it has\nwheels on PyPI so it doesn\u0027t attempt to build it but it fails to build\nit inside an aarch64 system.\n\nChange-Id: If0a90919cdddaf91a67ce96646dea87cf0b9632c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a7a31b18c44de2c44b2995af819bb77ee34522bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/loci/commit/a7a31b18c44de2c44b2995af819bb77ee34522bc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
